)]}'
{"id":"openstack%2Foctavia~810996","triplet_id":"openstack%2Foctavia~master~Ied5519087dac7ad1c6c4fd453dc6ce2af4348cd9","project":"openstack/octavia","branch":"master","attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-09-05 14:08:48.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ied5519087dac7ad1c6c4fd453dc6ce2af4348cd9","subject":"Do logrotate for amphora/haproxy hourly by size","status":"NEW","created":"2021-09-24 23:05:12.000000000","updated":"2022-09-05 14:08:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4dd6b22843a82119283fec1ec0869b796987b0b4","_number":810996,"virtual_id_number":810996,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-25 01:36:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-25 01:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-05 14:08:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"e6d284e5eb5ec616df3b4a9f827fb02960fdcc99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-09-24 23:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40968fd52a7aac7cc1572e91c6b575fce5212565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-25 01:36:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7af11a53501403ea083699b2171d4b4 : SUCCESS in 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b4b85f85b424653b9a028d1177bc8d1 : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c068d43975cf49a9bcb87176b0d8441b : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aaedc80365a495e8eb7095c8971e2ff : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf87aab9a12249a18a3fc4231d3f86d7 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25e9cb8479284747bc012b50d5f21998 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a97911cbfd845d6b99a7de5f8fddeef : SUCCESS in 13m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/047d7500737848c6ad168d68c4fd2b91 : SUCCESS in 6m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/53c6634c0430487d8ad37347eea4cfd7 : SUCCESS in 8m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d2e411d57d02455bb03574ebe1a22fba : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c309baaa32d46519ecfaed38e3ac9cf : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8ce7b76e1fd4303aa91aac157066318 : SUCCESS in 2h 26m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9dc74c9874e4b909984704fc439f2e5 : SUCCESS in 2h 03m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc7db5f6df4c4a30ab9ad703ad75e3ed : SUCCESS in 56m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/711b9c0a964a46d1935e24cb881caa25 : SUCCESS in 1h 24m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d50b0567975f4781bc72c7dbfa2531e1 : SUCCESS in 35m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a2ad44ba73a467dbf534df73793234e : SUCCESS in 2h 06m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6791046e859f44fa942647a791c96a74 : SUCCESS in 2h 07m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/411a33175ba14fe699cc3f0fd82aed92 : SUCCESS in 1h 57m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4ce34b73be484e1fb0d2f32b1f880496 : SUCCESS in 2h 18m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/04c84b75b7594e8eaadf889121ab1639 : SUCCESS in 2h 06m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/86e6ef94af0a42c993f0e0186848b3aa : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dd6b22843a82119283fec1ec0869b796987b0b4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-05 14:08:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bba6171b81300c22eba3326d7043d9d784429b9","revisions":{"4bba6171b81300c22eba3326d7043d9d784429b9":{"kind":"REWORK","_number":1,"created":"2021-09-24 23:05:12.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/96/810996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/810996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/810996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/810996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/810996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/810996/1"}}},"commit":{"parents":[{"commit":"4e02b7639dfe02ecad2fd22269d68dd5ca1a9672","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e02b7639dfe02ecad2fd22269d68dd5ca1a9672"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-01-21 04:43:18.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-09-24 23:03:11.000000000","tz":-420},"subject":"Do logrotate for amphora/haproxy hourly by size","message":"Do logrotate for amphora/haproxy hourly by size\n\nMove the crontab trigger for logrotate to force hourly runs.\nThis should resolve some issues with logs filling the disk if they are\ngenerated too quickly.\n\nEstimated total space consumed with size\u003d5G rotate\u003d10 compression\u003d10x:\n size + size/compression * (rotate-1) \u003d total\n 5G + 5G/10 * 9 \u003d 9.5G\n\nChange-Id: Ied5519087dac7ad1c6c4fd453dc6ce2af4348cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bba6171b81300c22eba3326d7043d9d784429b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bba6171b81300c22eba3326d7043d9d784429b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
