)]}'
{"id":"openstack%2Foctavia~812465","triplet_id":"openstack%2Foctavia~stable%2Fxena~I70cff2adc3c2554526c52f3879d16eadc6ba4ef8","project":"openstack/octavia","branch":"stable/xena","hashtags":[],"change_id":"I70cff2adc3c2554526c52f3879d16eadc6ba4ef8","subject":"Pin pylint on stable/xena","status":"MERGED","created":"2021-10-05 09:10:51.000000000","updated":"2021-10-05 18:07:56.000000000","submitted":"2021-10-05 18:05:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812465","meta_rev_id":"b5c7125e2d2ff89dcb70ea6a0d7254c8f2d748c5","_number":812465,"virtual_id_number":812465,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-05 18:05:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-10-05 14:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-10-05 15:54:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 11:23:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 14:47:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a4d188254353bb1e15008d41cf736b2155577939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-05 09:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"428a3e88948a38381a891cb913ddbce98499d8d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 11:23:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/97e4b5b7f9114216ae161eddaefb3eda : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75d06d2b6da5453eb238200234716a70 : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dff2d36aa5a4f139a4d1e9016b34c51 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/440b3336307c4909b92e713cc33e1bfe : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20beb3ff7e29480eb38c193dd187bdb7 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/196bef5f92864f3c9d0e9fbbcf9852d3 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a00f5bbb71b4fbc8454b304705c3bc5 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e25dc93338451483417f98bdb0e4d4 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d6522e92ea44120a727cf3c66d117ca : SUCCESS in 5m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b1a1b11b6944cccaf92673e58414d8a : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cde639e61f484152838e2b163c5558a7 : SUCCESS in 10m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4db8c8bea19a414a9c40e1a3d7ddd4e2 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d85666831bca4bf1b162af18930e88ea : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d61847b2fd8465a99d0dfce65127304 : SUCCESS in 1h 56m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff304b7f2af9445bba608c0f5a42d499 : SUCCESS in 1h 36m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d1bdffb7bee34e65aaf3530dcc761079 : SUCCESS in 36m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/259e24979d77493284ac8b0b8c55152b : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbe3cebd6cd540389ce8c78014a165e0 : SUCCESS in 2h 06m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5dc20e3478584b508e44295205cd30f3 : FAILURE in 2h 09m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/561960a8a3e147ee95d672bd97ac768c : SUCCESS in 2h 03m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/975e4ba1eaf84242828f89d27af65757 : SUCCESS in 1h 40m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/abf6063806cd4b80bbaf49b99c2216c5 : SUCCESS in 1h 21m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/658aa5d25e374fe8bb464ec18cc86e46 : FAILURE in 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6476378d97091e50f371ef320d2b19bfb13d793","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-05 14:47:45.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"a7e9d90a6fa739c5304524f16e6fb840c13b6cb4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-05 15:54:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba6d844f91aab6d3d6d017f72ecbd6962fcc4441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 15:55:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45a66340908131bd6364827d84b003409495c105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 18:05:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3334d30ada0449a7a58aa9c755f0b501 : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/028b457838234a929781b7771b8f79d6 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d2224f073247b687a9ca0757c7e783 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cfd80b1137e46e68a2a259c90358fca : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaf0029f605f49d4bc619f3242a460f4 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6df48f92b94ab5a3349a458166674c : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8efce37e24b40cd865bd6592fb848e3 : SUCCESS in 5m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/155f5f9b93d34f03b3791e10dbf342ca : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bde44bf7ff364c30a9e3c86fe9b78f4a : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2e1b410e40a4fd2a1e0c17e389c6648 : SUCCESS in 2h 08m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76b561717b344c50afb5355765af0405 : SUCCESS in 1h 50m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c83926e983314957ae3e2961965856be : SUCCESS in 46m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c84d1360f514022a32916657350beb9 : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"663517f80df154627819788d95b827658d073688","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 18:05:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b5c7125e2d2ff89dcb70ea6a0d7254c8f2d748c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 18:07:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9bc70fe68341b9937aae887eb93c62 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7acd838624541678bd887330e5fb33a : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26c75ccdf26c498cff005ea635e94da83a5e46c1","revisions":{"26c75ccdf26c498cff005ea635e94da83a5e46c1":{"kind":"REWORK","_number":1,"created":"2021-10-05 09:10:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/65/812465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/812465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/812465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/812465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/812465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/812465/1"}}},"commit":{"parents":[{"commit":"b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da","subject":"Merge \"Disable conntrack for TCP flows in the amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-05 09:10:02.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-05 09:10:06.000000000","tz":120},"subject":"Pin pylint on stable/xena","message":"Pin pylint on stable/xena\n\nFix new pylint issue by pining it to the most recent release that should\nwork on the stable/xena branch.\n\nChange-Id: I70cff2adc3c2554526c52f3879d16eadc6ba4ef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26c75ccdf26c498cff005ea635e94da83a5e46c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26c75ccdf26c498cff005ea635e94da83a5e46c1"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
