)]}'
{"id":"openstack%2Foctavia~812801","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I72796b3ed1c3f180cca8c005fdbad81d9dbef37d","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-10-13 19:27:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72796b3ed1c3f180cca8c005fdbad81d9dbef37d","subject":"Fix management network selection when calculating deltas","status":"MERGED","created":"2021-10-07 06:41:56.000000000","updated":"2021-10-13 19:30:08.000000000","submitted":"2021-10-13 19:27:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812801","meta_rev_id":"aaac34238a74a1af972ce80842f22310f12c1046","_number":812801,"virtual_id_number":812801,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 19:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-10-13 09:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-10-13 17:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-13 17:41:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 08:27:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 09:14:15.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-10-13 17:41:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6aafeed68927e6a3dadd25e0c33e6aeb0420579d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-07 06:41:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dce1f37df9a6cf6259cc75e1e7425484ec70ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 08:27:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bf3737c218146ddb32b062ee36c858f : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0467492b7424ac4937d044782da87f2 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7dee20d8484439cb874467b6a1d720e : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25532e9301434becbef2d1c8be553e52 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b0c41a33d2456cb01c9a7dd6adc75f : SUCCESS in 12m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fc38c0bcc20c4055b14f41e15c4c3542 : SUCCESS in 8m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3bebfa217c714986894d3e3cede56413 : SUCCESS in 9m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fc2d46d2ecd47c2842892bb292538b1 : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/abd67c2199134711ba4fc875a91c5780 : SUCCESS in 1h 08m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fab4c5018af4e6eb12a87561fe2049b : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0d9dd70b413497d84d7b9bfe404577c : SUCCESS in 36m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8d813d960eae4bfe9a4c0e34ba4e71e6 : SUCCESS in 33m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba832e2a92dc40b6a3a5a6def7bec59f : SUCCESS in 34m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f263ecc16ce545e1a5205da43291263f : SUCCESS in 56m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27ae464af0cf43858fd488a8bb025564 : SUCCESS in 1h 31m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/53d513669c4f4f6795f8003403cec983 : RETRY_LIMIT in 8m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e56f3c3148d0477b98167cf9ddd5c3b3 : SUCCESS in 1h 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4981217b4af8f568d0a9688135f6dcd5be97a66","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-10-13 09:14:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"515175eac51314617dbb468c2e37205ba58245ff","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-13 17:41:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9d30cff2c4168537b2ba6a3a0075e13e2b05f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:42:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47f39f9536807afb5912029d9adf3d65c8825a72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:27:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4970087cb357406ea4f614242a54293f : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb6cfd2bb6374ecfa2d050749c6ff55a : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4be92246e6f043f48fa055fcdef18f8c : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e505e06cdc5469b87571cb83c31946f : SUCCESS in 12m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9c885fd0d494bba8203a58d71712dff : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f55281ddd012425fa371fbc5e35e1243 : SUCCESS in 1h 43m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0dbc0bec87bb4f9daa1f01b55b364231 : SUCCESS in 1h 34m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3da33d45a6e842df95532f31752e4955 : SUCCESS in 52m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6078d0a10f9d4a8a982dd7be61957f92 : SUCCESS in 45m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85da484382884573b516e1d9be5cd89e : SUCCESS in 44m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14be8c61e80f4b8f8983ff714387ec48 : SUCCESS in 1h 07m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8774e76175551400bed3e1b11d85774f4f138409","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:27:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aaac34238a74a1af972ce80842f22310f12c1046","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:30:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e54669a684f479f9605d9b2ff0f715e : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19ce10f9529cffaff255bbc2b3110e832965e877","revisions":{"19ce10f9529cffaff255bbc2b3110e832965e877":{"kind":"REWORK","_number":1,"created":"2021-10-07 06:41:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/01/812801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/812801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/812801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/812801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/812801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/812801/1"}}},"commit":{"parents":[{"commit":"6fd98bbd15fe4268982fdb2cf3377bba6f964382","subject":"Merge \"Fix MAX_TIMEOUT value for listener\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fd98bbd15fe4268982fdb2cf3377bba6f964382"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-24 19:41:08.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-07 06:41:27.000000000","tz":120},"subject":"Fix management network selection when calculating deltas","message":"Fix management network selection when calculating deltas\n\nProviding an AZ dict without the \u0027management_network\u0027 field to\nCalculateAmphoraDelta task broke the calculation of the interfaces to be\ndeleted. As a result, the management interface might have been removed\nby the HandleNetworkDelta task.\n\nChange-Id: I72796b3ed1c3f180cca8c005fdbad81d9dbef37d\n(cherry picked from commit b41b275b8314f79cdafe1575d84aface2b5861c5)\n(cherry picked from commit 8681637da0f99b4dda128bfd8ffd45be20de6112)\n(cherry picked from commit 60b14c69dc9c6ab6aedf6928004dae0c958eafb5)\n(cherry picked from commit 4703bebd418c11adc119aece0c436f5401a9138a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19ce10f9529cffaff255bbc2b3110e832965e877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19ce10f9529cffaff255bbc2b3110e832965e877"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
