)]}'
{"id":"openstack%2Foctavia~813149","triplet_id":"openstack%2Foctavia~master~Ib550542820163be2bbef97df7b090834a6b6dccd","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-10-11 20:20:40.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-10-11 20:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib550542820163be2bbef97df7b090834a6b6dccd","subject":"Fix docs jobs and pep8 errors with pylint 2.6.2","status":"MERGED","created":"2021-10-08 09:04:55.000000000","updated":"2021-10-11 20:22:25.000000000","submitted":"2021-10-11 20:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813149","meta_rev_id":"1bcc20782548211145ff7449971173067a68f2ce","_number":813149,"virtual_id_number":813149,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 20:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-10-11 18:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-11 16:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-10-11 18:10:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 11:15:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 16:46:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-10-11 18:10:41.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"b8d1636aabf35033861e1f77cfd01190e125dd2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-08 09:04:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abb4736955c01f80ed6baa9cbdb7a8687386ea4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 11:15:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf6cd67d76aa4412a722b323b272fc49 : SUCCESS in 14m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ad942cdf3474b8785d202d9fd14b034 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c4d2583b27483eb786077430529eb3 : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28416fa789c442fbb894078d26edce24 : SUCCESS in 15m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa18bba1985c4ee884dd411eb976d294 : SUCCESS in 13m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13bf383c7e9c40098bdcede19326c538 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd2bee5968845f78d7d58b49ac45472 : TIMED_OUT in 31m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5378230f34b54ee6b7795e140f879391 : SUCCESS in 14m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6e63d7c55c7c4e3aa248b0cc3b1b96fa : SUCCESS in 19m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/609b3e5225c544f0849ec91dc337ec5e : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b33131069fd44da39cd99b799076fd5a : SUCCESS in 17m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/045bb84d4d33439db63a9e8061d02dd4 : SUCCESS in 2h 08m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c88c44fdd59e4184b22372ac9d832cd3 : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e567fb4fb6b6469789bff1ee1691a14b : SUCCESS in 42m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a098b5d29e8347f9a2e862edfe4bba4e : SUCCESS in 47m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da340dfa67be47b1b53e931986abd8a1 : SUCCESS in 34m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/115c0b05c61f49f3b2f9a67c003b487b : SUCCESS in 1h 47m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7dbd42200cdf45c0812576bdabe48014 : POST_FAILURE in 1h 46m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c91ccf234f3c4d2d8232e16d03c07678 : FAILURE in 15m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3736f3d496374216a2b7fecc1dd60bb4 : SUCCESS in 1h 42m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1817b264d96f4fffbce6aa8948c4ca28 : SUCCESS in 1h 50m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e2125e58ae74489d972eed5caeea5bcc : FAILURE in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c18df9c786cbca411b3daa9e70252dfc38067cf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-08 12:21:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e05fa01ef7029e35c85bc2243d351a12f1e4f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 14:30:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e32745e265164b85a158dae2bcaa9b13 : SUCCESS in 7m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a51ae0f9bb204ffea16d8b2c1620a4e3 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c95135092d4c479382f9b5ce7bb25fcc : SUCCESS in 16m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd9f735d421948ee99c45dcfd14dca8e : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/219626b862ca48719643ac31a4259b9d : SUCCESS in 12m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f62d22b950384f238dbd0e9862ae36f8 : SUCCESS in 12m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b973b52b39d44e13b9a0584ca877bd95 : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7261de1fc684a3ba2e38273b214108a : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66973e54d4be4f4a8f75568f3f7a9ff4 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e8458f7b2f64fb8a977f0c757b2b30b : SUCCESS in 7m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dfcbfe6f11b14d9590b99afcd2d95aa7 : SUCCESS in 13m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0515524e623444119553bcf91ddff7cd : SUCCESS in 17m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/22392b79a31d41c7ae8715b2c258a20e : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3eb0464c7cda4329981a0a30da17c8cc : SUCCESS in 19m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/572db6e9d1984ae58b7b1c293b4a3e65 : FAILURE in 16m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21da110ed5fe4eb99505717de982ede8 : SUCCESS in 1h 53m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ffc955e724e94980b8667e0a3e43b0fc : SUCCESS in 40m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d789ebafa6245ddb4801f5bbfaf4594 : SUCCESS in 51m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/818c0b1cc2a04b2f8ebfac1674f76bc2 : SUCCESS in 45m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d115cea70c24a4dbf6766996937863e : SUCCESS in 1h 31m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/906b7796cd034ca09e2561e84cdb4dcf : SUCCESS in 1h 43m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d353abce0885444cb3fa3710de84bb96 : SUCCESS in 1h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a8a77a1ecce4f1c9242033251aefc73 : SUCCESS in 2h 05m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4fa06bf11320482e840dd6e3059bb610 : FAILURE in 2h 03m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/528388b6330c4045975833d7e385b62a : FAILURE in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4fe408197d958ac9e338bf86790830eab4353556","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-08 15:56:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8c83b399f131bdaf99a49af167506299d7bcf774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:33:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9b0a9474c494ea296345c9735fa342e : SUCCESS in 6m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e89ec7beeb1544dca6cb70b405f5b6f2 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6c11dececc84ac5a244f5f6164ab08f : SUCCESS in 19m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c96341f8817846ad8cba79ae78037cef : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90021e48e1064cb0bbd429044ac52561 : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfb967f97b324897afd2f01519e1cd87 : SUCCESS in 15m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6657c64264b346c1b5e1a7ff61c1b205 : SUCCESS in 12m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b3b9333bf1243e392d32069a173459d : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21483c0d6f3b4b8a9208740254882123 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8d39b21082f44a7bf817db4533ad5b8 : SUCCESS in 6m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a1cca64fd71b4fbd8913224441f72247 : SUCCESS in 14m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ec4818baf32c4386a658650dcdddf480 : SUCCESS in 18m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fb63b9a85f1b47f6a1f49c8c686491cb : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/14cbbe0ba87e4a0981f47768e5139bdf : SUCCESS in 18m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/034125c4d2e94e9fa373f69293827468 : SUCCESS in 2h 18m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/820ac71a2d2a4183a07efadc4d2b50ce : SUCCESS in 2h 14m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f843cd0e9c04bb99ac9b3932f78df99 : SUCCESS in 38m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfee4dbe79534439b7097003873bfc3b : SUCCESS in 1h 06m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a23e5ed340c45fd8c6521f153e73531 : SUCCESS in 37m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b025dc4c1dc8483ab5f430a8921c5bb6 : SUCCESS in 1h 30m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5fd85c8d8a1e4c30a5eec10b80e0771e : SUCCESS in 2h 05m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d1cf736525844d0db70577353b638c3a : SUCCESS in 2h 22m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8fa2b15623734acf8aead6ae0f42f7a5 : FAILURE in 1h 44m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8d7e19433a3a47d7bf1e1819639010d4 : FAILURE in 1h 39m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/15954326f04b4722b9358729c1caa65f : FAILURE in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c2584a0a076e32152ba2ed8e361535a21d095b28","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-11 16:46:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"59e2b5bd4b0df60da57984f96ceb0aba42667d95","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-10-11 18:10:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f81c44e4aec08a2d99ff21325fd9ac29f6655312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 18:12:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d6d224143c85e89d21772efa1d5d500a6c494e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 20:20:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf6644f724e24c74b0a19621dd845711 : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/336b0c18ce584c99abd6507a795c19a8 : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64fb0fdefe414793b05731a66257fdb9 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c498105c4ab498fb3ba578efa1dfc35 : SUCCESS in 13m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4abaccbce09246a8acda32dc38151be5 : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21549d25e8904a9a8aba45ce24c56f9c : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ea474debbd4ea6b18b562ed2b8365b : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a21b6e93099d46aa8556141a4f5c9387 : SUCCESS in 5m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3401ec6f9801423da38dc19af375270a : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b58c16b0756841d69ce8d09ee706fa5e : SUCCESS in 15m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af0e1f70d1e040f1b8c38d6a54a1fa10 : SUCCESS in 2h 06m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bff535a378304644ab53c9ec7b8edb19 : SUCCESS in 1h 58m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e60e30e2a6fd4ebf90406b48992ea5c3 : SUCCESS in 35m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2207a3cdb0d4677b9b7df80d2d32c85 : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":2},{"id":"ca138a0394afa3e3c5a9436f0023760fd3a93f01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 20:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1bcc20782548211145ff7449971173067a68f2ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 20:22:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/63b52b8454924cad8c6645da388a285a : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ebdd1787f954c1789967419789b1bbd : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c30a09a59f064a3e9619cb941beb46ae : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ac5aa7cbec259d79b63e8777ee26bd8391c33b5","revisions":{"ae005b2387a84da303c3ef970cec034052faf44b":{"kind":"REWORK","_number":1,"created":"2021-10-08 09:04:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/813149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/813149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/813149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/813149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/813149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/813149/1"}}},"commit":{"parents":[{"commit":"b41b275b8314f79cdafe1575d84aface2b5861c5","subject":"Fix management network selection when calculating deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b41b275b8314f79cdafe1575d84aface2b5861c5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-07 16:33:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-08 09:04:07.000000000","tz":120},"subject":"Fix pep8 errors with pylint 2.6.2","message":"Fix pep8 errors with pylint 2.6.2\n\nRemoved some comments that disabled consider-using-with in pylint,\nmost of the flagged code is now considered as false positive.\n\nChange-Id: Ib550542820163be2bbef97df7b090834a6b6dccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae005b2387a84da303c3ef970cec034052faf44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae005b2387a84da303c3ef970cec034052faf44b"}]},"branch":"refs/heads/master"},"8ac5aa7cbec259d79b63e8777ee26bd8391c33b5":{"kind":"REWORK","_number":2,"created":"2021-10-08 12:21:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/813149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/813149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/813149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/813149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/813149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/813149/2"}}},"commit":{"parents":[{"commit":"b41b275b8314f79cdafe1575d84aface2b5861c5","subject":"Fix management network selection when calculating deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b41b275b8314f79cdafe1575d84aface2b5861c5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-07 16:33:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-08 12:17:03.000000000","tz":120},"subject":"Fix docs jobs and pep8 errors with pylint 2.6.2","message":"Fix docs jobs and pep8 errors with pylint 2.6.2\n\nRemove test-requirements.txt from docs dependencies, add hacking in\ndoc/requirements. It should reduce the duration of the docs job.\n\nRemoved some comments that disabled consider-using-with in pylint,\nmost of the flagged code is now considered as false positive.\n\nChange-Id: Ib550542820163be2bbef97df7b090834a6b6dccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ac5aa7cbec259d79b63e8777ee26bd8391c33b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ac5aa7cbec259d79b63e8777ee26bd8391c33b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
