)]}'
{"id":"openstack%2Foctavia~813743","triplet_id":"openstack%2Foctavia~master~I658a4097e120d4f6a49fa1a7945a8d999709d1bd","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-12-17 16:00:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I658a4097e120d4f6a49fa1a7945a8d999709d1bd","subject":"Switch centos8 jobs to centos-8-stream","status":"MERGED","created":"2021-10-13 05:54:52.000000000","updated":"2021-12-17 16:08:33.000000000","submitted":"2021-12-17 16:00:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813743","meta_rev_id":"d03cf340299a2c3412bf10752dd8439e139bbfa8","_number":813743,"virtual_id_number":813743,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-17 16:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-12-17 12:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-12-16 12:02:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2021-12-09 16:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-12-17 12:54:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 05:55:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 17:22:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2021-12-09 16:01:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-12-16 12:02:08.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2021-12-17 12:54:44.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"b9d934e68ff1b74dd3736be27748b821e6af3cd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-13 05:54:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a36290edeea42877fdd1ec13fc156968ce2b2f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 05:55:43.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job publish-openstack-octavia-amphora-image-centos-8-stream","accounts_in_message":[],"_revision_number":1},{"id":"0813a5d135b7eb768ea746e28b08b03094f1193e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-13 06:01:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0400a5595dbba9480656aa02c1c41184d9ec461b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:05:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27db39db12bc4324ae6164a0877df1ac : SUCCESS in 22m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/798d7ac9c7674485ae5ac9eb924a31e6 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbc6d1677114813a910a05f83836892 : SUCCESS in 16m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb7472e344ba4b8cbd07d6eae330b246 : SUCCESS in 16m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7242228dd82349469de0ba7633be153f : SUCCESS in 16m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/283b12f0238e4e26b0b080b082180f99 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b783bd88c74482eaae44b7f16309868 : SUCCESS in 14m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2558869f3b384414893fb6d86cdab0d5 : SUCCESS in 20m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4363c17a9ed942788d3e3da1824f40ad : SUCCESS in 19m 47s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e8410be51b64590a2e67765eff359a6 : TIMED_OUT in 2h 35m 45s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/2025f7fcbad644aead06da758c3ae46c : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c981b6d66b10dc237552494c416b5de2a4d52437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-13 11:50:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"479960ec34ae1fcfa90047da8653d53d527be3d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 14:25:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7449945d17f3440b99d764a1b5a1405b : SUCCESS in 19m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02ad3fe8db2042e39b825643b461cae8 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c5fddba09b4ddc897d2dd93d327564 : SUCCESS in 13m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59d54399952c4cb9a97875a03eaa3ba5 : SUCCESS in 17m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/716a26b8a32f4d7a9ec9d54512e671d9 : SUCCESS in 13m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/779e39ba4b154a54a782ce9323c4f4eb : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93042863c4b84520a3495f7b7294ebf9 : SUCCESS in 12m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d3bcb1d30924b03ba77b20e879f5a85 : SUCCESS in 15m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/19598998d8794b2884c55407ce8e641e : SUCCESS in 18m 49s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/16617c4e5ed049e6a5747ae6b0779f26 : SUCCESS in 2h 32m 30s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ef99882700e4ff79000988a463cb1a1 : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":3},{"id":"a5d67176a4e3e1217fffc382057e3143f02a827e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 04:57:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbb37c1db41bfc7e9195d86e1737a594f8bbb045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 07:40:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55063c5fc7b94ce8a77f770418b3d632 : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b54458e20ba4e17ba29963fb31f14c5 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f6a60d111846cdbf66b6bcb23c7d6b : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3657c3de33b4611913a903c6de2020a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c66c0a6e5a84485b70de54f8d776298 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22eb815039174e0ebbbe185c0ab27955 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e14566baf984d8785f21a62a3e23e46 : SUCCESS in 12m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2617a4c5cf6047c887ca9a07945a7b05 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4d5e0f780f2443699b49df68d681fe94 : SUCCESS in 9m 20s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9aba068a4e974af586934f81a42155a3 : TIMED_OUT in 2h 34m 14s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7ce60633e324d17bc24d463c6569722 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":3},{"id":"2502a2e9492d0f17a805a24144aefca4f2b2bf84","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 07:53:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d13444ab81bdc2a3084d4c8000cb85ba6201c55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 10:04:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/425e3079aee840a4aac981394245da96 : SUCCESS in 12m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25748642549a4a7eb86fde43a0b322c5 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57ec09f188c4e1981da995b73ac508a : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7c85ab57ef4b789a2f325375a523bb : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e04f83cf52243ae87a983aab5f45f20 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d66e2ed80954bb49d5c17a108f4323e : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77526f518b7435ba85ace2d82c10be5 : SUCCESS in 13m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/04400e2d2fac4676a703849a187a76aa : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/256b4a2cfcf443c8af398055a23db5f5 : SUCCESS in 11m 16s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cda4b0b0f444a258353ab99bedc66f8 : SUCCESS in 1h 58m 43s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ae8e1e3455643948c2a0c13d9de8ea0 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":3},{"id":"822465cc8b761300292d489b2c8b183877544bfc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-08 09:21:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc31f2e861d8e855506cefe8affaff79d179fa7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 12:00:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/703f9a9c4b1249778a4f0609e84169d3 : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ade887ec3c6f4ba78f841f8ce9a0aa73 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e3ec00f76c4ad2a28ce3527bf6d6bb : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3561cc7e4f174adbbe3f932da775c661 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7af7271d7284b949d1c9e7a8bec1d6d : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25138bd2f009432a890d2104bc12305f : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b5f060c0a24316862f04f57da93bf3 : SUCCESS in 12m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6bd16b870eaf467ca9b96afb7c9613c0 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b97d6fc1d87d47c6bf141cb0875c9abf : SUCCESS in 9m 23s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7a5afefe54e42dc8128fe975bb6924f : SUCCESS in 2h 35m 58s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa89d486a97f4323a500e62829295618 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":3},{"id":"eba8d81e6b7333ce0a9e90654bb0a9411857acca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-16 07:58:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b555db56faffa2239f8e4f901469d17eaeac1d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 09:39:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8653686e6b64fe5abd6e12c268b4f32 : SUCCESS in 11m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50c66d9f242c4f5b944695fd0b924c54 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdf03bcfa42142babde782181379c723 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a4ef38e594643e6b0785637921456ef : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6de21e9a1c4246dc901a43b14f0dd6b7 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b98ab55782dc4a82aa558ccd21c5cdc5 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6762e6960d5949ab9a51c714fdf0a8fc : SUCCESS in 11m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7f8944128e7b418f8cc9c16ab162da4f : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/89e818467a804738a5df8ed1ebab6be6 : SUCCESS in 9m 14s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ab622c316fa428792b6a09517c574ef : SUCCESS in 1h 33m 12s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ca35fde2f304e8cb151359c81eb4501 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":4},{"id":"f6b18b17eaf6f9579725b680b4cb2528fb40f21e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-16 11:21:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d4298b1da9ab6548d3adda1492eebdbb18a0312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-16 11:23:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bb8fc0c7fa2177efc5658e5eb229d0114dba6032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 14:00:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f6159815af745b587896cce483674ac : SUCCESS in 31m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bce48d343104fc28f874347b1c82741 : TIMED_OUT in 41m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e700cedb42484b668f1f42eeb0e9d39d : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/720ac483031747aab18f18af14065714 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/800abb47d5a840b2a830e179e08f3680 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e8437a150db4e2d886214d5d72f4b86 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1815afb465f04140bfd114df43978ee2 : SUCCESS in 13m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/89ad749d32274dafaf5e6ac602dcc667 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0acbdce8d9df4d8a9f45cb5732dc199c : TIMED_OUT in 31m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23c15497f8bb487b8a26eb74b0e48821 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/509df6bd3dd7436d8e7fd6417c4827e2 : TIMED_OUT in 31m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d33f792723b646c98dfb1e9af7a422d3 : SUCCESS in 2h 16m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7597c16beb2044659141daaa8f069123 : SUCCESS in 1h 17m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c3f39a2cfb445e4801ae4ae86eb72cd : SUCCESS in 46m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f76581d12ae64375bccbcb5dc73e2362 : SUCCESS in 46m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2fb025ae3f040369218ef82962e5c7c : SUCCESS in 31m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/706f7a195044464197d98ca8ba2dc9c0 : SUCCESS in 1h 28m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/888a4735d64246c19a64ca7a6460a3cd : SUCCESS in 1h 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2dea4c8779c9462f923782fd7f64416b : FAILURE in 1h 42m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7f8dd2e201854c0980a6402a5e0ba0dc : FAILURE in 2h 32m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c70d6f9498e64b98ae5095b1ee47a58e : SUCCESS in 1h 22m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3919635b484a4c6a87aa1aa15f4be4df : SUCCESS in 1h 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f8d91f75bf86ebee780c2fe62e18ceef7cf8bff3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-17 06:37:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e54fb15459ba647e3ca5561da794b60e119ccb24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 09:08:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd88da77e91e43d1b6f0ec459cbd0d3b : SUCCESS in 46m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc08972a38df4d1a9f4fd3d7eb73f310 : TIMED_OUT in 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/687015c697a8456b8f9c20c703dc958c : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0746d524b0d04521a7c5e2a039b010ab : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3c10db8b66247b6ae4fc28f4f863d5f : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/939906dd9b1e44e3802ba91327c87fa6 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d5e1c202404e40b2f110f63087081e : SUCCESS in 12m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1e73ad3b155b42098ea4e8da658347f5 : SUCCESS in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7513b2a2a486437c8927b24c50bc6728 : TIMED_OUT in 31m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fbcda2a699384374895133a360da8716 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d40a650382ca48a7aac4fcd7e42813f9 : TIMED_OUT in 31m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2eb1bc6e8a544342be5c6b5f3a87f92d : SUCCESS in 2h 00m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb09a7d1c58e4447a06528390f8b2a5d : SUCCESS in 2h 28m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17b4b62da74e4358a7f73a83ae57c178 : SUCCESS in 41m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7456513239704517977426610bed89c4 : SUCCESS in 58m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c2f273632904031b2d3a31cbc9a4717 : SUCCESS in 35m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3c9ed2363a874fb6aa2283f34c33542b : SUCCESS in 2h 15m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be49651b6d19403a91ff529df320cecb : SUCCESS in 2h 27m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f609037d9b1b4d6c9c0a65793981ca9b : FAILURE in 1h 51m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a0eeb63f4cca43c993acd7014ea12a1f : FAILURE in 2h 15m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/841234ccef5d480284e0c466576dc017 : SUCCESS in 1h 24m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b1dfad4f7f7421da981af93877a6e08 : SUCCESS in 2h 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"29315ca947d519dc3b27ecd475708ce346673558","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-22 07:07:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2c0a4a8988625da257227b457a3a74d6ffa82139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 09:47:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25fce7d52a7941c88acb5221b731f081 : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6374fa8107d47fb9479f800741d2234 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab9e30d1c4a425ea20ea9dc717de1ce : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7404ceabaf6b4a7984c4a6f303ba3788 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/578fc651c2bd4123b85900784bf85c41 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86807ce5e585437eac9b09c533ec4bbd : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee52827a86840539cc7ac9e5c4bf6ef : SUCCESS in 11m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/14592320941a49979037ffd507505f05 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7adc68585a724dada5b3d95744903b3d : SUCCESS in 10m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fa457d67f43049c494a75818c82ad16c : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/689834639996409a8b7c26255d42849b : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/981026c782be46df90340c19ad5c16a6 : SUCCESS in 2h 18m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa930e29ea4b4a2782cb81b7f4da832e : SUCCESS in 1h 54m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d1f4987b0a2846b4b76b18761b74fe90 : SUCCESS in 37m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1964533fee02456381eeaaccd33b0cfd : SUCCESS in 1h 12m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ccb9441e6ac345728cd60a08739721b5 : SUCCESS in 43m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0a29267343e14b16938ab4d983a38636 : SUCCESS in 2h 31m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56674410ad264114800c11a4af4508a1 : SUCCESS in 1h 53m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/16fde4ddd27044b4bf3eca41dc716f96 : SUCCESS in 2h 16m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/04c489441af94aada780151009cff358 : TIMED_OUT in 2h 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1095406f59254064bc0b114b4b416458 : SUCCESS in 2h 01m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bbd992cafde47bca0367bdf46aa3775 : TIMED_OUT in 2h 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3ddb91ba02916a5cdb3a14bc89b4f2e5bc666e3e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-07 17:22:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6298015912932570355772a98b51b85e8f75c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 20:13:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/853e24061eae40d3955f187678ce6d37 : SUCCESS in 11m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/284ad421c81f4fe3863802e4dc2b3c05 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08efa3330a94e66b67cea7524fabc1d : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f060942cedef453ca671b5686cd89464 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5cdbcd137874279a982a9ef0274c57b : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e25d43aef97407292670e43d5e1ae76 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06c210cb4df4a23a56dc3e264fdcd9f : SUCCESS in 14m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a2388c3b1db9445489293808d7942f81 : SUCCESS in 7m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e6d3db287d7e48f69147259e3853b759 : SUCCESS in 10m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c764bdff4d074ff88e09372948e4393c : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5914c21d7db1440aa5cfef0b144f9812 : SUCCESS in 10m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2392096167e94e34928db87042e17650 : SUCCESS in 2h 16m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d2d1e61d4c941db9a93f0ab6ed1eb6b : SUCCESS in 1h 24m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8b3e4a3273c748aab1f1a478797f3976 : SUCCESS in 37m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/575531e268924c468599c345addad556 : SUCCESS in 48m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1a49497c8c14385bfd56781350bcb18 : SUCCESS in 35m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/25e0354dcf8f4819a8281c474f74a2ef : FAILURE in 1h 55m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05125295713541c082a45cd5d0921b9d : SUCCESS in 1h 46m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/629be0cfda9140f584c1ca386c675f46 : SUCCESS in 1h 29m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/60eb4ac85981435bac1ff6c8f8ce809f : SUCCESS in 1h 46m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5af603a7b38c4680a7735a48cae64f7c : SUCCESS in 1h 23m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0ed8cc054eb4e0daa7a7c8d4300cbc2 : SUCCESS in 2h 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"637e2bbab2c9cf5b2bf4a7dda8a2874643daf1c2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-09 16:01:01.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1c48aa05f221700abae5b2f170f2a4688d9ff139","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-16 12:02:08.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3dee2d1d6d17c326d641a8df7bbc25d0204e8363","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-12-17 12:54:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ff0e7982545e3c0a6772bc0863b4147e67476468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 12:55:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dec8444276e150b6d1b67579395497364c23ed83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 16:00:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74829231da34567b284a969e213202b : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ba4537f0fc470ba13c02d76abe785c : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efef4dc617e64776b9b6b8403fb341dd : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f811bcd5fdd8418d8f8efc64062f7d92 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4864953cecd43a787bc32173adfd2df : SUCCESS in 11m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8f57a1f56ce4da5a5de293a151bd643 : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11eb5ff2b9624811a91a30a92405b02b : SUCCESS in 9m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/880327e397da4bf9acc5b13fded7d0e2 : SUCCESS in 2h 33m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ee6e4e960fc45dcb755d73ed4fc0255 : SUCCESS in 1h 23m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25ee2beeb8d746c582a83f9a9b3ce864 : SUCCESS in 34m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ef9b2cb1b184ad8809f3747717cc767 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":5},{"id":"e7041c2481d95b91e52ff6c9a9fe4f72ba10ff2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 16:00:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d03cf340299a2c3412bf10752dd8439e139bbfa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 16:08:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266d810045ac4ff58bf08b3f7942f3ab : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"76617f883e535a77008aac80d5350d0f0c889f8d","revisions":{"2dfaba25c6d1be27e2b29b42913b719ac55b2de5":{"kind":"REWORK","_number":1,"created":"2021-10-13 05:54:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/813743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/813743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/813743/1"}}},"commit":{"parents":[{"commit":"0b84631948bd3c5eb25a0921d5fbad657036304f","subject":"Merge \"Increase TCP buffer maximum and MTU discovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b84631948bd3c5eb25a0921d5fbad657036304f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 05:52:53.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 05:54:43.000000000","tz":120},"subject":"WIP Switch centos8 jobs to centos8stream","message":"WIP Switch centos8 jobs to centos8stream\n\nDepends-On: https://review.opendev.org/698450\nChange-Id: I658a4097e120d4f6a49fa1a7945a8d999709d1bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dfaba25c6d1be27e2b29b42913b719ac55b2de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dfaba25c6d1be27e2b29b42913b719ac55b2de5"}]},"branch":"refs/heads/master"},"d174776ff74167dacd113dfce8d92c8fe8b62315":{"kind":"REWORK","_number":2,"created":"2021-10-13 06:01:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/813743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/813743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/813743/2"}}},"commit":{"parents":[{"commit":"0b84631948bd3c5eb25a0921d5fbad657036304f","subject":"Merge \"Increase TCP buffer maximum and MTU discovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b84631948bd3c5eb25a0921d5fbad657036304f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 05:52:53.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 06:00:57.000000000","tz":120},"subject":"WIP Switch centos8 jobs to centos-8-stream","message":"WIP Switch centos8 jobs to centos-8-stream\n\nDepends-On: https://review.opendev.org/698450\nChange-Id: I658a4097e120d4f6a49fa1a7945a8d999709d1bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d174776ff74167dacd113dfce8d92c8fe8b62315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d174776ff74167dacd113dfce8d92c8fe8b62315"}]},"branch":"refs/heads/master"},"edf62d061cb931d876315cc43149071def35ee1f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-13 11:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/813743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/813743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/813743/3"}}},"commit":{"parents":[{"commit":"2043788473a8cd6707bfc9aaedbc2065688f0db0","subject":"Workaround to fix amphora-agent on centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2043788473a8cd6707bfc9aaedbc2065688f0db0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 05:52:53.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 11:50:00.000000000","tz":120},"subject":"WIP Switch centos8 jobs to centos-8-stream","message":"WIP Switch centos8 jobs to centos-8-stream\n\nDepends-On: https://review.opendev.org/698450\nChange-Id: I658a4097e120d4f6a49fa1a7945a8d999709d1bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edf62d061cb931d876315cc43149071def35ee1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edf62d061cb931d876315cc43149071def35ee1f"}]},"branch":"refs/heads/master"},"7cee511cdeb35cbc121aa1d6cf3ab85ce0fcb90d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-11-16 07:58:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/813743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/813743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/813743/4"}}},"commit":{"parents":[{"commit":"74f3242c8e29754d9981325e64440cdf87edbfea","subject":"Merge \"Workaround to fix amphora-agent on centos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74f3242c8e29754d9981325e64440cdf87edbfea"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 05:52:53.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-16 07:58:43.000000000","tz":0},"subject":"WIP Switch centos8 jobs to centos-8-stream","message":"WIP Switch centos8 jobs to centos-8-stream\n\nDepends-On: https://review.opendev.org/698450\nChange-Id: I658a4097e120d4f6a49fa1a7945a8d999709d1bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7cee511cdeb35cbc121aa1d6cf3ab85ce0fcb90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7cee511cdeb35cbc121aa1d6cf3ab85ce0fcb90d"}]},"branch":"refs/heads/master","description":"Rebase"},"76617f883e535a77008aac80d5350d0f0c889f8d":{"kind":"REWORK","_number":5,"created":"2021-11-16 11:23:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/43/813743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/813743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/813743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/813743/5"}}},"commit":{"parents":[{"commit":"74f3242c8e29754d9981325e64440cdf87edbfea","subject":"Merge \"Workaround to fix amphora-agent on centos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74f3242c8e29754d9981325e64440cdf87edbfea"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-13 05:52:53.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-16 11:22:33.000000000","tz":60},"subject":"Switch centos8 jobs to centos-8-stream","message":"Switch centos8 jobs to centos-8-stream\n\nChange-Id: I658a4097e120d4f6a49fa1a7945a8d999709d1bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
