)]}'
{"id":"openstack%2Foctavia~815585","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I435c40a5493828bce24c2850182f0ddc8e5bdd1f","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-10-27 12:40:49.000000000","reason":"Stanislav Dmitriev replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-11-08 21:45:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I435c40a5493828bce24c2850182f0ddc8e5bdd1f","subject":"Fix failover of az-specific loadbalancers","status":"MERGED","created":"2021-10-27 06:22:07.000000000","updated":"2021-11-08 21:48:34.000000000","submitted":"2021-11-08 21:45:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815585","meta_rev_id":"2a9dadc87014403201e371b1c0d2835f066f9bf5","_number":815585,"virtual_id_number":815585,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-08 21:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"recommended":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"all":[{"value":1,"date":"2021-10-27 12:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":2,"date":"2021-11-08 10:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-10-27 16:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":1,"date":"2021-11-08 10:52:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-27 06:22:07.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"state":"REVIEWER"},{"updated":"2021-10-27 08:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 16:09:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-11-08 10:52:07.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"27ee05913a7864514d8b6795e82116c1296f5a57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-27 06:22:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5021898e14bf54583c14c7ece209d1f93baebaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 08:35:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e033d1a7d0549a49c7f4ebd97241c5e : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/925d10acc67841c49070db743f65b889 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f32211205be400d96aa6726569f4f6e : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c810824b2974847b43b10c118e673e4 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68406dbad110438b8ac22442b18e55b3 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4824e69f28244304a2da41f62126c5f3 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f1d8fa7afa4a9493355319a33076a6 : SUCCESS in 14m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ec47aac0a6a743ecbcd2ec377cefbdc2 : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b77a510c8bc34e34ad252689e75ff041 : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/539ff415b37c414aaef1a99384e65424 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d65adec7b2044b9fbd1d672a264808fa : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dda5f863571d478795707daf77a43711 : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7c7900319904fdc9d7b86d693adcea5 : SUCCESS in 1h 48m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64bd79e2dcb942f3a4d8676b8c4157c0 : SUCCESS in 39m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a9827efcf1cd4995bc377cc535a62073 : SUCCESS in 42m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/978c7c53b3584964a88d74b49dfa585c : SUCCESS in 34m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/898d2767b822448cbccc1c050aa057a3 : SUCCESS in 1h 02m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dc95445287f84ff6a6aceeca58604e48 : SUCCESS in 2h 02m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/baca39d4e0284adcbe5637f85e0bb2fb : FAILURE in 40m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1061ba5c88f3447cb9caf96fa12e7686 : SUCCESS in 2h 06m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba7d38289cc64c3fa165bb2b0e0a57db : SUCCESS in 1h 42m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c623fea9c52a449c8ad5587f544bea40 : FAILURE in 2h 10m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6438ad70a354436ab7520bdc0df3f1fe : FAILURE in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2834a77efdf10334a359f377c9403eda3b531bf2","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-10-27 12:40:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2010bd246070ddaf8f9162a6d5b615ddf1fefddc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-27 16:09:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65f473af1f094af472e4f86dc9110e2829ff1eaf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-11-08 10:52:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4304deb380f2738ec33655875c208cf8f899afc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 10:53:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6ea11519ccb67f402cd1af770ebc88fc5f3945e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 19:26:23.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e93179d1472c106c32e6ee00953a2b89eaa3f3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:45:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db8aee4212c3493dbcfe5dde4615dbee : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4a7706080d9475aaad574747dfb15a8 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86a2d393f092458d8bcb1102e77827be : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdfb42e9c09c40dead9b65bbbc760e77 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a31a0b0d2cc417286b317f00f369576 : SUCCESS in 13m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f4d01e0083284fcc96124f11a9b38532 : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f42763b11d9c4ffab6ac025eedd23c5b : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/91edb4a5d92c4e8c81f951e696179fd0 : SUCCESS in 1h 45m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0899bb9942ec4ebd957fdb13e688213b : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47088880d8e341fca3eef9c3bdfb3e50 : SUCCESS in 42m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e876796030d4795adc575ae607574b6 : SUCCESS in 36m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2c7b57edecd485db92143a50f5f3c04 : SUCCESS in 54m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2750ae3cea24cd28a27112afec97d47 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":1},{"id":"64cf6cfaaf586090592077349e278814e450c783","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:45:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a9dadc87014403201e371b1c0d2835f066f9bf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:48:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62693f1ed29243f7b0dc0020925a2a82 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77a87a7f335464180933cd1859cd34998d7f2e75","revisions":{"77a87a7f335464180933cd1859cd34998d7f2e75":{"kind":"REWORK","_number":1,"created":"2021-10-27 06:22:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/815585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/815585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/815585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/815585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/815585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/815585/1"}}},"commit":{"parents":[{"commit":"60b14c69dc9c6ab6aedf6928004dae0c958eafb5","subject":"Fix management network selection when calculating deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60b14c69dc9c6ab6aedf6928004dae0c958eafb5"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:24:06.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-27 06:20:33.000000000","tz":120},"subject":"Fix failover of az-specific loadbalancers","message":"Fix failover of az-specific loadbalancers\n\nThere is a bug exists in CalculateAmphoraDelta class\nIf there is no such a key as constants.MANAGEMENT_NETWORK\nin availability_zone dictionary, method get will return None\nIt means that management_nets list will become [None], not\nCONF.controller_worker.amp_boot_network_list as it supposes to be\nIt breaks the following failover flow\n\nStory: 2008669\nTask: 41957\nChange-Id: I435c40a5493828bce24c2850182f0ddc8e5bdd1f\n(cherry picked from commit 81134bdfec79c8019c26e7846d31a2b2952965a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77a87a7f335464180933cd1859cd34998d7f2e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77a87a7f335464180933cd1859cd34998d7f2e75"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
