)]}'
{"id":"openstack%2Foctavia~815586","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I435c40a5493828bce24c2850182f0ddc8e5bdd1f","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-10-27 12:40:54.000000000","reason":"Stanislav Dmitriev replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-11-08 14:07:20.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I435c40a5493828bce24c2850182f0ddc8e5bdd1f","subject":"Fix failover of az-specific loadbalancers","status":"MERGED","created":"2021-10-27 06:22:55.000000000","updated":"2021-11-08 21:48:35.000000000","submitted":"2021-11-08 21:46:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815586","meta_rev_id":"dd70aa4575e4792b986b48c8cfa209deae68f897","_number":815586,"virtual_id_number":815586,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-08 21:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"recommended":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"all":[{"value":1,"date":"2021-10-27 12:40:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":2,"date":"2021-11-08 10:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-27 12:51:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-10-27 16:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":1,"date":"2021-11-08 10:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-27 06:22:55.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"state":"REVIEWER"},{"updated":"2021-10-27 08:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 12:51:13.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-10-27 16:09:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-11-08 10:51:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"9a3a30860288456e112e275bfabd6fbbfaf503c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-27 06:22:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56532b1d041e4d52d8483ff0464d4c3dbbb81389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 08:41:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7531e1a093bd456dabe1ef98fb75e708 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9833f5687ed0427da5e788eabbd0e318 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/253660caacc4416aaa1b083d275404e9 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/322d43d25c32462896ed26e72e3255ad : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaad03f6b8944e21bcf0f0fd5775e6b5 : SUCCESS in 15m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6127d6f169954eeaa14caea819bf939f : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d39750dce70b4463983884dcdd69baa0 : SUCCESS in 8m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ec7fbed487441be8b738a377f5d03d6 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ebb7d04349074ec4885f90f206768352 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a24191f38de842a18b048d426eff719f : SUCCESS in 1h 42m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7d23f1a53814befbc404f29dabc173b : SUCCESS in 1h 56m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1888c11546b1450898b92b9601e1c9ec : SUCCESS in 44m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be23afb29db842188e74274c64aed73e : SUCCESS in 52m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff34bcc9a510485c8a9cd2028c830969 : SUCCESS in 42m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9162bf6493f54e2b9a689147a4157d4e : SUCCESS in 1h 12m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/694e306ab6b04109a0e79729040621eb : SUCCESS in 2h 15m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05fd3029861143beb9356c7d372800ec : FAILURE in 54m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/62afba26733e4889933349257f9a3a81 : SUCCESS in 2h 00m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e2989fc8a2b4e2489a67775a2ca2861 : FAILURE in 1h 56m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d59b674ba7174de8987f972330ff8071 : FAILURE in 1h 41m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/026809f1c084407f8d08908e3cffda4b : FAILURE in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"083e19fcdd72343f55c29dbea8954eddfa745454","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-10-27 12:40:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b6a7a810fc3b99c5741a39a3ed5720f82d5675f7","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-27 12:51:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c47533d455b6549184b7e64aabd3df8e17b3ef6c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-27 16:09:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"633e1947cd3a486e14a667e4cd729155b65a9abd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-11-08 10:51:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94edd4ebce163920d6224f832d71e186a519dbcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 10:52:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4033fbe3a72dc232f5e676c073b1758148a48da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 14:01:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8775f787ad824fdfbbf5bf73d8b4237c : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef4a57c2988f4960981c0b17694b319f : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/346f645897594d2aa3cf96ee8fd96be3 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ecbcf3b41c44cd3bf8309ef5364ecb9 : SUCCESS in 11m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c79d8f564154eeb82afdd110d244cab : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f5b72ae827d41d69218c6adcf018cb5 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f62a217f4414535b61ce209fb7b29f2 : RETRY in 1h 17m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e622ad2892b4ba2b13b5f19093ce4e4 : RETRY in 1h 19m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c632596cb3aa406486560aa786d34d8b : SUCCESS in 50m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8591f28c88a24130816861a3acde1c02 : SUCCESS in 33m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ccf6ea7a38b44409f38aaada09a1f84 : SUCCESS in 34m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e38aa4afcf8d4a89bfb114efd66871c7 : RETRY in 1h 19m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e63bd89cf9e441e903f03db069e7c2dd6e4ee509","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-08 14:07:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7007ac062431f910fc86edf003725931107ed75f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:04:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b2a8f675821464e9fb7a90cbe9f815a : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc5089e132c4d87bd347d422bcf9288 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf7207da1624c2e81fb9747a1a868ec : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8fcfb3b87b64417a33a70dd459ae20e : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489e9d4586f74fa68172eb14b8d6f6a2 : SUCCESS in 10m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f957413ea28f4141b6b4e9eb679364b1 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5c5b97f62e424bf98b972cd030baa45f : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d19a18c45bb54f7387aa561063d17285 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c87b107fa98444dfb8deb92cb046ca6f : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/caf40db297f24e9f8a6b2422409056ff : SUCCESS in 1h 32m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32e6525255e44e1a97524e8e9cdc8a6d : SUCCESS in 1h 52m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af0ef0be9ee34afb959df721195ca0c8 : SUCCESS in 35m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4fd07d8cf5594fdcb03ad7633857e84a : SUCCESS in 34m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54ee50c4f3904769be33ceb43a14afae : SUCCESS in 33m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b41a234ef6a486a9c4d8dc631f25e61 : SUCCESS in 50m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/625b9f71370646b493e9e49b288da13b : SUCCESS in 1h 37m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/964da135f171414aab9f62a1b2f70ccf : FAILURE in 53m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bab6876b3e744681ad8b8d9ebf5514c7 : SUCCESS in 1h 29m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/446f01e05e0f4d348fceaaad826d2df3 : FAILURE in 1h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ef5f3804fe6041deac2e11e03cbfd5f1 : FAILURE in 1h 27m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8e5d2c3befcc47ef9fe876b79d3a7bfa : FAILURE in 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b65c079339d57e24109bf5031a048b30f4a0384f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:05:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66ae8f6a5bafbc97b175b7e021f378d242ab4b69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 18:43:34.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcd2b35ec6961415e128d305df3da0579466d69c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 19:28:13.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc509c7e22d2a6a9e4768e628321274b04945ef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:46:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85154c5318bd45638efa5d81434a8f9b : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdcdc4f1f7bb47ce9a479831adbe1fb5 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3596858f3f8412a99305432256455da : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ffef1f883f4e5db41f2f43e25d29a7 : SUCCESS in 14m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/12f3e94e0d1944b4888bd7bc620bbccc : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19d841e7bf344ca1bf4bc8677cc92ed9 : SUCCESS in 11m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d7ed5d099994c2d918f233a333ec8d2 : SUCCESS in 1h 38m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39c35a55cd4044a3bcd9105dafb6bb5a : SUCCESS in 1h 55m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ecd5f972f4048e4bce8bce42a43ec2c : SUCCESS in 38m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c34e01cd31fb4557bcedf009ed3b634b : SUCCESS in 31m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3197c9eeb0340d48b004dc3f43f8e86 : SUCCESS in 34m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b390c17eac034b83a3e9206e52ed5588 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ed6173f2c0efc4420c67a1305a92b98e65576387","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:46:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd70aa4575e4792b986b48c8cfa209deae68f897","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:48:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b6fcab329a433db2328aebb64918ec : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3026d35828b0b4541de3415e22582fa523f78e3","revisions":{"e3026d35828b0b4541de3415e22582fa523f78e3":{"kind":"REWORK","_number":1,"created":"2021-10-27 06:22:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/815586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/815586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/815586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/815586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/815586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/815586/1"}}},"commit":{"parents":[{"commit":"3f83471fcb5eafa4e9fa3b257a9dab88ed32bbd7","subject":"Merge \"Fix management network selection when calculating deltas\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f83471fcb5eafa4e9fa3b257a9dab88ed32bbd7"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:24:06.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-27 06:22:33.000000000","tz":120},"subject":"Fix failover of az-specific loadbalancers","message":"Fix failover of az-specific loadbalancers\n\nThere is a bug exists in CalculateAmphoraDelta class\nIf there is no such a key as constants.MANAGEMENT_NETWORK\nin availability_zone dictionary, method get will return None\nIt means that management_nets list will become [None], not\nCONF.controller_worker.amp_boot_network_list as it supposes to be\nIt breaks the following failover flow\n\nStory: 2008669\nTask: 41957\nChange-Id: I435c40a5493828bce24c2850182f0ddc8e5bdd1f\n(cherry picked from commit 81134bdfec79c8019c26e7846d31a2b2952965a0)\n(cherry picked from commit 77a87a7f335464180933cd1859cd34998d7f2e75)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3026d35828b0b4541de3415e22582fa523f78e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3026d35828b0b4541de3415e22582fa523f78e3"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
