)]}'
{"id":"openstack%2Foctavia~815587","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I435c40a5493828bce24c2850182f0ddc8e5bdd1f","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-10-27 12:40:59.000000000","reason":"Stanislav Dmitriev replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-11-04 21:07:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I435c40a5493828bce24c2850182f0ddc8e5bdd1f","subject":"Fix failover of az-specific loadbalancers","status":"MERGED","created":"2021-10-27 06:23:32.000000000","updated":"2021-11-04 21:09:16.000000000","submitted":"2021-11-04 21:07:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815587","meta_rev_id":"d35f0f52b324dd8d5f0d2f85f916332d9d5b69f4","_number":815587,"virtual_id_number":815587,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 21:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"all":[{"value":1,"date":"2021-10-27 12:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":2,"date":"2021-11-04 19:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-10-27 16:09:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":1,"date":"2021-11-04 19:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-27 06:23:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"state":"REVIEWER"},{"updated":"2021-10-27 08:57:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 16:09:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-11-04 19:20:14.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"493fae095e2002a1e4a3e407f34e4c3dfdb7373d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-27 06:23:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c5822eaf3031434f258aaffcab7c39d1c1e3303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 08:57:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/878c8ad28ccc4fa39fced1bac54dbd7d : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5790e947233343bd8d79cdef0a2636c7 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91cc4e27f3a14d60802d52a73ff4509e : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10eb9b98f75c459a9d2e222131b409f3 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a3b6d59082425e96cd4a701c13e347 : SUCCESS in 12m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/64d266679cb34a08bd9f8d6c89fc83b8 : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8a9d032d71554a9bb60a3dc6074c2caf : SUCCESS in 9m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/595911a25d7549ea922b1f81d7740602 : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e858a99f679a4b29acb122dad035873b : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf5f18a294df4c728ebf47a44ea823b8 : SUCCESS in 1h 49m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7302916363d437f9474f173e84c18bb : SUCCESS in 43m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ffa4fd14a48e4da5a68f4ca3737310e3 : SUCCESS in 43m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ae5950b497149e9a9214c41bb4d0d32 : SUCCESS in 37m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0243c5522a74cb2b088a7d944726b65 : SUCCESS in 1h 26m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/027ef21b1e5e47dc9cf23ae91278e3db : SUCCESS in 2h 00m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ac2de7ef3c041978f1a32f042c2f09f : RETRY_LIMIT in 7m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/82c1374a49534e088fdcd7e72bccab8a : SUCCESS in 2h 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"841e6acaf56a93a8aecd56a2714445147416e332","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-10-27 12:40:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a94b41437b6aed38c44dd6ecb3a0ba6e3bfa4668","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-27 16:09:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bb390edb4be352ab79f66a5b56ce475e29e9844","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-11-04 19:20:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f316ccf475d613a85ade2b0aed4a47e944b85e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 19:21:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef399212de01a02bb1e20161bc3405cacae4bfad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 21:07:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404c854b92d2404e8c30b27bd25756ff : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/247614dd55514d829b1bf5ef91e39832 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d95746dcdce42b4beb615a5db1e5124 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/940ed7c3112d4d7cab0f909c1b356f04 : SUCCESS in 12m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8762fcbbf3d482fa612fef159cf15ed : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd1af818d4cc49fb9dfd5e53aad88b73 : SUCCESS in 1h 11m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f1e66d705b947adb171700d901e5c54 : SUCCESS in 1h 44m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e76a99c213d4667a3bd0e500c05912e : SUCCESS in 35m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e8d5eead669040f582915b9034c8f450 : SUCCESS in 34m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c462ba8df48d497ba890925fd5d7475d : SUCCESS in 46m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9522c4a9af1d476e99bc41bd73614d07 : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"52216521ac133b21f8dfa5d325bedb90f5b94221","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 21:07:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d35f0f52b324dd8d5f0d2f85f916332d9d5b69f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 21:09:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e544b8769f4ac28f183c380581d5b8 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b4e05aeb6525a7f42c5f4d1193def75d9551529","revisions":{"0b4e05aeb6525a7f42c5f4d1193def75d9551529":{"kind":"REWORK","_number":1,"created":"2021-10-27 06:23:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/815587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/815587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/815587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/815587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/815587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/815587/1"}}},"commit":{"parents":[{"commit":"178481be9ada47558d0159bad1a474e959e56015","subject":"Merge \"Fix using host_routes in VIP subnet with amphorav2\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/178481be9ada47558d0159bad1a474e959e56015"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:24:06.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-27 06:23:14.000000000","tz":120},"subject":"Fix failover of az-specific loadbalancers","message":"Fix failover of az-specific loadbalancers\n\nThere is a bug exists in CalculateAmphoraDelta class\nIf there is no such a key as constants.MANAGEMENT_NETWORK\nin availability_zone dictionary, method get will return None\nIt means that management_nets list will become [None], not\nCONF.controller_worker.amp_boot_network_list as it supposes to be\nIt breaks the following failover flow\n\nStory: 2008669\nTask: 41957\nChange-Id: I435c40a5493828bce24c2850182f0ddc8e5bdd1f\n(cherry picked from commit 81134bdfec79c8019c26e7846d31a2b2952965a0)\n(cherry picked from commit 77a87a7f335464180933cd1859cd34998d7f2e75)\n(cherry picked from commit e3026d35828b0b4541de3415e22582fa523f78e3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b4e05aeb6525a7f42c5f4d1193def75d9551529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b4e05aeb6525a7f42c5f4d1193def75d9551529"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
