)]}'
{"id":"openstack%2Foctavia~815588","triplet_id":"openstack%2Foctavia~stable%2Fxena~I435c40a5493828bce24c2850182f0ddc8e5bdd1f","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-10-27 12:40:36.000000000","reason":"Stanislav Dmitriev replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-11-08 14:07:13.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I435c40a5493828bce24c2850182f0ddc8e5bdd1f","subject":"Fix failover of az-specific loadbalancers","status":"MERGED","created":"2021-10-27 06:24:24.000000000","updated":"2021-11-08 21:48:08.000000000","submitted":"2021-11-08 21:45:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815588","meta_rev_id":"e297aa4558092fddbbbe5e320c7be1aff9b1faae","_number":815588,"virtual_id_number":815588,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-08 21:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"recommended":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"all":[{"value":1,"date":"2021-10-27 12:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":2,"date":"2021-11-08 10:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-27 12:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-10-27 16:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":1,"date":"2021-11-08 10:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-27 06:24:24.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"state":"REVIEWER"},{"updated":"2021-10-27 09:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 12:55:03.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-10-27 16:10:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-11-08 10:50:40.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"1cdcbc71901fecaf9a69599238790ad70b569612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-27 06:24:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e293a17c2b6e4d171b2353de2b7bb2d67b0f4e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 09:13:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc6ec42d6b5e4064aa58586c5b983dc4 : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5056adfc8f2c45ff99f3ad1218d1cf5d : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4981e29501a4f19955da0946a2f47b4 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d71fac6f42c141a88cba3b163d8929a1 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0784cc29826047209a43699f59ac144a : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0b31c2e880488481c7d7e6ab8c4631 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199f372d4d1a4be783c5b5ea6b2e9338 : SUCCESS in 11m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a824dc44f56a4af686b2babfb6aae1c7 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/83c5450d4ad343d6a6f9b87f82fdc3ed : SUCCESS in 9m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/684e36b071b047c2bb971954bcf39106 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8291b186c042404c8111384830584b63 : SUCCESS in 9m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bdcf97141a0c42ecb16fb2fbe374b8eb : SUCCESS in 2h 03m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cec37d250ddc4229b0633ea69809187e : SUCCESS in 1h 46m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/134a1b64e9d948acaa9fd0d7f5d24fc0 : SUCCESS in 37m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f497cdbe641045cf99b3b09a6d55adc8 : SUCCESS in 48m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7f1725428424874a5eb60baacb8e9a1 : SUCCESS in 42m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0d2434e777ec424c800dceba5e0d1662 : SUCCESS in 1h 18m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2a573540dcc64ae78c3165ef766d5d7e : SUCCESS in 2h 45m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/849336ee072644f78859793335506937 : SUCCESS in 1h 29m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e321bf2a2f348eaa5291f0f7463af88 : SUCCESS in 1h 50m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/08eed7cec5e34034ab84ea2076968344 : SUCCESS in 1h 19m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/93c2c744121b49748386e99b63b6f33a : FAILURE in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"622f02d2d84385fd53e9df7e0b45d752346b21af","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-10-27 12:40:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"618bb31ae84f69f7023884b6e1df21bca9b9a0b6","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-27 12:55:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"38a5abff9d0ab7bf7ecaff5ba32b21fa1e4e0046","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-27 16:10:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7811f2d4efd1bdd6e8d2debc403c985b3e5307fa","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-11-08 10:50:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20c27db4955fb5314732a371bb4f8ef45e0a6307","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 10:51:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3caedc055c6e39a82f5de6019275f079cdebc6a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 12:33:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7355f5a341c4ea2ae27445dd437142a : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171be6a7595a4e3e96c9f5973a0d9ccd : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d38fd8ba7034c37b41ae97c685a9139 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc654b05621d4d96becce2236864c9c8 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5045c1ea250f417c8ebbf91a85587108 : SUCCESS in 14m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/31fedbf83259486e91fa0ecef6489412 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7524a601fa94d7ab6d02d10e02095c7 : SUCCESS in 10m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/679be7f839e244b58b8b91345a2a8d1d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99f08afc75ac421b9a18d17552548188 : FAILURE in 1h 36m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58423998883b44f1bd355666e0786753 : SUCCESS in 35m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/625967e2cd024308b8529341b88f93bd : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":1},{"id":"009dab7e30d97732f91450b6b13b7f6c869808c5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-08 14:07:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63f7d35c8c8dda5cd04ea57fd504fe7bde5f22d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:40:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae9f9ddc2a2248c88404d464a889614f : SUCCESS in 12m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb8a9308612d4563b80974b76608ee65 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316830f2ddc7417cac75c70ac2cbad53 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c04735c77c448c38d48786a445d210a : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c246d0bc89c4613b23efbf9b8ef3c55 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05ce8f8060c547428cd3aca9a1474e2c : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9052f25fb58744f5ad0b4b6c1003f245 : SUCCESS in 12m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a26f4ed7b65846babde6d158ad06e3a2 : SUCCESS in 7m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7c9aeab042ee46a59d22ece614b7133e : SUCCESS in 10m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/70158bb5a4eb4bad9dda92f16698d764 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d01f2ba66d9f45a7a1ce6be5f4752160 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/989a8cd1416a4f61a7c829e354cfde93 : SUCCESS in 2h 23m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b3ef4a6196a4614915b64d0bcb4e51f : SUCCESS in 1h 57m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ae292e7b8aa545ad8a83e3e52fb7b762 : SUCCESS in 41m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f11d8194f25b4829bf9b53db7f929758 : SUCCESS in 58m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66c43b7a97d34cf7be370d3c4a875df0 : SUCCESS in 32m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f206ad60d10d404d938cdc73816c82dd : SUCCESS in 1h 20m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9aceb49b30f45bc8f553999ab9678d5 : SUCCESS in 2h 02m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dd10c0c133fc41d1bca96be78a4793ee : SUCCESS in 1h 25m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e3f1324e10f4fedb74dfdaeafd4676e : SUCCESS in 1h 38m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/523abe7ec7464c1a8d207804c30777ff : FAILURE in 1h 49m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1dd3377065b94482ad36c748cbeef131 : FAILURE in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12852d995af8fa66a7777155ecf0182368918449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:41:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"436ccbe22c4dbb438dcdfbc1ad0512c2ff47877d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:45:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e9231cb29434950bc05ed5c25794874 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b631a7e05ff44e10b1d7000e8f9e66d6 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa98a968a52f40a6aa3a2f721921c7bb : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52a687d186db424fa0daea9ea9f05a83 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730567b196364e9abe3545d27cb97c47 : SUCCESS in 14m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c334ab9f0c824e32a30dbd92745afce3 : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f780d36b3634a0e9647b127776fdebf : SUCCESS in 10m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d808787ddc944b281e8aec54be387b2 : SUCCESS in 2h 18m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afcad286f3ba4fad92f5a81ba7719445 : SUCCESS in 2h 02m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ae28db134d24b3680ba7d99f70687ff : SUCCESS in 46m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f963978899e24a9da26c3a853541f404 : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"25547060a0d69d93944df52a85ba2fa5ceb7ca99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:45:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e297aa4558092fddbbbe5e320c7be1aff9b1faae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:48:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82adbe476e1046da97319ab844939c2a : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45c992c1f396bc66cb61bf8fa0e448648a0d5f53","revisions":{"45c992c1f396bc66cb61bf8fa0e448648a0d5f53":{"kind":"REWORK","_number":1,"created":"2021-10-27 06:24:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/815588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/815588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/815588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/815588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/815588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/815588/1"}}},"commit":{"parents":[{"commit":"8681637da0f99b4dda128bfd8ffd45be20de6112","subject":"Fix management network selection when calculating deltas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8681637da0f99b4dda128bfd8ffd45be20de6112"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-03-01 20:24:06.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-27 06:24:01.000000000","tz":120},"subject":"Fix failover of az-specific loadbalancers","message":"Fix failover of az-specific loadbalancers\n\nThere is a bug exists in CalculateAmphoraDelta class\nIf there is no such a key as constants.MANAGEMENT_NETWORK\nin availability_zone dictionary, method get will return None\nIt means that management_nets list will become [None], not\nCONF.controller_worker.amp_boot_network_list as it supposes to be\nIt breaks the following failover flow\n\nStory: 2008669\nTask: 41957\nChange-Id: I435c40a5493828bce24c2850182f0ddc8e5bdd1f\n(cherry picked from commit 81134bdfec79c8019c26e7846d31a2b2952965a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45c992c1f396bc66cb61bf8fa0e448648a0d5f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45c992c1f396bc66cb61bf8fa0e448648a0d5f53"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
