)]}'
{"id":"openstack%2Foctavia~816370","triplet_id":"openstack%2Foctavia~master~I3f9d08fd4ce038208b8c2d160e6abecae8c8383d","project":"openstack/octavia","branch":"master","topic":"c9","attention_set":{},"removed_from_attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-02-07 12:19:04.000000000","reason":"Adam Harwell replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-07 15:02:28.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I3f9d08fd4ce038208b8c2d160e6abecae8c8383d","subject":"Add support for Centos 9 Stream","status":"MERGED","created":"2021-11-02 16:10:46.000000000","updated":"2022-02-07 19:02:28.000000000","submitted":"2022-02-07 19:00:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"816370-c9","meta_rev_id":"79aeaa7d90f2e4ab94c310d67ff58e0c34da80cc","_number":816370,"virtual_id_number":816370,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-07 19:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2022-02-07 12:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-02-04 19:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2022-02-07 12:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 22:35:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 11:16:29.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-01-31 12:34:33.000000000","updated_by":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"reviewer":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"state":"REVIEWER"},{"updated":"2022-02-02 16:14:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-03 12:40:45.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"b7164d9fd31179acf764d1a7b29dca1926f63825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-02 16:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"000b0c14234d85d002e3eea97e273046e8ca7062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 22:35:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/255264776fb1403c873d2053e4718eee : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5feb71f3be14e0c821e6fcb311f2620 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af20c33ee0a44bb28fb50b6a4ef40149 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88eefb20d8e24c84bd2f9f13e96ef956 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111c58567d654c81828292e88c58eac2 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/554ad9b3cafb43d083fdb51c0dcd71fc : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a54b8886bc4fbbb3b7c0a7508ffc93 : SUCCESS in 13m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b2f65cfa22e94470969aa4e4e5d368d7 : SUCCESS in 8m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6b402fd49f454b0a9f723ee73c9c958c : SUCCESS in 12m 37s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bdf2fcc65834a59936e781f9654376b : TIMED_OUT in 2h 36m 36s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/88ed6d5791b8413585a5c906ef24ae01 : SUCCESS in 13m 10s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c47b58d885141bfbaae68d25c7bd87e : TIMED_OUT in 2h 36m 21s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/84c2fd23cc9a40628a88b91fd30e341e : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c3400bb225aeb531c55c07e16b8b1bee42a8a748","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-03 13:22:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3602f2dc54245798a12c9c5673f6565ed2c9d862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 16:01:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dee624917b59435ea03300705563f6ce : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6453e6ffa6d0421b86daf5b866a60c4e : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8e3e6dd1954aec90d5f4c32d5834b9 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e3cd0d2620d4513abb3069e151623aa : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81dab1303ca24570acaf863bc86f0d2f : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1de6504590b04ebb91b9ee0d629d8961 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b47d3c66c3cb4eb8adcc651f295313bb : SUCCESS in 12m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/49b5ec9cd64f470bb87da310bad659e9 : SUCCESS in 7m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/65492c65210c4362af742f1e31920561 : SUCCESS in 12m 03s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed1fcb62dde1463b9a1b2a7ea7935a7d : SUCCESS in 2h 18m 18s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e54174d1af240bc8111db38abe969d7 : SUCCESS in 14m 34s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3ef74d1640f430ca0deedd6917861ea : TIMED_OUT in 2h 35m 27s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3e2bae83314476f91688e61e6bb25e2 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a5c0aaefbdd3c0de5bae9f2bceb75c19bf7d28df","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-22 07:07:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3477f2d4b07cb3680175d80f435ee0f04b36803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 09:48:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6df354fd26f24a35a60ee4f01df93bac : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba83290705e04dd9aca1068f2919480b : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c45583341b541b1b28b531843be3402 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1d5e78deae6461ba803b909e94f6f01 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc54f69bcdb498f96a554bb4998b1b6 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b57a871ac64486a97b1db2dcb03a1ae : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cfb77cda4404553a6128df85aeb5d0e : SUCCESS in 12m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/180b1d1a35c44a6db52abd8bd0944b4d : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a0f4e34d92f84581bac32d0fdcc9214d : SUCCESS in 9m 03s\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3ae6117067146d096ace4228c76d21a : SUCCESS in 2h 12m 56s\n- octavia-amphora-image-build-live-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5e2695755ca4a9e8d229dcf66e7ee10 : SUCCESS in 11m 44s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ae94ffd7619493eb13756a1372b6607 : TIMED_OUT in 2h 35m 08s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e0f240e76ac4c8ca2638e5a0e8b133f : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"94bcc674b174cb0270e80a82b7192a09fd02e1f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-08 13:57:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"52ab0f795eb86a8059d206257f625e82aff6427c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:46:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/686677bbf734493cbcf101de0bc2489c : SUCCESS in 13m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecdd7af6507a4f7dae192f923afa154f : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d838ae72f614abbb0d8cc2cbe43bda5 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/693db70e161545ddb0cbdc8ae34b47d7 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94c9b3dcdbc74ab6972e5f33fb81bbc0 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb22cb6de8d547bfb7ff78c1e161a02f : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431f0f80c8bc4578a12e939f5374df21 : SUCCESS in 13m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/abb564ffbce84070a098ab4e4b72e9d0 : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/911f6394817a4d18a5d31574f70869ac : SUCCESS in 11m 19s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/66de2bc946c542f39bbdc023268d2da9 : TIMED_OUT in 2h 35m 12s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff040f8f9d1f4f88845c27a13afa1f5c : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":2},{"id":"7ae4d63e23c659a0250f1ee17e9582c57eaed23d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-09 15:37:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c3c5b0fdd552375e92bdfd1759f18f30ce4267e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 15:38:22.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"516b00b11734a933656b474d7803e99465fe4e70","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-09 16:05:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"13f385a917c9c2b2b5e991d0cffaef62e9b16643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 16:06:36.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  The nodeset \"centos-9-stream\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: octavia-amphora-image-build-live-centos-9-stream\n      parent: octavia-amphora-image-build\n      nodeset: centos-9-stream\n      description: |\n        Builds a CentOS 9 Stream amphora image using diskimage-builder from Git\n        master. This job does not publish the image.\n      required-projects:\n          - openstack/diskimage-builder\n      vars:\n          amphora_os: centos\n          amphora_os_release: 9-stream\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 179, column 3","accounts_in_message":[],"_revision_number":3},{"id":"391350954565987c4e230082c60934a5ce3f3db3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-09 16:24:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"732e5eec771739936e1613404abf3fa63c0873c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 06:55:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eedd7f46a6b489bb2a44de2a3727383 : SUCCESS in 12m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/046be3db460940d1b1e4178d1b6fbb19 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e1f639b7aa44bda32ccbe8a2598bb5 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d28f0017f7bb491c9499c146090cb681 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f04b718233c4644999606bb6d8e36bb : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a5fab7a0e394bf99626958524d1dba7 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c84e6f176124f2993816e91a3f142b2 : SUCCESS in 11m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7ddb4ea6c3f44e74814298df51375bd9 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fc032355cdb94c3184516492d987f5f1 : SUCCESS in 9m 24s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b038adb8f404beea09d6980a57b8425 : NODE_FAILURE in 0s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/618a04d4a32847a3af6e7461fd10b54b : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":4},{"id":"20b34074cb86f55fc64105c969321fae0d4c8d1e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-10 08:27:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8078a7dde40aba7c65d692634fc1fc7c58d53217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 08:42:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/975b4d6d028741789349dded735c77bc : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74b0d17392284fa9a931897302126102 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc9304fa6c24cc89d480d847090e873 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecd89bf5fa224348883843b052eebfbb : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1574e9507fa04e968fe1931f4afcfb97 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/549abf282b18406eb399273b1270ee66 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bafbf2bd69cc4738a3e343e4e9bf0232 : SUCCESS in 9m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d9ac0725d7a54685aca61a05110e8f81 : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9e63641034194bc2a8f79db7c3463d4a : SUCCESS in 8m 57s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2be100e34d4d4823bcdb6feebea12f74 : NODE_FAILURE in 0s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/d934aaffe54a4d2e8ed2860e4a55044c : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":4},{"id":"21458f281c5f96802bbd61fd3cb5f02456426259","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-13 08:05:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c326a6eda9156e7982cb8cc5cf95c5341b6a9631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 08:45:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a34f947d5ba44020ae701de8a6b4c87a : SUCCESS in 12m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85f7205b500d4fce971e38e4acc5f172 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcab28fa9b224057b6771e0c1581e49c : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70a8db74492f4a1b874178f1169045c0 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5b2c86c72064e618a1a2959af989dcf : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e73a9175ee2248e3abf7b94040d33ef2 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d35515b8f94dbaa2a8c23de8cad0b1 : SUCCESS in 9m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06e68fa87ad242c791b0b91aa3c559dc : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/501f5ed9ac3b4958b5eaaeb1ea2dc505 : SUCCESS in 10m 00s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab0bc81390ff45db95620e45fcb212f4 : FAILURE in 20m 32s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aa6bc34c9a344eab38edbd59364c6ac : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":4},{"id":"eae3de20f45f93445d343763aa9e33e32403c8fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-13 09:05:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a42f94dea4b87ede97d66bdfa3d334a661abd25f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:23:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf59a2d969784a38af5ce4a3bc1169fd : SUCCESS in 10m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da0ba179c3824b62ac16cae40440e46a : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4ba9641ce54c9281ff1e6043bc49b6 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b480191bc77547a699716149cf0438bd : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b4acea234c1440ba6e50bbfe51e1015 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15ca8350a94c422bae80145036063d9d : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a76b174719490ca37aed0d90c7c263 : SUCCESS in 10m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b634e2b2fb94547b2f87d3fa91b5b87 : SUCCESS in 5m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/64a16de24e354979936538646ae4b36c : SUCCESS in 8m 24s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/625ead172ff84b8582a1d996db6b9b97 : FAILURE in 1h 15m 41s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/138250a74b334b239a1a1f1c1714093b : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":5},{"id":"c2875730b46c571aee2c452297d5017390ab0140","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-13 13:43:37.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e4f7e215fae855f57979abe636417dbd4a1a2ca","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-18 10:39:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"73bb782ccb07cf5e4a5131a39e3516c42e99d522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 13:12:02.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d3fe9cc42914d7da5b6ce8229403534 : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00cbcbf56ada43d5a260a34921ec271b : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bda622e135d4b6e8c1aecc0528b89c5 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17b3415dd4ac4c768555a17851cb9606 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/167b54582c574710b87393056d999308 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/784fb73f17da4076924c681dbe809600 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da3c6c8c6854e0287aced2e2fe96f7b : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5f012b464f794553891a9f083db51ca3 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0ff5a8f871034bb595cc55f776d837b4 : SUCCESS in 10m 01s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b85717638084b31a211f7714e7d760f : FAILURE in 2h 16m 41s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8a1ebc934b24ba0b28cbe9362d56ed8 : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":5},{"id":"9d1dc1d548760b877ce989eda1b6eedc0da2f5d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-20 14:54:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e31b65606eb9d89c1b9a22b8cd7e20d3fa346557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 17:07:50.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7175d1d7f73d4a24b4f278854c367f9c : SUCCESS in 14m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9ad784eadc24f87a7b7ca9e87a2c982 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a705b170f944d118b785aa014b686be : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/171f3a6eb1b44421927ab8872b65145c : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a033c2255924973a8b0cb3f248413c4 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/761a383c48904647a908ff213d03f8dd : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b94b17ccc7b46ee9eca4773ee8488d1 : SUCCESS in 14m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06441520015c42799982c6db5ae632ee : SUCCESS in 7m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/10df47404ea446138912ba6772338240 : SUCCESS in 13m 28s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7629aa67b0864285a9fb7b35d7e1969b : FAILURE in 2h 10m 46s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5c5ef3a740d4322b17a4ff57172873b : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":5},{"id":"05e18f569339d7f5fb522a01fc87320849a3d580","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-20 17:15:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"546dbb0269f467a3809537a1eb125ce4a93041a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 19:14:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0317eea8e78644a087fc37ef49ac3f00 : SUCCESS in 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b4ec7d9f5bf4e78868289861eda877b : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf3fbc120274725946398aec3437e07 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4690ff4ab2c44550b2f050c54c588e39 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72be295a1dcd442284ce3c60513d7d77 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f98ebfa19164f728644b5193aabc0bc : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96fe556556543f4bbd0661c7ae998de : SUCCESS in 14m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5406662d3cff4555966b4c459a845a1a : SUCCESS in 8m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e3fbc5f3bbb447c3b3cb5fef10ef217b : SUCCESS in 11m 30s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e7ecb939f7d46a889f5f6bd4e29b4af : FAILURE in 1h 32m 54s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/097719a7d13448e19676dfafc75d3a48 : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":5},{"id":"c930b49903cc2326cf7b221d1e1e7d7508560243","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-21 07:06:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c0be398e97f83f3a4dadea8a95801e1fef1e3712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 09:51:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af02cd6098224d19b3aa5f103ac56454 : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f287843400844939d6c075370679d88 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c81d1172a04db3899ec9956ad5bfe9 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f36267b74e74b26bfb74070f0b7234c : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b73cf4598d455a9cbea9544e7a7edf : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/add5b3b48151418499bc8128b93e0324 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c84418e81be4d308c412aa91952241d : SUCCESS in 12m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e568ce056a9444c7947ceccb48cbfe67 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f370c774747d4856902fe956bafb2c8d : SUCCESS in 10m 07s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6093cad3228e4e6cbd5c953df518c342 : TIMED_OUT in 2h 36m 50s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fe04afc346c402bbd2356e83bf6018a : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":5},{"id":"16ca6ff1ae8c529ae76ad6f7b06fbdb656777391","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-21 12:28:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"471db66f48828428eb018cc2dd47bf8c177a78ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:53:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c77e9888f97e45f094117946c6cbc48b : SUCCESS in 10m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ab6ef1bc39a47cd92cbd0e9570c52b9 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6308fb5e0d4695bbb92b2e392fdf12 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d61175ac13a94483b9a01e457aebcb7d : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da3a833824fa4e97a933b101aaea1156 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f62c845808a74d48859aa26b5edf994a : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77073d31df5741c69fa70a55d8cb8597 : SUCCESS in 11m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b6e3860c55654098ad07e0d389e10ccb : SUCCESS in 8m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4885a949408249ff90b182f849e73943 : SUCCESS in 8m 48s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8827a960df564d77897a49a0408853fd : FAILURE in 2h 22m 52s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b70f4aefaa34edc95d59b0dc7901b4e : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":5},{"id":"8dd495344df951191a51fe9f80d5e69688c666b3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-24 10:07:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"67b0dbfdf520d301ead091400bc0d128d07a8d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 12:23:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28e398225c5c4cddb794cd861924b6c0 : SUCCESS in 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eadaf5e64f374f51b4b86f238f83e72c : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2a27e80e3444b487d8294e06117be6 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0032539359a84e4c8c2a398c483235ec : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dade625c20984ce2b8320493c5d4e6ed : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92158a73a25e4258add8a37009d8817b : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21900a2cebfc48dca5db6b4bd4546706 : SUCCESS in 11m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f1ceff9c7ee640deafaa497e87f4880c : SUCCESS in 6m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c0526b732ee4408ea61a6b16d2980039 : SUCCESS in 11m 06s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4291ba4c08514cb58e4dea5fbc4eb78f : FAILURE in 2h 13m 37s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3422ecd3e974087ba02819ce5d9925e : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":5},{"id":"a161619072aff20e8f085d95ae7b322364cf5b4a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-24 17:15:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e7aec5a1048bb8937693e5fc151e955a25f3fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:44:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c74c4c7327f647a49f6cd93e09ff91ad : SUCCESS in 13m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21d8276704e047d28a9a0096374104d3 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cbee254d82435caefe16b2e6716f22 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39fa042cb8d5463da0a305c01fafbc60 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a245010335d45928e6e27d0481fb50c : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc260d25e48b48b3876a8ec00a6cdddd : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41785cefe4c54b918822e6470099b677 : SUCCESS in 13m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0d850b5648a6456182b134acf646dab8 : SUCCESS in 8m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ead7d5d2b2d84da69c2f0f4b230947fa : SUCCESS in 10m 23s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b15ec90f22a24cdf8c8d3d0ded99996a : FAILURE in 2h 05m 22s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/577c8dc12c424cca86111f63f9cb7ba6 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":5},{"id":"98e2e7f36d044e789b9940cf4ce08ac6687015c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-25 07:27:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"897b31f3fa3e51e4aa462744c82146bdfeb9f1bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 07:42:46.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"1451ddb2e3d72b66b7fe66d762e6a5ae32047be0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-25 08:11:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1dfeb57dcc75b3621605c435b8160fcd7d1029a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 08:46:50.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae8f7d0562a14acab0510642f617bf45 : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/429df367df2f49198652275c99aa9c02 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097ab84c07cd495f9d346f85413af3eb : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e4b4a2bd11445b98f864061a25b0f3 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f86bbc03980f448d85b343e0e663b2e8 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18e1bb8aa3b84d8cb984e0cc838e0fb5 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb2404a5f8f49ca921aaa5b4d4d8466 : SUCCESS in 12m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e553af7fb3534471a10c63c5c5ef0775 : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f21c8a9eff44e5b8a96bef963e03d51 : SUCCESS in 9m 15s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c09f8f723e0b4a9d9592a075590ff0b7 : FAILURE in 30m 16s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/e801b0b01ebf474091a1edb6a4b7a7fc : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":6},{"id":"bf0202759bbdfe4215e4597f30a5d5a65aac38f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-25 10:01:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"faaf056d3672e2c2e67ef33c24a22442e114835b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 11:54:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7e99ad2cae849f09494d9bf15802667 : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d547cfd3ed3545a8b1fa647f37a58d76 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce3337191b964a418f8392332159e15d : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/403fca614ea648019902b28d11cb5c62 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4843d3a55542471bb3d87152d7bf8f7a : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31ce4e695f0145688c68baa137c21fe3 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a1f17e8c7f4bcd9b0b4e3ac53428dd : SUCCESS in 12m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/513afcd01cf24b8fa2034aaa7a9c2dee : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2297dc1a96ff49ae9f499dc38b31061f : SUCCESS in 9m 41s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/53efd1aa053b462fadebc1c767308f4d : FAILURE in 1h 47m 42s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/4aefdcbb6950497f8375114654e0fd29 : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":7},{"id":"32371ee05fd09eda669110c9631801d1867ac412","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 08:33:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dfd5dc944ead8b2d419375f32dd462bdddc397e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 08:48:18.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ac82856a19c4d2e8e6e94c13d60e57c : SUCCESS in 11m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35101e6f6d514430bb46209dc30042be : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f19f5a39f0648e185dd7c797720036e : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d8273ba94924e59afe7c91ea9bf96b1 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e6934cd74d94969bd233dd476b0bf4d : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31a81c6ab432470c817897b386edf23c : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d8473fdfe64ab887824fcb897ff0e0 : SUCCESS in 12m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9da64e365f2846ba9484b05f1b8379f1 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6748cfdf32c04853b94b5438efbed2ed : SUCCESS in 9m 00s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8fd20425399480ebe65cd9cf4d9b226 : FAILURE in 11m 04s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/607a5f429b834c56a95f55d0809d2d5f : FAILURE in 4m 42s","accounts_in_message":[],"_revision_number":7},{"id":"f12d74f4e2dbd057adf3e52b60c77a9df2e01741","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 08:54:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"07f408fd067ba935de856706b2318202dce186ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 09:13:46.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6bc62ff20dd43fc9ad56566967a0c3e : SUCCESS in 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ee98b42076e40ab9f1600beb656fea0 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027846be77914404a2e5d12b6273a94b : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b15bfe067bf4951ac13ba93a86785d6 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26874555230f4080844e6c41ebe3f10e : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac4de34449fa4f1f9b388131894ac372 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011b9c29e6234227a36e065575921d39 : SUCCESS in 16m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8a5c81b6eee64c57b56cfe6b05791471 : SUCCESS in 8m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ac0cd16abac841e39a59faf9ca28b399 : SUCCESS in 8m 45s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/68c090bcd6834212a8c33de425ed47c6 : FAILURE in 13m 05s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea7590e7fb5c4e129c33636624753616 : FAILURE in 6m 16s","accounts_in_message":[],"_revision_number":7},{"id":"d7f6ae749039b6be2667226634e69a340b24ecf8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 10:31:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"895736ad3a299ad25df14a558ed06b1b946e2b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 12:32:44.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00319844e6624801bb362ead591acee0 : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/368928af283a44efa81d9862cca72231 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ddb0389b7444e61af1663b929f8bd69 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3003088ed8d84a1aae022890d1e627ef : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/646238182f5048d2bea2aa4de3e16e5a : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2d6bd046676473bbb9012337a0274a2 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8eb1e4b66744d1af8c30bd6ba04e5e : SUCCESS in 12m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5ad4d8854eba44368487d22325996629 : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/82ec26425701407186550cb1a319840e : SUCCESS in 9m 54s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/243e7e9ff7144cc7a8824a540d42f07d : FAILURE in 1h 51m 14s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1d941620c85476f91b7b4f0f4df7661 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":7},{"id":"128ca4fc6e78bdb386da207a14b577cfe22d9a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 19:12:29.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"86eff37c20fc26611d82dcee374fe2f749d60d39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 21:57:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f765f69b51fe42dc842929d3d6163706 : SUCCESS in 14m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c7dd7c253aa4f7490cbfe18674520d6 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32af2202d0714d54890f59699b60bf48 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e1418ad027e46eab69871bdd25348de : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a7c879e08b498db76f23352e8d0bce : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e49778544f384a3e984cd569e26debd7 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa05899e74c4c9fa42c3a8ecc8e61af : SUCCESS in 13m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/665da04ce6ac456785cfbc06ec393383 : SUCCESS in 8m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fea6882eea0744e696f80eef0019b016 : SUCCESS in 11m 49s\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fa123a5f4fc4ce89c59b047f94925af : SUCCESS in 2h 00m 13s\n- octavia-amphora-image-build-live-centos-9-stream https://zuul.opendev.org/t/openstack/build/77e5eb1d36434c349501fbbd8985bde4 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":8},{"id":"f87bcd3e4d49ecb6c3f54ff9cabd600d3a345d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-28 06:03:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"75816d5085560e5be89fbc79599b12b2d4ec26a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 06:22:59.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"e3df2b8f85bb937452e57096f66c949e5105f4ce","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-28 06:45:25.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6d2f9895cab0e4e2d64e527596dde6be1002b5ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 09:11:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2f19bb646464358897427541b38b225 : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ab5fbc29c744fe1a797101b96dd82cc : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2fee9df26fd41df80122196edbf61cc : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f755294ceb06443c81840e82c3152edc : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f875dab4825248089b3598341368c8ca : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffc08e998bfe4467b391129e59d61e10 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad909f33b7345778cb071742d09bdd4 : SUCCESS in 11m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3348cda661434bffb1a230badd62cbbd : SUCCESS in 6m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1b57743f7f4f40dfa093bbbe48c83fb0 : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/40f14c42ebd74aea99a32ca9e4981acf : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af34aa6f7c7f4b27be359293b1303a24 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eae9c160e5cc42b7a424cc4023e99b74 : SUCCESS in 2h 00m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b9587a2111842b9af6b0c08ce5800bb : SUCCESS in 1h 30m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/61386a40449744b084719ce74889d3b4 : SUCCESS in 31m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc21eca888954acc83d0b42072add7ec : SUCCESS in 1h 07m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59cd0a6b1aac4abf91240127edc0f713 : SUCCESS in 29m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/222aa898fd154b83bace6106d1a335f5 : SUCCESS in 1h 59m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c8b60a4240e4c29bcc8c382a0a5c56e : SUCCESS in 1h 53m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3eeaec2e2e48eeba2df57b18d2edc5 : SUCCESS in 1h 56m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2bbde548594c42c5a90644aa7ac1a96f : SUCCESS in 2h 12m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/82ecf928c70e45eb9b04b5fcb50b45c5 : SUCCESS in 1h 48m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e05a7a9b5dda48ba99e833f263bdda2e : SUCCESS in 2h 19m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/309ca9783c3c43e0b006de4304c2e53f : SUCCESS in 2h 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6f0a4df54aecd4077c0be2a055216df69076a662","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-01-28 11:16:29.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"f3d5579400da8019afedb6672e0dc6e81b494dae","author":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"date":"2022-01-31 12:34:33.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"aa3a7817252701983ea70331bbc01bb482a66d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-01 16:35:35.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"dfe4af3adcfaeea2f713806ef2f2f92ad2023d57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:36:20.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job octavia-v2-dsvm-scenario-centos-9-stream not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python3-xena-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n      check:\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              irrelevant-files: \u0026irrelevant-files\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-scenario:\n              irrelevant-files: *irrelevant-files\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files: *irrelevant-files\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-v2-act-stdby-dsvm-scenario:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-cinder-amphora:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-ipv6-only:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-amphora-v2:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-amphora-v2-no-jobboard:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-centos-8-stream:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-centos-9-stream:\n              irrelevant-files: *irrelevant-files\n              voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-grenade\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-8-stream:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-9-stream:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n          - octavia-grenade-ffu\n      experimental:\n        jobs:\n          - octavia-v2-dsvm-scenario-nftables\n          - octavia-v2-dsvm-noop-api-amphora-v2\n          - octavia-v2-dsvm-tls-barbican-amphora-v2\n          - octavia-v2-act-stdby-dsvm-scenario-amphora-v2\n          - octavia-grenade-amphora-v2\n          - octavia-v2-dsvm-cinder-amphora-v2\n          - octavia-v2-dsvm-scenario-two-node-amphora-v2\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":10},{"id":"1addaa432d363442896db0f3b11dbe28e795d6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-01 17:30:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8929b41869fbfd7368a06b610c75f9cb5a59f413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 19:32:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e95e74972ca4dcda497ebe822c71e4c : SUCCESS in 10m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f0391976add4ae88524c01403fac38d : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348e27787ba643b298babc09d8c1638d : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37db2627ac4845ab9620c71edc24e7d5 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc0f08ba0a5b46deb6c35e9714f48218 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db6768327e5547e2b90d3b1d0a99fabb : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434797304b8e470c9d95d8f3a3fdce46 : SUCCESS in 10m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef184b1723a8400caf0c887b43c02a34 : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/147877bf240c4a878c6b2149069d3a4c : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a7e651cd85804c028492d88226157498 : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e1dbe3909a1412eaaf38d2d58c01b2e : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1cdbc18f7c5347759517f4e30489a443 : SUCCESS in 1h 55m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ee301f9b52d4e8db627ae78ea5df0b4 : SUCCESS in 1h 33m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e059aee7f8374675a7d0237243520e98 : SUCCESS in 37m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b70591f40d1b4b00b49277f529a00e95 : SUCCESS in 46m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82ffc8e705d746958ff5282122cec3b2 : SUCCESS in 32m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0e130f5e3d42428ebaf9c0dd9c3dad5a : FAILURE in 1h 38m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca6f7661432a422b9e16068463b75656 : SUCCESS in 1h 42m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/48f7bce7a86549c7812c48ab5a102ca9 : SUCCESS in 1h 42m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a5dfce58bcb946d38788d8e861d6c231 : SUCCESS in 1h 24m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5adde730853e4bbfa2fd8c0a51f78119 : FAILURE in 1h 22m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e50cc5be62ab4cadbf7697c83fef9c8f : FAILURE in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"74e2d6e3d49044ba914cf7a12a406d1814707067","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-02 16:14:17.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f0b5b76d99188945eab34ec619e0fbe789e8676c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-03 12:40:45.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"85d3364ab5086c845707ed71ca176d0b25f21517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 12:41:31.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"708d157830f9f1b56b2176b9785fbd1a678a50b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:28:18.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"81624d5e7408dd5cdb47df51eb1b10f106978d8f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-03 13:46:43.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3eeeb60f6bd322e5286094b9cf0a36dc0231d7ae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 06:06:50.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7b95d60b56aacdd0d16f64d8b8eb018c08c59190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:45:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd0030b2991a4df4bf258a60f4fbbd6f : SUCCESS in 12m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f923eaf1e264b24a88aaeebd67d031f : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da059a865b5846b8ad575b5959d3df04 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ff4579b01974bc186f9517d16dbc3da : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6f04e50563f480385d44ae55b514dd0 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/392def9342dc4af3938b96396d62bca2 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81a1e20d6dd4aee8462a0f232c0393b : SUCCESS in 11m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/79f0ea3e971f411c9700534ebe36e61f : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/91cf23433745488ea033017c949ee8c2 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b659e01889e454a93fc0a8c42df6bb7 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad76adeabb474501a82ec6493e46ecf0 : SUCCESS in 9m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d100cf55aad4390adab4c08a4807609 : SUCCESS in 1h 48m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbab6efa750e468e98efc9582d772e9c : SUCCESS in 1h 35m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/28761681beb048ec9190cfa59b957a1a : SUCCESS in 36m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0eb85cad3fa044d287e21633b9fb1b81 : SUCCESS in 1h 10m 48s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/f40a42691c664691a9543c2da4ead7c4 : FAILURE in 27m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fd4a50db0ac24ddf8608328241cadab0 : FAILURE in 29m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cb61f8680bd4939bbcc04de48180363 : SUCCESS in 33m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27cc9025558c4ae6ab9f8977877a1d85 : SUCCESS in 2h 04m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9e6327ea556437b87fed6133ee72d28 : FAILURE in 2h 25m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3413a7ce2e224659a10255f2c26a9211 : FAILURE in 2h 16m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/892ac6bca305498995dca7916994a9e4 : SUCCESS in 1h 44m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1425bea7a39947b1aa6efb8bbb8b6674 : SUCCESS in 1h 49m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e75c02459b874b67960c7c6cec1f479b : FAILURE in 1h 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d60aacd918edbb56303db80bd233d04079516538","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:47:12.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"0b507412f91ffca92f89b5540411772a8bb7630d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 09:28:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f1532ec1e4e434397102af78114a5ff7052af439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:50:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d8853a1027480cbdd52125d899404a : SUCCESS in 13m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29090fd908c34673a20a845eec235691 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5140c210b52746c49481a4e58ecffd2f : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332cdf4740fc453a991934fc7ebe7ebd : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d03e8104be6446b85dd36bdfb6e6498 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f8cfb507966439786cbdccb87cc13f1 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0d4c0ab0cf4a69b0c33d73689f1ff8 : SUCCESS in 15m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1ff8d569fa7542f58790928147b46e8f : SUCCESS in 9m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4c70f6dc80b0467490d3a75b78387872 : SUCCESS in 10m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a08dba65b9624538af9aeb11825144d0 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fcd21144f84845df8d9981f084d7876b : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ecaf0ef1a2746e59b0b9fe2eac0a254 : SUCCESS in 2h 07m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfbf4fc3bbd3436d8dc7921c2359a3a0 : SUCCESS in 1h 53m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11ae2614bb55441086cf94ddafa44ffb : SUCCESS in 38m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1b1119b4a0154737a9080eb16b293cfa : SUCCESS in 46m 09s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/eba42d65286e4f89aed7a4cc95d6e8b4 : SUCCESS in 2h 14m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/304badbb9ea74160a653d3f6fd263c40 : SUCCESS in 49m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8815291f5aff4935b82aa7f61580b5af : SUCCESS in 34m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2b0c25c061a44f17bcd486ac05ecd592 : SUCCESS in 1h 24m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61feec82b5d5472a95a16904b1b55951 : SUCCESS in 1h 50m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ba73dabfbdfd4c7386d5fd02faf0667e : SUCCESS in 1h 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/efd5d5d063054b31977259650010cec3 : SUCCESS in 1h 20m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a3377f3f34f04bf59fe23431cc34be3d : SUCCESS in 1h 16m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/22e7e3b8a8a94baab6b7016768000d6f : SUCCESS in 1h 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"979ca8fac341b78a6a2a98c32e56d8cce3ba1179","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 15:02:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"82c7656e260bc3cf8cac6d157e2fb0e78ce40c37","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-04 19:34:15.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9a28e16263cb00ac20a8ca747cc71d18163c3be7","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-07 12:19:04.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"702badff00b5d6f37a787387ff17a65b2ea10666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 12:20:10.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"7772df0bf14b77b11798e033c6e0cf6c7d376168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 14:41:59.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f64f4bb409b94af3b43d7f7e25f29cb8 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d5783dc97043faaf0c7fe0b5c9e9bc : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abc3f7c4869f461cb3839c94eb4095c4 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcaecf3be6984220ac976d3fba90e5f8 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241663b9dcec4093b1f036472c9fe4e2 : SUCCESS in 10m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fbf0a08f07f04e5cac0819c286d81214 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9e95d525e3249a6bc292ea289154799 : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b71e807fcc364cf4bc4c0fe366615870 : SUCCESS in 2h 02m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f657a5f769b443498434de82eca6c404 : FAILURE in 2h 20m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/913c0a414aa6422aa929002fb6794540 : SUCCESS in 35m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d22007a5d0774dabb87a4575b0266b12 : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":12},{"id":"bea1a6a3284f85a5ae1969364932e7c155903e89","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-07 15:02:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8ddb6373ce04c81c009ba2ca7e6d0dbf5e2a616c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 17:07:50.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/581e7fdd81024e16b6850f15c66ac1e5 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb7a360b1fa64531b51df1011acee5cb : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c15d0c47e34fc0b9ff421cb6296a2d : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b04b2648174044a1b212bb5852e07cd7 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b98a3ce24090452890e7769c219d57ed : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9732b5fd8ead469fb4074d66c5438095 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3bc2ae7752546b7846c60cfc16de804 : SUCCESS in 11m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0d4c6b13268d4aaaa3f838f4dcc5e03b : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/25cdf28931ec435c821217c3774aa1c8 : SUCCESS in 9m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75d3a77bb353448ab861d570f9ec6ac0 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34acf51507d04f3ba29b87831cf55b1f : SUCCESS in 9m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4a18f25b12748a087f6a7a98e5550ab : SUCCESS in 1h 53m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8c2d897da4a49f8b35eeef868624a41 : SUCCESS in 2h 02m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/403b85bbdbd44e52b4ca5a94ba24949e : SUCCESS in 35m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab6ca5617dc947399d1bfd58651825f5 : SUCCESS in 52m 24s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/79adc4fd9715448483a4b3094242a72d : SUCCESS in 2h 01m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bf14b0769836496ea780f03430f6700e : SUCCESS in 49m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa1eebf83f9e47db9f8fa9c91783f7e3 : SUCCESS in 32m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f2e262cf27494285a3d64bfa4928c249 : FAILURE in 1h 27m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/842c178d16c34184b6a527f8349501ba : SUCCESS in 1h 44m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e83a66b3199849e1869d4adc9ac85f8a : SUCCESS in 1h 49m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/55d11df8030347c688237c51aed100d1 : SUCCESS in 1h 36m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a9155596f531415a984da5d88257850b : SUCCESS in 1h 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/456fc5d4c1fa459b9bcccee1de796d11 : SUCCESS in 1h 49m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1667f6c7e0787e773fbe6dfea3a05b670d087afb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 17:10:47.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"4b4c92a69328dc227a6dce5e4f2515a77200ad62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:00:40.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12930b0b76b94e589fb89f2ea557c2ec : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8011a66ae99470cb36e7873ef582b5a : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c782a91bb862459ca384be9eb4b59107 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc5b5947a353484c91f2630fff68245e : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/683c0abf838f4d77924277f38a4645b7 : SUCCESS in 10m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/648f647a6dbf46a9b383ccf91981eb6e : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2799c45292c49638e1297d1767a0735 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72978756ad4345e6b6ca58a91af3c35f : SUCCESS in 1h 45m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/089847fd7fdf420bb8c2d2b86b63b60f : SUCCESS in 1h 30m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f0d88fd9e034a9ca9c718a00791e2a6 : SUCCESS in 35m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d31cccad17143fbbc86b69ba667d7e9 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":12},{"id":"731dfce0cdadf0d379772aa83efbbd4e8b08a941","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:00:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"79aeaa7d90f2e4ab94c310d67ff58e0c34da80cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:02:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d343bf49894ce38f5e677e2913d4b8 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0","revisions":{"0ed66f5344b7ffcc5fd2363e4eaf09d606c7c27c":{"kind":"REWORK","_number":1,"created":"2021-11-02 16:10:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/1"}}},"commit":{"parents":[{"commit":"edf62d061cb931d876315cc43149071def35ee1f","subject":"WIP Switch centos8 jobs to centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edf62d061cb931d876315cc43149071def35ee1f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 16:09:00.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream amphorae","message":"WIP Add support for Centos 9 Stream amphorae\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ed66f5344b7ffcc5fd2363e4eaf09d606c7c27c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ed66f5344b7ffcc5fd2363e4eaf09d606c7c27c"}]},"branch":"refs/heads/master"},"40f1856e321b50da6f739e87bb40dc36dc8f46af":{"kind":"REWORK","_number":2,"created":"2021-12-08 13:57:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/2"}}},"commit":{"parents":[{"commit":"76617f883e535a77008aac80d5350d0f0c889f8d","subject":"Switch centos8 jobs to centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-08 13:56:48.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream amphorae","message":"WIP Add support for Centos 9 Stream amphorae\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40f1856e321b50da6f739e87bb40dc36dc8f46af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40f1856e321b50da6f739e87bb40dc36dc8f46af"}]},"branch":"refs/heads/master"},"a25656b5c79e4a76830457d0fcf941e50806e5c8":{"kind":"REWORK","_number":3,"created":"2021-12-09 15:37:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/3"}}},"commit":{"parents":[{"commit":"76617f883e535a77008aac80d5350d0f0c889f8d","subject":"Switch centos8 jobs to centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-09 15:36:20.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream","message":"WIP Add support for Centos 9 Stream\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a25656b5c79e4a76830457d0fcf941e50806e5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a25656b5c79e4a76830457d0fcf941e50806e5c8"}]},"branch":"refs/heads/master"},"8155025a1c9f9ca455faaf8177e1978023b55253":{"kind":"REWORK","_number":4,"created":"2021-12-09 16:24:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/4"}}},"commit":{"parents":[{"commit":"76617f883e535a77008aac80d5350d0f0c889f8d","subject":"Switch centos8 jobs to centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-09 16:23:59.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream","message":"WIP Add support for Centos 9 Stream\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8155025a1c9f9ca455faaf8177e1978023b55253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8155025a1c9f9ca455faaf8177e1978023b55253"}]},"branch":"refs/heads/master"},"30e638c99b50bebbdd624be86fd8f0045af6b779":{"kind":"REWORK","_number":5,"created":"2021-12-13 09:05:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/5"}}},"commit":{"parents":[{"commit":"76617f883e535a77008aac80d5350d0f0c889f8d","subject":"Switch centos8 jobs to centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-13 09:05:02.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream","message":"WIP Add support for Centos 9 Stream\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30e638c99b50bebbdd624be86fd8f0045af6b779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30e638c99b50bebbdd624be86fd8f0045af6b779"}]},"branch":"refs/heads/master"},"5ba54c9973bcc7c352a38db940619354bc75ba6d":{"kind":"REWORK","_number":6,"created":"2022-01-25 07:27:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/6"}}},"commit":{"parents":[{"commit":"76617f883e535a77008aac80d5350d0f0c889f8d","subject":"Switch centos8 jobs to centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-25 07:26:09.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream","message":"WIP Add support for Centos 9 Stream\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ba54c9973bcc7c352a38db940619354bc75ba6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ba54c9973bcc7c352a38db940619354bc75ba6d"}]},"branch":"refs/heads/master"},"40438b15cf286f19b5ca9791c1f52b78b73bf733":{"kind":"REWORK","_number":7,"created":"2022-01-25 10:01:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/7"}}},"commit":{"parents":[{"commit":"76617f883e535a77008aac80d5350d0f0c889f8d","subject":"Switch centos8 jobs to centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76617f883e535a77008aac80d5350d0f0c889f8d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-25 10:01:48.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream","message":"WIP Add support for Centos 9 Stream\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40438b15cf286f19b5ca9791c1f52b78b73bf733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40438b15cf286f19b5ca9791c1f52b78b73bf733"}]},"branch":"refs/heads/master"},"3713ba57169325904aa5c1dd0b4502c009890557":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-01-27 19:12:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/8"}}},"commit":{"parents":[{"commit":"cbffae6eb684629a19844b5f2f4b7523b77a725e","subject":"Restart rsyslog from cloud-init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbffae6eb684629a19844b5f2f4b7523b77a725e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-27 19:11:24.000000000","tz":60},"subject":"WIP Add support for Centos 9 Stream","message":"WIP Add support for Centos 9 Stream\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3713ba57169325904aa5c1dd0b4502c009890557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3713ba57169325904aa5c1dd0b4502c009890557"}]},"branch":"refs/heads/master"},"7c8bdd7c78f933db5aaefdd63999ed282b492633":{"kind":"REWORK","_number":9,"created":"2022-01-28 06:03:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/9"}}},"commit":{"parents":[{"commit":"cbffae6eb684629a19844b5f2f4b7523b77a725e","subject":"Restart rsyslog from cloud-init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbffae6eb684629a19844b5f2f4b7523b77a725e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-28 05:01:26.000000000","tz":60},"subject":"Add support for Centos 9 Stream","message":"Add support for Centos 9 Stream\n\nUpdate elements for c9s\nAdd a periodic job for building and publishing c9s amphora image.\n\nDepends-On: https://review.opendev.org/816369\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c8bdd7c78f933db5aaefdd63999ed282b492633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c8bdd7c78f933db5aaefdd63999ed282b492633"}]},"branch":"refs/heads/master"},"feeb07bbaf65bdf064db2acc2c24a314df42cdeb":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-02-01 16:35:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/10"}}},"commit":{"parents":[{"commit":"cbffae6eb684629a19844b5f2f4b7523b77a725e","subject":"Restart rsyslog from cloud-init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbffae6eb684629a19844b5f2f4b7523b77a725e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-01 16:35:24.000000000","tz":60},"subject":"Add support for Centos 9 Stream","message":"Add support for Centos 9 Stream\n\nUpdate elements for c9s\nAdd a periodic job for building and publishing c9s amphora image.\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feeb07bbaf65bdf064db2acc2c24a314df42cdeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feeb07bbaf65bdf064db2acc2c24a314df42cdeb"}]},"branch":"refs/heads/master"},"00a3546e13b7433ead53a89aba502385dcf9731f":{"kind":"REWORK","_number":11,"created":"2022-02-01 17:30:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/11"}}},"commit":{"parents":[{"commit":"cbffae6eb684629a19844b5f2f4b7523b77a725e","subject":"Restart rsyslog from cloud-init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbffae6eb684629a19844b5f2f4b7523b77a725e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-01 17:28:55.000000000","tz":60},"subject":"Add support for Centos 9 Stream","message":"Add support for Centos 9 Stream\n\nUpdate elements for c9s\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00a3546e13b7433ead53a89aba502385dcf9731f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00a3546e13b7433ead53a89aba502385dcf9731f"}]},"branch":"refs/heads/master"},"0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0":{"kind":"REWORK","_number":12,"created":"2022-02-04 09:28:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/816370/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/816370/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/816370/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/816370/12"}}},"commit":{"parents":[{"commit":"cbffae6eb684629a19844b5f2f4b7523b77a725e","subject":"Restart rsyslog from cloud-init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbffae6eb684629a19844b5f2f4b7523b77a725e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-04 09:27:52.000000000","tz":60},"subject":"Add support for Centos 9 Stream","message":"Add support for Centos 9 Stream\n\nUpdate elements for c9s\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
