)]}'
{"id":"openstack%2Foctavia~817377","triplet_id":"openstack%2Foctavia~master~I3859c7fcefc89e91fdaecd139143e6a74d8b3c1b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-12 16:50:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3859c7fcefc89e91fdaecd139143e6a74d8b3c1b","subject":"Fix serialization of lists in data_models.to_dict","status":"MERGED","created":"2021-11-10 10:02:34.000000000","updated":"2022-09-12 16:51:28.000000000","submitted":"2022-09-12 16:50:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817377","meta_rev_id":"0d40fca99c9cfea0e0bec42e0d859ea054f95a61","_number":817377,"virtual_id_number":817377,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 16:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-11-15 12:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-12-21 01:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2022-03-07 11:10:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 14:31:53.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 14:33:32.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 14:33:32.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 12:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 12:01:13.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-12-21 01:24:54.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2022-03-03 09:27:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-03-07 11:10:28.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"1157cb91cc9f3e811461f0b1768538341671ce4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-10 10:02:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd10ce5aac286b471ed7ad8c774c48cc78fd3e72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 12:14:19.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ad4fd2cf0d6404fb82fcf47d1f2e870 : SUCCESS in 12m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba6177a2859f422e9ca850f652c41ed1 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac786a2b09e408cb09fa23969a76d40 : FAILURE in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bbb1b8475bc45e195744ad4b23df2e2 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50d3a8a9784a433793bece2e87d3c37c : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02b484ef2b9f4e23a1a2e6217a017ca0 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196b150c285b41b39035aa6b5348b709 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7a33e3067594469a755577aee5b73e0 : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30c9d45dd8c74b189d50ef4418da0429 : SUCCESS in 7m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68050ee3bf744582870321cb00eaffde : SUCCESS in 10m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bac6bb01cb0748418f367a2f8e9b32ad : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da882eaa42034730b36b71944a9f96a6 : SUCCESS in 9m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64fd50b09f884fc08f04052551e6c410 : SUCCESS in 2h 05m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e7f7f36863748c1a55cb5982b4cb363 : SUCCESS in 2h 04m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7dc981de428e45eabcb507550c4c7562 : SUCCESS in 39m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b46599c7ad7a4dc78bced7ce3e670595 : SUCCESS in 47m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/995e2746d00844d8b4e3650c12b14805 : SUCCESS in 35m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c3be9283b70b4d7c927c97adb8b7f6be : FAILURE in 1h 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c8bc4f2c03f40f893fe5a9cf2950748 : SUCCESS in 1h 47m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b1fbd7edab1c415ba2e9e113a14e5bb8 : SUCCESS in 1h 37m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/67307b2a80354c74a028ddd1c69fa2c9 : SUCCESS in 1h 30m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/957d27208ede4b409a9777e82c1d577c : SUCCESS in 1h 33m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0b761354206c443490ea56226b1e762e : FAILURE in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0b2e9cbb5e1773caeea9e759f3aaf5437f21ff0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-10 12:40:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37cfbd3e60b6dbeea80cf126d3f400ac8402d163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 18:40:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10720cdc92384bbc8ff821d0ff5d12f5 : SUCCESS in 12m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82b895d02e86497f9fcd1ca258a8c1f9 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1471b08fb4a484fb7e3cd3b202a4e70 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/224087d75a1946f385f1ac4acdedd18a : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c469965860d4961a52f38489ac7bcde : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bd383aaf259407d80c96f453ab75d06 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5efe38942ff47e79a6c15c53a176148 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54c9947020cd40f68a5001adb3fa0346 : SUCCESS in 7m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3265ba4a6ed44380a61fc81ccee51b5b : SUCCESS in 8m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0c4f79973fb0478b91091860862a9fc4 : SUCCESS in 11m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7434977242c742d294bff7c83dad1e30 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79f6a465798b4e95aac59436754f0b04 : SUCCESS in 10m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e7a01ec594a1479b8b6d9eb5af2b557c : SUCCESS in 3h 03m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fa935d4cf2846b2a20fa6fe6d3d73ab : SUCCESS in 1h 27m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e56ad2a6fd00456bb311310a4d93d084 : SUCCESS in 38m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/df6a76e2fbf34b059913cc816a5b0738 : SUCCESS in 1h 23m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdee87c79c944f929b7bbecd6830b6a8 : SUCCESS in 1h 10m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/469aa63529ec412bb7e3cd16f36a37e3 : SUCCESS in 1h 25m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cfc7cbb677e8438887e6a429e701c75b : FAILURE in 1h 54m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/15f68d04e1f74b0bb6442eadad42ef97 : SUCCESS in 1h 55m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f4b6e1184524ff1be9f9300174e5c0b : SUCCESS in 2h 25m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5356f7e7099643a0bfe88a0fce24ee49 : SUCCESS in 1h 44m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1e9213594fd0477db59a547a59581352 : FAILURE in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d886dab236b8503dfa589f7dd94ad44934d7b26a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-11-15 12:01:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2a90776d81a9b987f9542e08794c421154127013","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-21 01:24:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"748e183c9dd672560a0820c8fb6834a4a617a138","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-03 09:27:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"88f04d8fd2503458b9c6dcf8103a136d96fa86da","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-03-07 11:10:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8e9b30f3b9324c694f53413cd2bf7da305ab81ad","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 11:47:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2c3dcc2628c6989c413201804ebb37ecf785608","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 14:31:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c00efbd519a23256c124955cf5b8b549a912e181","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-12 14:33:32.000000000","message":"Patch Set 2: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b0468bb1703c2f5f0f0ed321dda21ed42ddc7dc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 14:34:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80b6362e03d94a30c1ac01db27297e16959a4be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 16:50:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e7630386ce64faea82d66ea86584a3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c69a4148d0e4e9db4f768ee9c402126 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bb9d6e0d6ba454d8bc969f8f77123b1 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9d962a61bb94a5cbecda7439a1f8a47 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d9e7ffe4794f9ca8ca172087be65e6 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3f70a8432354e37bee0034d58c37ed6 : SUCCESS in 6m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29e8f1f9fc884d4e8a32e8451957ee8f : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/88bae2e6731e49e49b5b88ab45e3a018 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e953941381bd4c06afc2b575498d228c : SUCCESS in 2h 09m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbfe344b0ac647c9a7800846cfa9b2c4 : SUCCESS in 1h 29m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49b340a177984e9381b77cd7b7711849 : SUCCESS in 45m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e17e7d084d4e4b1ba8f0011cc0e0d0d8 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":2},{"id":"17117627441a7cbb274398f01b6376b50bea2bae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 16:50:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0d40fca99c9cfea0e0bec42e0d859ea054f95a61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 16:51:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e273ca6e424fbaa6b2f4f70269ba6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd3541ad95d4cdf8840149893b197cb : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17ebcc2a261442e39c0d17dc5f2e9259 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d25ed7aba6753285671e4f3b8c478c0e5e022e10","revisions":{"0d6ea3a3222357a7e0d8ffebde22cf32263bdf87":{"kind":"REWORK","_number":1,"created":"2021-11-10 10:02:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/817377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/817377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/817377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/817377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/817377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/817377/1"}}},"commit":{"parents":[{"commit":"81134bdfec79c8019c26e7846d31a2b2952965a0","subject":"Fix failover of az-specific loadbalancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81134bdfec79c8019c26e7846d31a2b2952965a0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-05 16:08:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-10 09:54:31.000000000","tz":60},"subject":"Fix serialization of lists in data_models.to_dict","message":"Fix serialization of lists in data_models.to_dict\n\ndata_models.to_dict didn\u0027t convert values inside lists.\nWhen serializing TLSContainer, the intermediates list that contains\nbytes types wasn\u0027t converted properly.\n\nto_dict now converts the items of a list recursively.\n\nAdd a test that checks if TLSContainer.to_dict is serializable.\n\nStory 2009310\nTask 43699\n\nChange-Id: I3859c7fcefc89e91fdaecd139143e6a74d8b3c1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d6ea3a3222357a7e0d8ffebde22cf32263bdf87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d6ea3a3222357a7e0d8ffebde22cf32263bdf87"}]},"branch":"refs/heads/master"},"d25ed7aba6753285671e4f3b8c478c0e5e022e10":{"kind":"REWORK","_number":2,"created":"2021-11-10 12:40:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/817377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/817377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/817377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/817377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/817377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/817377/2"}}},"commit":{"parents":[{"commit":"81134bdfec79c8019c26e7846d31a2b2952965a0","subject":"Fix failover of az-specific loadbalancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81134bdfec79c8019c26e7846d31a2b2952965a0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-05 16:08:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-10 12:40:33.000000000","tz":60},"subject":"Fix serialization of lists in data_models.to_dict","message":"Fix serialization of lists in data_models.to_dict\n\ndata_models.to_dict didn\u0027t convert values inside lists.\nWhen serializing TLSContainer, the intermediates list that contains\nbytes types wasn\u0027t converted properly.\n\nto_dict now converts the items of a list recursively.\n\nAdd a test that checks if TLSContainer.to_dict is serializable.\n\nStory 2009310\nTask 43699\n\nChange-Id: I3859c7fcefc89e91fdaecd139143e6a74d8b3c1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d25ed7aba6753285671e4f3b8c478c0e5e022e10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d25ed7aba6753285671e4f3b8c478c0e5e022e10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
