)]}'
{"id":"openstack%2Foctavia~818088","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I48b0f5a773209b1c1b056d71c0da05d6fd82ca73","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-12-02 11:41:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I48b0f5a773209b1c1b056d71c0da05d6fd82ca73","subject":"Fix LB set in ERROR too early in the revert flow","status":"MERGED","created":"2021-11-16 13:54:55.000000000","updated":"2021-12-02 11:43:38.000000000","submitted":"2021-12-02 11:41:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818088","meta_rev_id":"6e30353d5f6de52ca114c3f3d823199bca0e832b","_number":818088,"virtual_id_number":818088,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 11:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-12-02 09:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-01 22:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-12-02 09:12:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 16:20:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-01 22:05:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-12-02 09:12:56.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"e8e389c0ae6634bb33a1c8d8ef6a270ca914f5ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-16 13:54:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fabad513a2235e833c71fa5ccf3db01e8c1cf9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 16:20:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71b43bf523fe413f9d8ac48585202d4a : FAILURE in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3764ec1de23a44acbed8ddba29e2346a : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fd9f5eef8d347babe039860f7a76d9f : FAILURE in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a11d80b95c0745afa2cf47d902c72e16 : FAILURE in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4605ba811a2a4f94b99c3b8d0ca72033 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4975b8eb363f4d81a588d62c7766bd3e : SUCCESS in 8m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/169285f2c04a4edd8443fdf46f776ec7 : FAILURE in 7m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6930cbdcc3f54b16af5a758b51f3dd35 : SUCCESS in 10m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef04fcebae1c40a6b483cfed3a83fc90 : SUCCESS in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ecdeacbb17de4861b44aa73aff67c385 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e6df8be679743c189a278013c2a8632 : SUCCESS in 1h 38m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/357bce2b16154fb595eb621cf4c7f828 : SUCCESS in 1h 45m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19708f7453dc484cbee8c8b5b1b3fa8c : SUCCESS in 35m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e3b73d144cb644ae9ab7abac06bd2802 : SUCCESS in 34m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4b4d2f876d24d9ebf33effed9516319 : SUCCESS in 51m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/df6ef9b2f4084c04b4d82a1369eb547d : SUCCESS in 48m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e0388ad34d1a42359485b9e525da1149 : SUCCESS in 1h 30m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/715c46cf371949529463a25907725f74 : FAILURE in 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac31c06f663743b1bd6eb13ce2a70b3c : SUCCESS in 1h 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f5228c9b02844323adce64388c07b6a4 : FAILURE in 1h 28m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d757adfd2ec44783a4a2bee59e6a53a2 : FAILURE in 1h 41m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9d8b270e274047c0aed43bf60b87bff4 : SUCCESS in 2h 08m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"906b745ebff7f96350921df3526e5e4ad51e3079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-17 06:51:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20e906e5dcb61f16140b5f58253b261b5ef461ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 09:16:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c18e5a46b1b4488b820539e7e02690c : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a79a1fdcea247b38b4b14d38d420765 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52b45019ee6841de8af2956f1c184710 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cdac7a4cf5245049b3ceb7ff4f52fc4 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5371f12d261f4fb4ad4fa27d3f925be1 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4869a6ef677946e1bc53e661d9b42c6d : SUCCESS in 5m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b54faf8a738a4e9680a8aaa5e5931ac2 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6a1eaa9012544ec18cb9d85f8f541c36 : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e163c17ed5534967828585a85113a202 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c1c217482ef4b62bd3a15eb0fb204d6 : SUCCESS in 9m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/32381d40f8de4e759d4466fb277fb4b7 : SUCCESS in 1h 38m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fcfd86751514e2fa56b0ddb308c30bc : SUCCESS in 1h 59m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e1d7d16564348f0a68684d1e97edb07 : SUCCESS in 44m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e70f18c9a7e24dec88b356831275d6be : SUCCESS in 43m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cca30951513d4d51a724df4d2477eb90 : SUCCESS in 36m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/023eebe809d3468484d51aac3b13f858 : SUCCESS in 1h 21m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a574eeb259b4eccaa758275d3214917 : SUCCESS in 1h 45m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b6bc66d1b7114155be5f8d5d90e3b0e5 : FAILURE in 31m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/33bf31fe13f34e0a8fe5e47d18de30de : SUCCESS in 2h 21m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4e9c6c9dac5b42619ee7635d43377308 : FAILURE in 1h 31m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/52bc7c1372e046dc8afcd414d8f09168 : FAILURE in 1h 57m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4640d1ecad784ace978fa52d980decb4 : SUCCESS in 1h 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fde6d00cebf80006eeeb6cb05247a2407cdf61c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-01 22:05:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64051fb54274aa739c4c0a0f532354514a22418e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-12-02 09:12:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15b2ff1814fb3b51b124a73a0318682f6e2f9b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 09:17:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36856e2342797f50e63046c7a769641d0829cb0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 11:41:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3df3403e4e34b48bafd0a8244023dfc : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49199f40905d44a382fd93722f64fb4a : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0941a8f930b147da869246ef95ce6695 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40f2f4eab46469f8e4314418a9bda3b : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/950ec6945253435b85d6104d89d7e24f : SUCCESS in 5m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/462978cfe2664309b94227f818c31f1b : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d08e153eb8eb49fc8a1a6f39b2a88824 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84c0aaba978e4edcae593dd934ccb57d : SUCCESS in 1h 05m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b520dc168134094b3687a79d655ab02 : SUCCESS in 1h 29m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/333a7adfad9e47a685bb3829acc48534 : SUCCESS in 33m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e93ce2fcb0044d7a8f601062c41e801 : SUCCESS in 33m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c94c97d5fbd54e719bb43ceb60fdebcb : SUCCESS in 34m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a4e8604a5d0548b4957581641f560696 : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7d59e090943d93149b5cd61457d8e07720843e9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 11:41:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6e30353d5f6de52ca114c3f3d823199bca0e832b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 11:43:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8447f2b4ba7b4dd7ad08ebc4a9da847c : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/154266a46f6e40f1af2faa068db3ed72 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"17f2c674916584d70ba11fd2d8ba3bae439d5b62","revisions":{"cf2a8bdf88d544e47d2aa2e43388aeb3de99e21b":{"kind":"REWORK","_number":1,"created":"2021-11-16 13:54:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/818088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/818088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/818088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/818088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/818088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/818088/1"}}},"commit":{"parents":[{"commit":"e3026d35828b0b4541de3415e22582fa523f78e3","subject":"Fix failover of az-specific loadbalancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3026d35828b0b4541de3415e22582fa523f78e3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-28 20:13:26.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-16 13:54:21.000000000","tz":60},"subject":"Fix LB set in ERROR too early in the revert flow","message":"Fix LB set in ERROR too early in the revert flow\n\nWhen an error occurs in a flow, the provisioning status of the load\nbalancer should be set to ERROR in the revert method of the first task\nof the flow. This update acts as an unlock of the LB object and cannot\noccur in any other revert method because the API might consider the LB\nas mutable before finishing a task/flow.\n\nRemove all occurrences of mark_loadbalancer_prov_status_error calls in\nrevert method of tasks that are not specifically designed for unlocking\nthe load balancers. Add a LoadBalancerToErrorOnRevertTask task in the\namphora failover flow to prevent a LB to be in an immutable state.\n\nStory 2009651\nTask 43810\nStory 2009652\nTask 43811\n\nChange-Id: I48b0f5a773209b1c1b056d71c0da05d6fd82ca73\n(cherry picked from commit 4b8b198fecee802e1031c014f811eb766329a1a0)\n(cherry picked from commit 4039d35ce221b58490d4aa5b3497e29dbf7664ac)\n(cherry picked from commit 844f1348ea3d83f9fb5f920b64c9234dbdf9948d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf2a8bdf88d544e47d2aa2e43388aeb3de99e21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf2a8bdf88d544e47d2aa2e43388aeb3de99e21b"}]},"branch":"refs/heads/stable/victoria"},"17f2c674916584d70ba11fd2d8ba3bae439d5b62":{"kind":"REWORK","_number":2,"created":"2021-11-17 06:51:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/818088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/818088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/818088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/818088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/818088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/818088/2"}}},"commit":{"parents":[{"commit":"e3026d35828b0b4541de3415e22582fa523f78e3","subject":"Fix failover of az-specific loadbalancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3026d35828b0b4541de3415e22582fa523f78e3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-28 20:13:26.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-17 06:51:08.000000000","tz":60},"subject":"Fix LB set in ERROR too early in the revert flow","message":"Fix LB set in ERROR too early in the revert flow\n\nWhen an error occurs in a flow, the provisioning status of the load\nbalancer should be set to ERROR in the revert method of the first task\nof the flow. This update acts as an unlock of the LB object and cannot\noccur in any other revert method because the API might consider the LB\nas mutable before finishing a task/flow.\n\nRemove all occurrences of mark_loadbalancer_prov_status_error calls in\nrevert method of tasks that are not specifically designed for unlocking\nthe load balancers. Add a LoadBalancerToErrorOnRevertTask task in the\namphora failover flow to prevent a LB to be in an immutable state.\n\nStory 2009651\nTask 43810\nStory 2009652\nTask 43811\n\nChange-Id: I48b0f5a773209b1c1b056d71c0da05d6fd82ca73\n(cherry picked from commit 4b8b198fecee802e1031c014f811eb766329a1a0)\n(cherry picked from commit 4039d35ce221b58490d4aa5b3497e29dbf7664ac)\n(cherry picked from commit 844f1348ea3d83f9fb5f920b64c9234dbdf9948d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17f2c674916584d70ba11fd2d8ba3bae439d5b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17f2c674916584d70ba11fd2d8ba3bae439d5b62"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
