)]}'
{"id":"openstack%2Foctavia~818236","triplet_id":"openstack%2Foctavia~master~I59664615e1ef5e46a99c12256dbe02f89ecf4c0b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-26 18:23:44.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-02-22 18:02:27.000000000","reason":"Michael Johnson replied on the change"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-07-26 18:23:44.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-03-14 13:36:00.000000000","reason":"Tom Weininger replied on the change"}},"hashtags":[],"change_id":"I59664615e1ef5e46a99c12256dbe02f89ecf4c0b","subject":"Fix update listener certs doesn\u0027t work","status":"MERGED","created":"2021-11-17 14:56:06.000000000","updated":"2022-07-26 18:25:09.000000000","submitted":"2022-07-26 18:23:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818236","meta_rev_id":"2e8ec54f66c38e5d0844f22d46b97e21db5a2fa0","_number":818236,"virtual_id_number":818236,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 18:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-07-26 15:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-11 16:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-03-14 13:36:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-07-26 15:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-11 16:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-17 15:44:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-11-17 17:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 08:53:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-03-01 14:31:47.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-07-26 15:36:43.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"9fb3a72d7be39d871763fee161822a6b51c398db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-11-17 14:56:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6adba0536fc14b3196e79efebb0bd46488610cb4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-17 15:44:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"55fda38afb20f73de423ba502346eac7b356c52e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 17:35:03.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c835e0356124490b7f5382d5720b178 : SUCCESS in 29m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c7305ab051142ce8180de7ddcb82afe : SUCCESS in 26m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a956542c4b554820a3c2951e4e58f257 : FAILURE in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac0f557c4914bdaa83aa9e12d47d89a : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9434c848dbad4648a701693969e4e2e2 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b15428622c24987aac52ff437716c54 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b7da7483004ea0920d5bc0954168be : SUCCESS in 19m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8289adb4d2c041838db0cccdd7a4e4db : SUCCESS in 8m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/742a7d360c5c4f68893495c9bcb56ccd : TIMED_OUT in 32m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/073201b0b5cb4024b527174480cb8fd0 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d779ebeb49554bd0bfb1382a7341644a : SUCCESS in 31m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0e8b4a2c0024e1ca9dd2f70da3715c6 : SUCCESS in 2h 12m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1ded711efdf4350a062ca6f82f874c9 : SUCCESS in 1h 38m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52a6962868ab4389b10b6be1ce813a6b : SUCCESS in 40m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bed9a1bc21b34784b794719e3bba93d2 : SUCCESS in 54m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f715e86c3f5473ba619be55b71c4911 : SUCCESS in 39m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a46c0523d024a59977c934cd85bf900 : SUCCESS in 1h 33m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c8e923260e0446ca7110d898cf7db1e : SUCCESS in 2h 03m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97c857a7c356481680cc6cb22dacda3a : SUCCESS in 1h 52m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5bcde6f461804ee09e8b0ac764e8a0f7 : FAILURE in 2h 16m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/eb9c76d3f531453c9d9a450acb5f1730 : SUCCESS in 1h 39m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8316064a1ef84e23ac297fa340eb9121 : FAILURE in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89011664f945636df1b3cfe49652787ac39490a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-11-18 11:55:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8724bac8032ec757303a0c17fdbd7054c854cca6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-11-18 11:59:09.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7b951bddbb9a59e0e665b66d3a9ce50720bc58f3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-11-18 12:01:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e056d5d8ce75e61ff3043bbd839752141a3e2c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 14:25:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d38beaab17404368bbde22782d6cd35a : SUCCESS in 31m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f8c0118160047d3a3d401950d9a83ea : SUCCESS in 27m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9346298faa144ba8a299d0dbf838e33 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c57493b8d8340cabb61f51477d52cbe : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cfd0800fa894f72ade7841f6dd52cdd : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b7fe21714b542a98981f3c4f8d99b7a : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628287e5f8ab4e4aa6acbacf1a36337d : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b72d5eeddea4c1bbc2043fec68347bc : SUCCESS in 5m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f9e900e041ea4f9888b7ddf7fb6e9f0b : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cf75d98b1394cf4a6dafbb56161814b : SUCCESS in 30m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b698a3205a7c4c4e9be3b43ada3cdc10 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4511fc58e8424cc2a75ffbc64adb8b46 : SUCCESS in 27m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f160cf2726f94503b5656f8973cdfe5a : SUCCESS in 2h 18m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19a6126c1c4f42c1920ec384f71c9776 : FAILURE in 31m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b12ab105a2c54a8487bb17de0010be8a : FAILURE in 36m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/73e3250e29b14fc4ae36d3a56857635c : SUCCESS in 1h 08m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7faadfd0b73c45ffa77e8bad927a447b : FAILURE in 31m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/88e48cca3de04c1aa8dc1f48573059d4 : FAILURE in 1h 06m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56edeed663dd44afba15d727eaa47ee1 : FAILURE in 47m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d8386963cf334df9b9dccde48422bdf5 : FAILURE in 30m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/80a7f586b0ba48a2974d622d9ac11066 : FAILURE in 29m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dcb6ebd3a7b041f2bc8b7d9e5d529523 : FAILURE in 34m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ac687bf9ea964b9d8a81e56bd78636a7 : FAILURE in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6960a37dc283d86d36d313f553dc5269ca34878f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-11-19 12:09:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12dd6f04d9ff54fa97c8c7edd327c4ac27ff9eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 14:54:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/414c07ded1464709ad79af3ad5dd8f2f : SUCCESS in 47m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6ddf2315fdd4b7cba58f92c901659bc : TIMED_OUT in 41m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e928dfee465242c68ba720eba2cf5414 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8dc8dcf92b446ed86761886e7201eaa : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f21c5da926d64b0e844b1d7ef9b44804 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2b624c955034981a93eca8b81a6d2f6 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9db6a4ac804ed19bc5c687b2803db9 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9690ebf104945218554a2b9669cda22 : SUCCESS in 5m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cc69b4d862d84b699e18d17d323fc853 : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4e5e6df7e3704256af66c2f536274e21 : TIMED_OUT in 31m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a771708cd61e48ff9681b08e3fcedf12 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe1f89fb56cf448cbd9439805ddaafe5 : TIMED_OUT in 31m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5037c382dbc4411189d87fcb8f5f2a2d : SUCCESS in 2h 36m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb85831da39d4527afd93b3c27c9d621 : FAILURE in 41m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/975a787c0b544d12baf4e93db2d98288 : FAILURE in 36m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af008ee68f8d41ebb89d484bd973a196 : SUCCESS in 1h 00m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64e657e5b7624c938dd60c78d74c5b94 : FAILURE in 40m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/97562e3cfe1d4a06862b00d70d1a4b4f : FAILURE in 35m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd123e9f9b8547909c1c4d4ed8bcca84 : FAILURE in 59m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6249bfda8e0141488bcf0ecebf1ae141 : FAILURE in 36m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f3becea4dea746f88ba8182bf7c2cbd2 : FAILURE in 35m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/27d90da765eb4481a3b808b9359b7f86 : FAILURE in 46m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/920309bf626a4a6d978ad01ef095bc80 : FAILURE in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"26568da19588378e521bb2836b5c5f5e59776d53","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-12-09 12:48:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"88b06d97c71964e376cd502a1da259c6c5c44356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 15:23:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2ff0998a78f453facfbc668efd8cebd : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a13c371656e44ec89fb211cd7065893 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edcc98067d04fe49cc0f48bc685c500 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebe4e5ee53394c8db26d583353474d0f : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab7f77d42ef4ea19fd9265537ffd5a5 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80633d759a3149ce8aca5fc42ed849b5 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6ef5947cca4f3395a03012d34552a4 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f924813c63b42c9a1cffb6fc6e35e13 : SUCCESS in 5m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eda264d65c384abd9a919aff4fd8da29 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6d7bca75d2ae4d27a05cee0637139e92 : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/21df817ca54641b08e278407741be58e : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e178423034c4f74a01e28c37c1fc8a8 : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6d7ec7460f042fc89995797183af986 : SUCCESS in 2h 03m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19b49c21f3714fcd9fdceb9cfc672d86 : SUCCESS in 2h 04m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3fc542d8535a45a49915a562ef90b916 : SUCCESS in 46m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/15b3ced5a1ab4259866a58c91f1e23e0 : SUCCESS in 1h 10m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8685c0bc4a6e41b7b46f8e1bb0d59d92 : SUCCESS in 33m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c4418fe9073741a4912a70b0028bc5e5 : SUCCESS in 2h 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/540eb47751894849913306e8dedf8057 : SUCCESS in 1h 38m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ff1eb68ddd13421d839dba8aaddd4ed8 : SUCCESS in 2h 07m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/93547600d3464c99a15732ad413b3d97 : SUCCESS in 2h 03m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/34b6912f05714203bcd023e3f0e9cf40 : SUCCESS in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/da44fad8ef0646219b4df711ff492bf5 : FAILURE in 7m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"594caf0e5496ebc97e71ba4820782aeb51561099","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-22 18:02:27.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da80ff9f7826d0e0b56e5c220625305aa3be99ff","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-01 14:31:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32e101f8505f2cd682a3b31fc5bfbf76d9a61bf4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-03-11 02:12:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8dc2c395b3be1a16ab41f996364b52ca6a2be245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-03-14 12:25:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fb2cb0713bab5c96d42476e987d70fbaa323ef1f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-14 13:36:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"95505e8583a50eb71c5a26e0c4dddbc2cbb529ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 14:39:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01c91ac628df43259f4622df7680b1c0 : SUCCESS in 16m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/762d21b08a7a49b5a9b3dcb9fda2bc74 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ccea71d9a245b495307fe779ab7403 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8491138fb49f44eebc4035a96648952c : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a365652c7b3c4376bae728a74c6ca85e : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3412b9c918945cd99e2d19c607ab4a5 : SUCCESS in 16m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66e3515a5fac4be99b781ca942c5ed52 : SUCCESS in 6m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9a2655bd63464f09a46d5b3fe89b17eb : SUCCESS in 8m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b99b92ab21724ffc91a1ebad3022b4b1 : SUCCESS in 10m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b9db197d60e44cfa8532cebb47fdf88b : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27530e7be9fd4d41b109d8d395ad338c : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e229a9ccbba41ff879da81fc8d41324 : SUCCESS in 2h 07m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e3a636389574b7e89a9c86c3562fc7c : SUCCESS in 1h 23m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fcafb93091884bc8acfadee8668fb34c : SUCCESS in 38m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/be8200bbd12c45a6bbd980bf0a697bfa : SUCCESS in 49m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/93cc1cd29c80427683f4a02a9038fc64 : SUCCESS in 54m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f6e0eb32ad54f38901583765d186106 : SUCCESS in 38m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e360c04e66bd434880a0ac48747bd84e : SUCCESS in 1h 30m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5bc97b134ff947e6af97ff8db0cb9cc7 : SUCCESS in 1h 46m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8720254fc323445b9a6fce13bf09c9eb : SUCCESS in 1h 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f2bb02a50084d609ff12e5f2842b6ab : SUCCESS in 1h 43m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/92df17d199fc4602a2d42df862bac1db : SUCCESS in 1h 25m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c74abf1ebc84f30bc81cea341369c17 : SUCCESS in 1h 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef3b112f33d71ea5c4ba8dc5d9a1ef77aa1c9665","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-11 16:43:52.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c18e64e13c2b7093bf36b4e6d93c184779dc0ec","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-07-26 15:36:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"778b3e0c01e149d4475f8c457c1b416798c63254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:37:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b7846ea04f6be1196082d8c5c56cf83680989708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 18:23:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d974a778f74b6c829ee4b8edd17eb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fad67f8be5747138c02ec451f36c2e9 : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b045a18dbe45cfb20e3c2662477389 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8364cbee06b4628b3e58fd8270ea87e : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f53df91c82642a7a13785635f2f3bcd : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b8e200ff3cc4b79b692df64484948a7 : SUCCESS in 6m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0c27ca87b864d77af2b291dec04d797 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92fc3fd2941246f68b24dae1f5130589 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3976c75d3971427da0faff7bdab77eee : SUCCESS in 2h 14m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f28c24fb8424f53a275523c75fcba16 : SUCCESS in 2h 00m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ff08b264bdd46c3b3f81477c5cf2be5 : SUCCESS in 41m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1c8263d7f6f24f088d8516691eaf80d7 : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":4},{"id":"eaa646d86c4ebf85ba13591a44151c155a6f62d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 18:23:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2e8ec54f66c38e5d0844f22d46b97e21db5a2fa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 18:25:09.000000000","message":"Patch Set 4:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1673474b9c34755b1191d37206b5431\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edd1d5eca9034a0090c89ce05b5f1907 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21f3cda815ab4d038edc907278fd28a1 : FAILURE in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d340a1de50b55a73f0e3a4d398ee735b0ccc128b","revisions":{"0d3aafb9ccf2c2bead2ceae46f2e0e88e200e88c":{"kind":"REWORK","_number":1,"created":"2021-11-17 14:56:06.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/36/818236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/818236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/818236/1"}}},"commit":{"parents":[{"commit":"f48141f43b4856b88092335daf04301add9f7250","subject":"Merge \"More logging during load balancer creation.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f48141f43b4856b88092335daf04301add9f7250"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-17 14:43:42.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-17 14:55:21.000000000","tz":480},"subject":"Delete crt-list file before updating it","message":"Delete crt-list file before updating it\n\nwhen the destination pem-list file\u0027s md5 is not match the expected\nmd5, we need delete it before updating it.\nfor example:\nwe have a listener with 3 sni-container-refs, the pem-list file\nshould like below:\ncert1\\n cert2\\n cert3\nafter we update to 2 sni-container-refs:\nactually:\ncert4\\n cert5 \\n cert3\nexpected:\ncert4\\n cert5\nThis patch also fix the unset sni-container-refs doesn\u0027t work.\n\nChange-Id: I59664615e1ef5e46a99c12256dbe02f89ecf4c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d3aafb9ccf2c2bead2ceae46f2e0e88e200e88c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d3aafb9ccf2c2bead2ceae46f2e0e88e200e88c"}]},"branch":"refs/heads/master"},"583de203e7c9da0713d567a5067804b028829e92":{"kind":"REWORK","_number":2,"created":"2021-11-18 11:55:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/36/818236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/818236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/818236/2"}}},"commit":{"parents":[{"commit":"f48141f43b4856b88092335daf04301add9f7250","subject":"Merge \"More logging during load balancer creation.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f48141f43b4856b88092335daf04301add9f7250"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-17 14:43:42.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-18 11:51:52.000000000","tz":480},"subject":"Fix update listener cret doesn\u0027t work","message":"Fix update listener cret doesn\u0027t work\n\nThis change fixes listener update sni-container-refs\ndoesn\u0027t work by adding O_TRUNC flag to upload_certificate method\n\nChange-Id: I59664615e1ef5e46a99c12256dbe02f89ecf4c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/583de203e7c9da0713d567a5067804b028829e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/583de203e7c9da0713d567a5067804b028829e92"}]},"branch":"refs/heads/master"},"9ac5c71cec97e92703af41910229417086df3011":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-18 11:59:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/36/818236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/818236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/818236/3"}}},"commit":{"parents":[{"commit":"f48141f43b4856b88092335daf04301add9f7250","subject":"Merge \"More logging during load balancer creation.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f48141f43b4856b88092335daf04301add9f7250"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-17 14:43:42.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-18 11:58:39.000000000","tz":480},"subject":"Fix update listener certs doesn\u0027t work","message":"Fix update listener certs doesn\u0027t work\n\nThis change fixes listener update sni-container-refs\ndoesn\u0027t work by adding O_TRUNC flag to upload_certificate method\n\nChange-Id: I59664615e1ef5e46a99c12256dbe02f89ecf4c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ac5c71cec97e92703af41910229417086df3011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ac5c71cec97e92703af41910229417086df3011"}]},"branch":"refs/heads/master"},"d340a1de50b55a73f0e3a4d398ee735b0ccc128b":{"kind":"REWORK","_number":4,"created":"2022-03-14 12:25:51.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/36/818236/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/818236/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/818236/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/818236/4"}}},"commit":{"parents":[{"commit":"3c5bbbc189efe9f3ed4caae04a23d7a13718d368","subject":"Merge \"Fix Amphora RSyslog configuration for TCP failover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c5bbbc189efe9f3ed4caae04a23d7a13718d368"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-17 14:43:42.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-03-14 12:25:47.000000000","tz":480},"subject":"Fix update listener certs doesn\u0027t work","message":"Fix update listener certs doesn\u0027t work\n\nThis change fixes listener update sni-container-refs\ndoesn\u0027t work by adding O_TRUNC flag to upload_certificate method\n\nChange-Id: I59664615e1ef5e46a99c12256dbe02f89ecf4c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d340a1de50b55a73f0e3a4d398ee735b0ccc128b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d340a1de50b55a73f0e3a4d398ee735b0ccc128b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[]}
