)]}'
{"id":"openstack%2Foctavia~819265","triplet_id":"openstack%2Foctavia~master~Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6","project":"openstack/octavia","branch":"master","topic":"nft","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-12-06 09:13:38.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6","subject":"Fix nft command line with negative priority values","status":"MERGED","created":"2021-11-25 08:15:34.000000000","updated":"2021-12-06 13:50:50.000000000","submitted":"2021-12-06 13:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819265-nft","meta_rev_id":"31663cae858ad712d3de2ffd8a75edaa1e333e80","_number":819265,"virtual_id_number":819265,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-06 13:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-12-02 09:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-11-25 15:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-12-02 09:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 10:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 15:53:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-12-02 09:13:43.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"8283de79f8bb6a86510ee82364e969ea096a49a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-25 08:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9063c8d9f6a34ffcd2e0d6e7e3504239f07fdff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 10:22:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aa0706c97f84585bf11dae5956157a4 : SUCCESS in 11m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/706335deba084636a36f3af0c5e996f9 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86e29c7d4ed4436c939a669642a65daf : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48bc248211754a2997aa04491fa9e703 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee4fb3c110a49f1b117417aea83e97a : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d5a796636c2487a9397ed69fe1e4003 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7297c6d1d6840ed8877bf9642573175 : SUCCESS in 12m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fafc920d9f52496d80c05d028a2e3653 : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/415417ead7f54dce8df9f84c2c3643bd : SUCCESS in 9m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bf732eaf42104156ae659e59bef7f610 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95073da97312488db016fce64d318364 : SUCCESS in 9m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/346438495c8743a5ac8489fd690f5f1c : SUCCESS in 2h 05m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c71cd1a2ee547cfa88f31e005fdaa9b : SUCCESS in 1h 30m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/097363452fff44d3a2ae2c8d31e8bac6 : SUCCESS in 33m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50dc4628697140b8a7611925236e701f : SUCCESS in 46m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59ac35c39e354dc4a2cf4a70cf82a680 : SUCCESS in 35m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94cb22538971408894ddbf25d76a8fce : SUCCESS in 1h 49m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fdd367ae95d945a1b7ba397ab3179f04 : SUCCESS in 1h 45m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f311427c73ef4623a023d87ad97a1159 : SUCCESS in 1h 42m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06d03366278d49e690fbf5fe8eb2a2ca : SUCCESS in 1h 20m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c6a9576b32e8474e918f47769ee2ae4a : SUCCESS in 1h 24m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cec6382e58a84b9fb40effa1f1026366 : FAILURE in 7m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad70827a712a84e1874e23951a7747f010a42fac","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-25 10:24:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41761fc54fa7bc02b67ee577c9ccc12d2029ecac","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:02:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- octavia-v2-dsvm-scenario-nftables https://zuul.opendev.org/t/openstack/build/dfb18f675a27494f8be021a6a765906c : SUCCESS in 1h 56m 20s\n- octavia-v2-dsvm-noop-api-amphora-v2 https://zuul.opendev.org/t/openstack/build/4c66f298675f4b568a545fe9f4e5f039 : SUCCESS in 2h 33m 57s\n- octavia-v2-dsvm-tls-barbican-amphora-v2 https://zuul.opendev.org/t/openstack/build/53058ad1f553485e9c15120c19c98a46 : SUCCESS in 34m 17s\n- octavia-v2-act-stdby-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64d8e13f1f374757a034780e50d1c9e0 : SUCCESS in 34m 51s\n- octavia-grenade-amphora-v2 https://zuul.opendev.org/t/openstack/build/e631a1435a07445488810caad1e54065 : SUCCESS in 45m 35s\n- octavia-v2-dsvm-cinder-amphora-v2 https://zuul.opendev.org/t/openstack/build/9a67b538685b4eff9e8611fb95c7e2cb : SUCCESS in 1h 41m 14s\n- octavia-v2-dsvm-scenario-two-node-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e189ab5c22645bcb7e2ad15c664eaf0 : SUCCESS in 1h 45m 49s","accounts_in_message":[],"_revision_number":1},{"id":"98721ad5444293275e856275343c0435934b4c70","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-25 15:53:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dd352402bdfdbe8dc143105c5e0dbcd3f502a72","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-12-02 09:13:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30f93ab82bfab669f0d9cc5dc2534fc611930657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 09:17:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99cb4f715280b899a2697808be3675a6d07fe971","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 11:41:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2080112f67e47439886a7d0f36d9049 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bbf1d7524734f24b47af576ca57073f : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d82ac9ada7ce43f1b6ce008bc0140269 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88892d24d6ff4fc7aa65dc4d19c003ba : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ea653b32b34d9ca8b5b1313b43b199 : SUCCESS in 12m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e829a969ecc49ecb69e4f3ef51ec36c : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbf0bf1008bf40ca882181c43bd68782 : SUCCESS in 8m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95179f4fc1744c2ebff5b3c45a546468 : CANCELED in 1h 30m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41cfbca45d5f441b94c139281f9a0e5d : FAILURE in 1h 29m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8b554b7628994265b685be83ace06d33 : SUCCESS in 33m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8505c0f36f5748f49681104e218ce2f0 : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e9daefcc308b333dd834332eb613e5ffb3960e41","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-06 09:13:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbde5b9c8a5d23a4b1fd969fcf46c7130e2b5737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:30:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e5a9e68a37476886163173a2aff89f : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/840140fd2bbd41cfb5fdcc01731e70c2 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a96f3cb7e9495599040e2316585e6d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fb266461b1d4b609367d1463423a04f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9004c5b45dc24319955781be2a4023de : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71d1336824a54bca933d83771430de63 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c690624fb44db180f23044175e02de : SUCCESS in 11m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ab6de7bc9f69404fa89c5471b5c933ea : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/046a39b6d2e04c48a9b6483af36ecd9e : SUCCESS in 9m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/62dd888d646d40d69b2fac34bb8498d3 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/788760fb2d55477386ef8988b8d9b2f8 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e561f784f7b2400f851eb37ff14e7730 : SUCCESS in 2h 10m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c4c14f1575a4e808ff60519252bdfc3 : SUCCESS in 1h 21m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2415137d2dd646018193898e6bb4116d : SUCCESS in 37m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1cbd190508cc4758b241ed298337b541 : SUCCESS in 47m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/155fbd372f084b869c7b1eee48c91f4f : SUCCESS in 34m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/035eec73b3574cb9bd5a377701ea5f0b : SUCCESS in 1h 25m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/019f53c193e74f01b9c96d79cc89084b : SUCCESS in 1h 40m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2bb9351e94b9401c95547898e88b8d83 : FAILURE in 1h 44m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/313ef671fc304755b4facfa0a6345857 : SUCCESS in 1h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/45a65b003692463a9bd592d75572d8d2 : SUCCESS in 1h 15m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/11b7bf09acd14d0d9486b1502c2c3681 : FAILURE in 7m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee57c73755a6240be43d0d4b8842e8194be84501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:30:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a484cc02f39ec2cab30f7a090d0b69308f5179b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 13:48:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4685911aef1b4dafba6fb944dc853924 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7405ba017a4e3f9d890084a0e984dd : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbe65b77918d49188fac21a3f85c8e14 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3934607095a474b8bc4e328ab3e9bbb : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/671770bba3e94e25bb4851f66c23d334 : SUCCESS in 12m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e15732265403463ca8af37e0aba6dbdd : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b95fed2076d84c0e832a523d7a5f7398 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e629f7a3ca7748f9b442a5e6b9261fe9 : SUCCESS in 2h 13m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17660e2e15ce4d07878b162c4002e3d6 : SUCCESS in 1h 37m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac75054b74cd42838d97c38f6e68a2c2 : SUCCESS in 31m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/70a7b6cb124f418f8c1e8535cc39a556 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"83ec7999b840a35ef77f8cd7a3bcf1b67d286b67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 13:48:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"31663cae858ad712d3de2ffd8a75edaa1e333e80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 13:50:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e88b7f412604517a2fe97362b0eb296 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4791cfe93cdde2cb375033164c3bf51e8c65f4c4","revisions":{"4791cfe93cdde2cb375033164c3bf51e8c65f4c4":{"kind":"REWORK","_number":1,"created":"2021-11-25 08:15:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/65/819265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/819265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/819265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/819265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/819265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/819265/1"}}},"commit":{"parents":[{"commit":"237d4436493edeaaf00e544280e10727b44dad76","subject":"Mock log_opt_values in API tests \u0026 fix auth_url in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/237d4436493edeaaf00e544280e10727b44dad76"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-24 15:55:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-25 08:08:58.000000000","tz":60},"subject":"Fix nft command line with negative priority values","message":"Fix nft command line with negative priority values\n\nWhen using nft with negative priority values it is recommended to use --\nto prevent the parser to interpret the value as an argument.\n\nStory 2009710\nTask 44065\n\nChange-Id: Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4791cfe93cdde2cb375033164c3bf51e8c65f4c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4791cfe93cdde2cb375033164c3bf51e8c65f4c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
