)]}'
{"id":"openstack%2Foctavia~819633","triplet_id":"openstack%2Foctavia~master~Ia6fe8483ab76c6b58ca983b7c7f95b48e97ad67b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-11-30 15:33:22.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"Ia6fe8483ab76c6b58ca983b7c7f95b48e97ad67b","subject":"Fix new pylint issues","status":"MERGED","created":"2021-11-29 12:02:46.000000000","updated":"2021-11-30 17:48:04.000000000","submitted":"2021-11-30 17:44:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819633","meta_rev_id":"a2f9df1e365c15fbcdf25dacc0b0034b5a1d11ab","_number":819633,"virtual_id_number":819633,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-30 17:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-11-29 16:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-11-30 15:33:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 14:41:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 16:15:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"925f781306ce3f511f42fd1c83a6183c281641d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-29 12:02:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa872645aed437d459f405d6d242c3a37783ad36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 14:41:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9d9541ef209431c9c6984960bcb8889 : SUCCESS in 11m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f875048f14342b0bb38d68d29faae0b : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8902370a3014f948e7d309e1ea0fab9 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab76765b8f0c4c298da2dbcaa3467496 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e4cc48ab06411e8133c5f0f929baf1 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dba7a2b6c3c34182b6b7f3f0360cfaeb : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbb07a69cd743df9bc8a004cc2fe6db : SUCCESS in 12m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9c9bf02159e24d38b4bdc5bd2463c31a : SUCCESS in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a29fe925fcb1481dbf6a0dfcddc05724 : SUCCESS in 10m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/432ae52a184d48c7b492dd1225f66de5 : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a635aa082a64e1f880f3b9e3720325d : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2103d475ab9948a9a42a7e9d05973957 : SUCCESS in 2h 02m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94e0abd5c02e45d0b590a81308e89f74 : TIMED_OUT in 2h 36m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd188195fb3e4c2484e2e2b9adbe9c84 : SUCCESS in 56m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/061b67ada5434288b10f0712c086db7b : SUCCESS in 59m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d298e3e7ec09485a9f16af018a204b95 : FAILURE in 1h 18m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/811528c8bdf645919439cf870e90b609 : SUCCESS in 1h 52m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1abf2293928b453995abb92d413c6083 : FAILURE in 2h 22m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b583ea554a9c4a55ab6d394b130fcab2 : SUCCESS in 2h 08m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/986e52469df04c7096b6daaf19cd8c7a : SUCCESS in 1h 53m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ad93f3effd0341a0b35a6eb696c72d87 : SUCCESS in 1h 50m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1257a5270f004d0ab7f22be07fd73497 : FAILURE in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08b46a10ff0e312f1ec91894f583d5ab068a2c29","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-29 16:15:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b3928d516435751f994cbc79295c406a7c28d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 19:17:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/393a4a32799e4fcaa5779d0563ab7633 : SUCCESS in 11m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57c971e46b9b4f298a05d3dbf53b57d0 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b726faff0a484a0c919d0e12b8caabe0 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d4353cfd90243648b7326591681163f : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/618609fb24264a8d970d28a11b931042 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f6de67c376947d9be9b124d8ea1f7c3 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe91f0199d94f419a5ae1523a3c4610 : SUCCESS in 11m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c2a4453dc0894754acfc6da71f15d6ed : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f0c56030aecb4ae2a5b10e03d3a2eb7c : SUCCESS in 10m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d792b179ffd41c6bb111620b24be1f3 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4744f75592a94d78b3f07734ff25d086 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ee5437e628f4dec8416865e29fdb743 : SUCCESS in 2h 05m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bad4a24dd90c4092988e226600f57af8 : SUCCESS in 1h 24m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85645b61e4ad46ea87f34493165437a9 : SUCCESS in 38m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a339a3e3b75401e82cdad225f4d496e : SUCCESS in 45m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4be08ece3b241a4ab7828da3dfbb7c6 : SUCCESS in 33m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8f1eb8d361a247cdb2bbba5d978fb5e6 : SUCCESS in 1h 15m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/370b23d2181b4c4291941f8237d608b1 : FAILURE in 2h 58m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2de44c93009c4aeeb2d8bef4877a9073 : FAILURE in 1h 35m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9fa48650bfee4eb5aca68da1e40e3d2a : SUCCESS in 2h 03m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/49140ab937634cdb9afeca5783d09cc0 : FAILURE in 1h 29m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d7ec728fc7194715a3e2e41e8b0a0f1d : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e7a3fb4a137878536c99f0b2548ba7ec887a23c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-30 15:33:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"186fcf0bd5e71374ef20746f9f0f1e4b0c0cc536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 15:34:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24724279a6bd323b1e97bec2b8a34967a229fb36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:44:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4753777db486477791566ecf69a967d7 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9d31b70946248248b741ef7a75d8cf0 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8333c1264a41a6918fad13d2fff854 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e22f56ef9147c185b6e3720e00cfa0 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477e36a19aad437fa985e00d2d495184 : SUCCESS in 14m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/61ebfd197499400bbdfa7a5aaa4c4b9f : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/96d03f48e39d41129b6662cd128e10d3 : SUCCESS in 11m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8acde6ad498b40deb931da87ad02253f : SUCCESS in 2h 08m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2b76a0deb504d58883bfb4e788bf4b6 : SUCCESS in 1h 25m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10f39f9ac3b440fe9f24be56756cc4d0 : SUCCESS in 51m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cbd93945d3234a09b20590ff273f5d46 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e3423f2c2f53f97fc9810d0273d4f37628790f06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:44:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2f9df1e365c15fbcdf25dacc0b0034b5a1d11ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:48:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c039fa309255409f81083e830d8c9841 : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5135d70aaffbd8d93db3fb6eff48a2b9fccfeee4","revisions":{"5135d70aaffbd8d93db3fb6eff48a2b9fccfeee4":{"kind":"REWORK","_number":1,"created":"2021-11-29 12:02:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/33/819633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/819633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/819633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/819633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/819633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/819633/1"}}},"commit":{"parents":[{"commit":"237d4436493edeaaf00e544280e10727b44dad76","subject":"Mock log_opt_values in API tests \u0026 fix auth_url in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/237d4436493edeaaf00e544280e10727b44dad76"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-29 12:02:16.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-29 12:02:16.000000000","tz":60},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nFix C1803 - use-implicit-booleaness-not-comparison\n\nChange-Id: Ia6fe8483ab76c6b58ca983b7c7f95b48e97ad67b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5135d70aaffbd8d93db3fb6eff48a2b9fccfeee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5135d70aaffbd8d93db3fb6eff48a2b9fccfeee4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
