)]}'
{"id":"openstack%2Foctavia~820034","triplet_id":"openstack%2Foctavia~master~I00c3083d7dbcf4473af3386d114c4af9e07751a0","project":"openstack/octavia","branch":"master","topic":"fix-nova-service_name","attention_set":{},"removed_from_attention_set":{"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2022-09-02 16:51:39.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-02 16:51:39.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-09-02 16:51:39.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-09-02 14:07:59.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I00c3083d7dbcf4473af3386d114c4af9e07751a0","subject":"Fix ignored [nova] service_name","status":"MERGED","created":"2021-12-01 14:44:38.000000000","updated":"2022-09-02 16:53:00.000000000","submitted":"2022-09-02 16:51:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820034-fix-nova-service_name","meta_rev_id":"72d4cb452002c4dbfa8bdd2f77b04ef973859e23","_number":820034,"virtual_id_number":820034,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 16:51:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"all":[{"value":1,"date":"2022-08-09 10:02:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2022-08-09 10:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":2,"date":"2022-09-02 14:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-04 14:01:39.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2022-08-05 07:19:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":1,"date":"2022-09-02 14:07:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":2,"date":"2022-09-02 14:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-01 17:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-16 06:06:46.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-05-16 06:06:46.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-05-16 06:06:46.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-08-04 14:01:39.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-08-05 07:19:35.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-08-09 10:02:48.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-08-09 10:04:57.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"}],"messages":[{"id":"a61628859a87e85b212cd76cb03e98d41d38fdcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-01 14:44:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"804770c75d25380afa39e0aa41ec421b474b2eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 17:22:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffb734940b214626b2d746c2db8d50cd : SUCCESS in 12m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a20ffb582264888918134f827cfbf50 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ddb6362ee74bab856029aa2fa2e73d : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aedea54d396e48e8af00a4839a095ed8 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7a7696111e24e16b84bbcac5c36d085 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2028095be1d8412d9444ab6d8f47cb67 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18c0797ede649cb8694d959012834a8 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50c39b041f014f4096e900ee36894c8b : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/140c7955207c4b8c982564ce8f52490d : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2e51caedb4a94fff9b972a314f56214d : SUCCESS in 10m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dd58a8d65e744a64955905284a39bb9a : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b9aad95def645938de0048e47fdb29c : SUCCESS in 9m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/597b9863c1844e86a3cc94ef33c6ae3c : SUCCESS in 2h 30m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48b71f68f6e849e58e6911e576383bf8 : SUCCESS in 1h 42m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8305c023481e47c8b33ceb8d466a3f00 : SUCCESS in 36m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a0900e8f3f2499fae56fa17d1022e7b : SUCCESS in 47m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9cfbf07609f48a4b49881840a9dfff6 : SUCCESS in 33m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/194259cfa3234b6dae9d47c7ba88a75c : SUCCESS in 1h 27m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/69b715c1f8364c9ea88edca4df281508 : SUCCESS in 1h 51m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa843c5c815e407bb8191eb40f80fa93 : FAILURE in 1h 58m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06c1f68fa66446f4a77e073063398e6b : SUCCESS in 1h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/23265ba0cde546f5b10ed0f705793019 : SUCCESS in 1h 29m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/48260ec740b44da6b7353d494a0d2b0e : FAILURE in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10aac19f1c11b1f68a407b4611cdf2f821fd8f90","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-04 14:01:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc0cda9e8c1b66265b32f358e81eabf8ca703c6d","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-08-05 07:19:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"816e20b5b99ceb929bd03bb6b9e990f8f5b7af6a","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-08-09 10:02:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"226e33843a3600dbc6adcdf5c6fadb728ea658cf","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-09 10:04:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad903354ad0d44abb2f55313f42555742e92c51b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-02 14:07:59.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb18e5ad73d5e769e8f685d951acf04dbdd74a4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:09:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67250275e8502bfe31b070ce279241603f7be148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:51:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e891bd9ce47244e8a04ec8ddb2a2c09d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c5201d8803a438c82cfb766c5fdcac5 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32d0647071e44ce5a21efeb3a86873b7 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8127ef6fb1a94b10893266cd604ddb7b : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e72871eff8426e8c0d6be9cdd94463 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91738baad97f4b3fbacd39f9e492c6c8 : SUCCESS in 6m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4473b35efca742adaa70e8378af7f6b5 : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/692234a733274635b6d7905ff05efc72 : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b43b0ed1a3a34f3283cf7cddf68095c1 : SUCCESS in 2h 25m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b290668e0a24f92a3e3816ebcfd5038 : SUCCESS in 2h 40m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea9fe29b830c40df98949e636e135877 : SUCCESS in 1h 02m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9c3d82b7e2848c596a8d8bbb6cba130 : SUCCESS in 1h 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f7c4165c7ef37ec041cdb646f2943d3173b012bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:51:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"72d4cb452002c4dbfa8bdd2f77b04ef973859e23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:53:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f5f5d233e544b7940736136ded576b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da297332b1594c99a6ca0387ced67198 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b2681cfe9394387a1e85cb118db9903 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d8fc69f94003d64e6b017218965e6867426a738","revisions":{"6d8fc69f94003d64e6b017218965e6867426a738":{"kind":"REWORK","_number":1,"created":"2021-12-01 14:44:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/820034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/820034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/820034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/820034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/820034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/820034/1"}}},"commit":{"parents":[{"commit":"81134bdfec79c8019c26e7846d31a2b2952965a0","subject":"Fix failover of az-specific loadbalancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81134bdfec79c8019c26e7846d31a2b2952965a0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-01 14:43:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-01 14:43:32.000000000","tz":540},"subject":"Fix ignored [nova] service_name","message":"Fix ignored [nova] service_name\n\nThe [nova] service_name parameter is supposed to be used to find\nthe nova endpoint from keystone catalog but in fact the parameter is\nnot used and has had no effect.\nThis change fixes the ignored parameter.\n\nChange-Id: I00c3083d7dbcf4473af3386d114c4af9e07751a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d8fc69f94003d64e6b017218965e6867426a738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d8fc69f94003d64e6b017218965e6867426a738"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
