)]}'
{"id":"openstack%2Foctavia~820329","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I7692155fb113442b692267ab12b69ab90200bcb5","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-01-06 15:17:31.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I7692155fb113442b692267ab12b69ab90200bcb5","subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","status":"MERGED","created":"2021-12-03 08:24:22.000000000","updated":"2022-01-06 18:52:34.000000000","submitted":"2022-01-06 18:50:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820329","meta_rev_id":"61e0a7f847c58770e6583913622025757341db94","_number":820329,"virtual_id_number":820329,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-06 18:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-12-17 12:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-12-07 17:25:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-12-17 12:55:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-03 10:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 17:25:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-12-17 12:55:11.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"3edf011b2f0058b854c74a8bb1ed72bf7ed397bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-03 08:24:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dee24315976d8e8dc0f6adb7024cc8ed6aa27fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 10:16:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28b017c0f8d94448a61a6b6124c39d5d : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a493bc6cda34936aa46adc875072468 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5275b61214d64764b1b91bde93d7e21f : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c777a197cd0e449f8dfc287a0c4d16ba : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552c762a38474807a9b4dd20b2787506 : SUCCESS in 13m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/642055e97d404111933d72e3a899eeea : SUCCESS in 5m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/65ac5d7001d549cfa6f820eaa023c935 : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c9a800bc0ad46b1abc8bf547ddd2987 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f6d7e2b1b2c4113b58fc4cee66eddeb : SUCCESS in 1h 04m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8627ad6ab9f44fd8369fe8795910337 : SUCCESS in 1h 51m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2109687203544f53b9e20ec2ae4fb632 : SUCCESS in 1h 03m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/695a7c9c8fd6417fb1387cfe0398bf46 : SUCCESS in 1h 00m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b5e7d4175dc4d11b68df037294c2620 : SUCCESS in 56m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c68d6b4cd9b4a439e944d6253fdc6fe : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5413250b083d46dcb832320f2472fde4 : SUCCESS in 1h 49m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90c10d74fad14cd3917ebb5bc524a2ab : RETRY_LIMIT in 6m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/795bf3a6f811424fa02f0396a6e2810e : SUCCESS in 1h 48m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e20f95179e218d1335d8fd434dca1cba7139df00","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-07 17:25:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dd4e81acd819a1e503ee4b2dda26bfbae4403e5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-12-17 12:55:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0305459212a1796c0d2b308dacc8a1a4bb481685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 12:56:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bce4e9eb0ea91708b946893189b360ea097a9c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 16:00:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d565a11fb7554578b1430327bd07241e : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee8db88f010349598968059a6a44ac63 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6aeeebfceb404a1aad82a4dfcfd7061e : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beafa0d24fd04523bb29f52810720c45 : SUCCESS in 11m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0a59d0bdcc046f7b07713f040536150 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73b968ba8c454b5b855bbb903fe3f502 : CANCELED in 32m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6ce70bf589b41f4bf00245506693136 : CANCELED in 31m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b55c7138f4804a88986431f7c847b9ae : CANCELED in 31m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/393628944d4b451fa11558cdca613932 : CANCELED in 32m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/865332b868a14505aca55d2fd9e8f704 : CANCELED in 31m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d1c19a4b396d4029abc93773261ed789 : FAILURE in 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"548402803186b20cafd1e925e9e665f428f91dfd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-17 19:04:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"932eae9fce72fcca0408b5231aae1ea324705a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 20:19:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdba6cf4a3a3463eaadbbd7bddcbf476 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74864923664547988bf5f0d6f0c812b8 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca741f108d8d4b98839848ee6b02c595 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5704255ff17d4b0a8416110ed003a5da : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ca90a3f09f430087957aeec295781a : SUCCESS in 12m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f02943251f9144abb7fe5b33398347bd : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ddca2ae108364b15b1a559582b974a41 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f983f29720a840b79151e8db6491d350 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08a7410125e748e5b4be5b65165475f8 : SUCCESS in 1h 12m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a36e996c0b844de88cb95c10bbb10631 : FAILURE in 41m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97cf37aad1e24a21840683f5e9b474a7 : FAILURE in 30m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/111021c0395c474db15776ccd3b6ba84 : FAILURE in 56m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c636cf08b68405784124d37f8864453 : FAILURE in 44m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3b2c0921cf848899a7a2d70a1aede43 : FAILURE in 31m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/09998f4b01c24d39b218515927e4f95c : FAILURE in 1h 01m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f177b5d40db42b098ea0dc42af67560 : RETRY_LIMIT in 8m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d523628f35e04b7f930638bd6c2044f0 : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f89609ea81008244552402cd4f2926b1965bd270","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-06 15:17:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53e49bd46834dd6da61639a59aed923d3a248297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:07:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5295295b658d4f57b0dd9337b6254295 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09447d12b214465cb145e491f951dae9 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e234040f04403082319ba2783c1d2f : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17dddbf21d464759a721d9fecd70cec1 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df79118a94346feb7a575c197f8cc67 : SUCCESS in 12m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3a9fab8a8fff41ffb26f84e867d2550c : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/886fd203dd5c49c980df62cdcc296058 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ea6cfde78804757b4eb3119743eb50c : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2b66b1640f94b61a361298cbe15f656 : SUCCESS in 1h 01m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c83a87a1c664449a8a9082ad7bc5294 : SUCCESS in 1h 19m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa8121941bea4490aefc512946a479bc : SUCCESS in 35m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/093790c93a1c48debd4c4e15aab0b990 : SUCCESS in 32m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48fcfaeab4f740c583b54218d7d49a9f : SUCCESS in 33m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e5d8efd052f45c689fd61454168407e : SUCCESS in 50m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f91511ab55d0415cbcefab3d5b03bd63 : SUCCESS in 1h 25m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc662a944cd04c3bb3717dec11a41bfc : RETRY_LIMIT in 9m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51f8d643c5d84285a88e08d117f3e875 : SUCCESS in 1h 47m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3959c3e9de5f6619602f05f258865eccde8b8fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:08:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9fa6658f3ecb92153a81e5480d89d3d95e1693b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 18:50:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ad2e2e8a0e47128362cd0c85367df2 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21f28d87c3fa4eaf81d25895c9c40492 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18fe9bc40a5f45bbb3e5d5f66b69ce48 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771cbf8a2f1c4aba90fbf849919b725b : SUCCESS in 12m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a90a0cc7806943efb9d3d1bd87b3e167 : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/241ceb05c6cd4224bfb1986abad16a3a : SUCCESS in 1h 05m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86d59e8bbeb14083a535ddc171ad1053 : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2294bd2a39694482acf72cc05c5109b7 : SUCCESS in 34m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ef959c6d1d3447ec87c92087f80cf0f4 : SUCCESS in 33m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3f09e48c80a4a3cb80b0c1bde1eda68 : SUCCESS in 33m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d782beabc3c542568e9b87277041e3ca : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":1},{"id":"92f7c8a0a3e5e37f2cb0e3a16c559dd33a2470c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 18:50:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"61e0a7f847c58770e6583913622025757341db94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 18:52:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00bf4fd6d3442f1aef5b5826e3e29f5 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d25fe5fff0ab4c2815e8ec2fc9da7c8fe70171b","revisions":{"0d25fe5fff0ab4c2815e8ec2fc9da7c8fe70171b":{"kind":"REWORK","_number":1,"created":"2021-12-03 08:24:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/29/820329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/820329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/820329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/820329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/820329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/820329/1"}}},"commit":{"parents":[{"commit":"a79f74b2178e8cbfe456d9fadff6853e7d44c4ac","subject":"Fix LB set in ERROR too early in the revert flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a79f74b2178e8cbfe456d9fadff6853e7d44c4ac"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-16 14:30:02.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-03 08:24:13.000000000","tz":60},"subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","message":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora\n\nThis is a follow up patch for I48b0f5a773209b1c1b056d71c0da05d6fd82ca73,\nthis fix applies only on Wallaby and older releases because the code was\nremoved during the Xena cycle.\n\nIt removes calls to mark_loadbalancer_prov_status_error in revert method\nof tasks (except in LoadBalancer*ToErrorOnRevertTask)\n\nStory 2009651\nTask 43810\nStory 2009652\nTask 43811\n\nChange-Id: I7692155fb113442b692267ab12b69ab90200bcb5\n(cherry picked from commit 9b7791db522e7c4a30862cb4d2fb86ffd2de9ad2)\n(cherry picked from commit de20868ab74e5d0b3b24e254b8c4cefae7b78fb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d25fe5fff0ab4c2815e8ec2fc9da7c8fe70171b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d25fe5fff0ab4c2815e8ec2fc9da7c8fe70171b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
