)]}'
{"id":"openstack%2Foctavia~820536","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-12 14:23:34.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6","subject":"Fix nft command line with negative priority values","status":"MERGED","created":"2021-12-06 15:18:38.000000000","updated":"2022-04-12 19:33:24.000000000","submitted":"2022-04-12 19:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820536","meta_rev_id":"3945e65998302e2ca18f2eec27aafb4d48255cc4","_number":820536,"virtual_id_number":820536,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 19:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-12-17 12:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-12-21 01:25:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2022-02-09 17:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2022-02-09 17:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 17:25:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-17 12:57:11.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-12-21 01:25:04.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2022-02-09 17:16:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"bb3c16c8da40134125cf0a246696691bf67321f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-06 15:18:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a28857d2b7d9f8fab72c821528a5352ef91f186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 17:25:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d556625cd94c45e2a0e5541d2847daf3 : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5c8b64832e148afb276e67fb473d06a : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d2345b7826465c8766cc630f4f4514 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d07d27add05b4d85af91b89195d3d164 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4427a0909f8c4c42a1e54565990ad84f : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dc38db16ec84e80ae0994a5166865eb : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0faec7bafc454a77aada2c778d2a8472 : SUCCESS in 11m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de02afd86916491191b6bce14445643f : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/de6141a04ea5437ba01da02f4ba8c262 : SUCCESS in 11m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c4e35cc167d47a0925e1db65e1bb3f7 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/911ca2bf976e4668ab5981f607a6acb3 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b01bfcbed574be9900aba2edbd26eef : SUCCESS in 2h 05m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcbe7571c0824c3c94017cc51b10a32f : SUCCESS in 1h 50m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f1eaecc17e642c0b0e7267df873a8ce : SUCCESS in 57m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/115530d719f949e3a0a8b1621832e593 : SUCCESS in 48m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50f91b8ab3f34300885e37aafa9422e2 : SUCCESS in 32m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8bc9a072d934141857fff746d4ecf8c : SUCCESS in 1h 36m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f97d5c050eb4e94913f726ddcc4c8ce : SUCCESS in 1h 39m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/03af6db1dd62417187ef31c75ff0437f : SUCCESS in 1h 27m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/88abc4335a8147dc9da96b9954cb4a47 : SUCCESS in 1h 20m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1101c4fa51534cb791f1000711642b8d : FAILURE in 1h 46m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7995afc6e44f41ef9c879ba87e02a55c : FAILURE in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8be28d9344bf090236d1bf63c1e1fb9d42a7c182","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-12-17 12:57:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab83a4b5dceea36de85e52f4cb3ee597cb70d6dc","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-21 01:25:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0683ec4d2312b8d453f92f7b83fef903892484b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:16:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44759ebffc4a35de39a2abf73383aacf77524f79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:19:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82e08915725e19f0145af29ac82f8f5bedd5b55a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:30:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee020811f7554778a19a31c7f7d15e69 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7e220f89114770a9ec07d5d083c393 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7fef1d626ae4c29bfb8d8445dd5c58b : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3edb3de5cdf74b82adc96ffd9bc3bef7 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400b7da9c6c04ac0882906263081db5f : SUCCESS in 13m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1473efc73e0f4c808db1794cbe717b23 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ee8b8d99bbb49e5ad330aa23ee5fcaf : SUCCESS in 10m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50285785a53c4e51b15bca4fd756c627 : CANCELED in 1h 21m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31971273a5e64296ac29d8df8331fbc3 : FAILURE in 1h 28m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4da7925bfca647d6af07d6a10db5dd07 : SUCCESS in 37m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0ad7ef42e704f3b998510464f81d754 : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":1},{"id":"eaa6c3a37d83d7dc29b5d9fd6ec53dd4b220674b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-12 14:23:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8310ac61f8637ed3c1733a7f4557ec54a9eb12fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 17:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdf93b0a36c24299baba2543efd07965 : SUCCESS in 11m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b60b277e33d34c1b8265c457ba48df0b : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdeeba1d8bd940e2a2e3ad87670d976f : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ff6da44fe44e588e24f0dfb80b1d03 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbcedd111ec74eb185d9cc9aab1f9502 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45efdcf5dbdd4ea09e99b7251920daca : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f549a494f34502906c754801091671 : SUCCESS in 12m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/19852aa4508848728bf2457caf4dbb2b : SUCCESS in 6m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7ad9fb5d12a545c1ac38305de1bc81b2 : SUCCESS in 11m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/88e67876a31848608ac6506061b6a00b : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d362d5876e2484dbebef15122fa96d0 : SUCCESS in 9m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee4b0b5308194a90b96255af2495e5f9 : SUCCESS in 2h 24m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b59586f5619414fb6c65829f278d960 : SUCCESS in 1h 28m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50ad404fd7c848439cd5150d2bf06cc0 : SUCCESS in 38m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59cc73fcaef949a7b5ad44142cf21054 : SUCCESS in 52m 07s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/1f354354a1d441d19294ec061bfcf201 : SUCCESS in 1h 54m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ac6fb19584746d49d5f57cd525c4dba : SUCCESS in 1h 04m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5cf99238f90d45e3bbf53b7fb436e718 : SUCCESS in 35m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cc49446a82244425ab3e63bcd932f764 : SUCCESS in 1h 41m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/54f7be7beee8412e912c05974a687d49 : SUCCESS in 1h 47m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e3a6f9e00a454ec18f84fe199b80d6cb : SUCCESS in 1h 36m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/704967494b2349789cf23a4da723cc13 : SUCCESS in 1h 38m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fe6a0bad7c8f4dc2a6954120952e4f23 : SUCCESS in 1h 38m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ceb7c0c88a124328b550406c650aa8a6 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80cb08ac468b7de7c9b5618849f96ed63bc053ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 17:07:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3b4c8b011635d43b3213bce7de049f5a2f803fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:31:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fa969a5c78b4f138cbd8605c34bb8b6 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17828ad412734dbaa8af96413c0e443f : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/439475a33d114a2aadff22cccd03f371 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df466a47a4ba415bbd21085d39bcc116 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5547900ef6b1404ea94ae63424e37a10 : SUCCESS in 13m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/621f4285bd7e4efcacc4fd49a3889e07 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06e22531548f4a2297ffb2db46727096 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9bdd6251ac184be1968f88d69e2da374 : SUCCESS in 1h 56m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1ed9c171b5b4b3c9178013bb84f55d6 : SUCCESS in 2h 04m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/046e922f3fd24b3ead2dc074e06c8ffb : SUCCESS in 37m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/409b9c179a6d42c594b0c5da0166af1b : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"707297f2cdbea2f62395f8ad2b8aad74d24f289b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:31:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3945e65998302e2ca18f2eec27aafb4d48255cc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:33:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d16b5c139f490fbc4f1866cef1cf42 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47d5dfcf8bfd0502c08a200056e21efe22d8d75","revisions":{"b47d5dfcf8bfd0502c08a200056e21efe22d8d75":{"kind":"REWORK","_number":1,"created":"2021-12-06 15:18:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/820536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/820536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/820536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/820536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/820536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/820536/1"}}},"commit":{"parents":[{"commit":"4039d35ce221b58490d4aa5b3497e29dbf7664ac","subject":"Fix LB set in ERROR too early in the revert flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4039d35ce221b58490d4aa5b3497e29dbf7664ac"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-24 15:55:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-06 15:18:14.000000000","tz":60},"subject":"Fix nft command line with negative priority values","message":"Fix nft command line with negative priority values\n\nWhen using nft with negative priority values it is recommended to use --\nto prevent the parser to interpret the value as an argument.\n\nStory 2009710\nTask 44065\n\nChange-Id: Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6\n(cherry picked from commit 4791cfe93cdde2cb375033164c3bf51e8c65f4c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b47d5dfcf8bfd0502c08a200056e21efe22d8d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b47d5dfcf8bfd0502c08a200056e21efe22d8d75"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
