)]}'
{"id":"openstack%2Foctavia~820542","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-09 20:30:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6","subject":"Fix nft command line with negative priority values","status":"MERGED","created":"2021-12-06 15:41:01.000000000","updated":"2022-02-09 20:37:41.000000000","submitted":"2022-02-09 20:30:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820542","meta_rev_id":"d3306fbd5ec11a71396465b87e1c768911f91f58","_number":820542,"virtual_id_number":820542,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 20:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-12-17 12:55:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-02-09 17:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-02-09 17:15:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 17:53:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-17 12:55:56.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-02-09 17:15:36.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"38220ed8ccaff81dcb2b27e96994c21a2ad5834c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-06 15:41:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a336cb39daa7123c6826030b6e717848bda17675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 17:53:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db62e568d0524b538bf415c13d845d72 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde7251f404140fe8ee28fdb6db2cf7e : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8c114ce15d941798bb234183e063347 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03c1be86fbe44542a3f66fc911d76c87 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e21d2cc05041b5af9808830c392c9b : SUCCESS in 10m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/547add5aa177442bb8f3b7a67cccbf59 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7e8b670caae848e280a1a5e376176278 : SUCCESS in 7m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/63ce8ddd64194737a721d8c9de82df2e : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4506854b2f140e3a27829cd8c7754e7 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70dea81c7921468594e6b61b6730c311 : SUCCESS in 1h 52m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f89d30553f24f23acac5c597c1ebb22 : SUCCESS in 1h 33m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d1184e44d3749bc9620ef5933148692 : SUCCESS in 34m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/910fe1893e3d4348a45acb6d3b9a3664 : SUCCESS in 33m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd1d7d658fe146a98e55dd21fefa67a6 : SUCCESS in 33m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7b07da9ad4c141fc8ae36f2089209430 : SUCCESS in 49m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cb9cc987da4745f18526cc2430114b80 : SUCCESS in 1h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f051834909a470e82f2dcb10c9acc15 : FAILURE in 30m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/03bbc09d8008423f846210f6839a1075 : SUCCESS in 1h 25m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b761240d3bf042a2824a2de0d958b08a : FAILURE in 1h 29m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c15c89f5fcfb4e75a83da5419fdf7720 : FAILURE in 1h 29m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/33b1e5ac24b04dbea5af96cb3e7270fe : SUCCESS in 2h 04m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"090cb11af1c543ef183556258c73851f4da81830","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-12-17 12:55:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a5d3547576ea09f16e8d0bd2dcf1a13d89aafa30","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-09 17:15:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eafd4e010eb8d7dc4f18b4ba2aad3a1f67a1f4ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:18:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64d4f4f7919e6007a240d7b48216244f5614cc5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:30:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b35501aca54515af4b8e7cf89a8d57 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3013d57bf01946b2a23787301ef5cd41 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d08b256c99141afb860ed7e12a1a859 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297617321acf4714b9461204d8e72032 : SUCCESS in 13m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6b0d6eaacfce482e981857ac46d0f42c : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b88057cc43874045990b00c96de8d528 : SUCCESS in 9m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01f92caae4c6498892cc32aaeabff61b : SUCCESS in 1h 35m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8141fa1cee4f449daa3e718ce637c0d6 : SUCCESS in 1h 23m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24dc3be1050a473b8b88f33a218dadd0 : SUCCESS in 36m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b9604c5038fd4efea38f56ef6e0aeab9 : SUCCESS in 33m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b7b488836f84b0ea9c5288156c02001 : SUCCESS in 37m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0d443fe03504a45b0cb6a83ca3ef85d : SUCCESS in 48m 15s","accounts_in_message":[],"_revision_number":1},{"id":"53014e385c0ace9ae3a8979f201de4c6e2c02fb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:30:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d3306fbd5ec11a71396465b87e1c768911f91f58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:37:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b728df10f24f8d9148036105c0a2de : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2653851e75544f8453cc7a0bd371956bcb86e99d","revisions":{"2653851e75544f8453cc7a0bd371956bcb86e99d":{"kind":"REWORK","_number":1,"created":"2021-12-06 15:41:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/42/820542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/820542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/820542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/820542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/820542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/820542/1"}}},"commit":{"parents":[{"commit":"4ec61cd4c4d9b8f566a6d875d890197147a68552","subject":"Merge \"Fix LB set in ERROR too early in the revert flow\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ec61cd4c4d9b8f566a6d875d890197147a68552"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-24 15:55:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-06 15:33:43.000000000","tz":60},"subject":"Fix nft command line with negative priority values","message":"Fix nft command line with negative priority values\n\nWhen using nft with negative priority values it is recommended to use --\nto prevent the parser to interpret the value as an argument.\n\nStory 2009710\nTask 44065\n\nConflicts:\n\telements/amphora-agent/static/usr/local/bin/udp-masquerade.sh\n\nChange-Id: Ia6bc6eee3df30bfb3c0acccf902267fd2a4d37f6\n(cherry picked from commit 4791cfe93cdde2cb375033164c3bf51e8c65f4c4)\n(cherry picked from commit b47d5dfcf8bfd0502c08a200056e21efe22d8d75)\n(cherry picked from commit 1efeb8d5a4a2e1c2f7e06b831b2c1d3555c0ab8a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2653851e75544f8453cc7a0bd371956bcb86e99d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2653851e75544f8453cc7a0bd371956bcb86e99d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
