)]}'
{"id":"openstack%2Foctavia~822375","triplet_id":"openstack%2Foctavia~master~I4fe9c551b421fb78a23c1bc00c35dd0badf709c4","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"32291":{"account":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"last_update":"2022-01-19 23:53:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4fe9c551b421fb78a23c1bc00c35dd0badf709c4","subject":"Updating python testing classifier as per Yoga testing runtime","status":"MERGED","created":"2021-12-21 02:18:01.000000000","updated":"2022-01-19 23:57:59.000000000","submitted":"2022-01-19 23:53:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822375","meta_rev_id":"de6d93b94426b532458857301f0e2fc7d7ee1b2c","_number":822375,"virtual_id_number":822375,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-19 23:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-01-18 08:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-01-19 21:47:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-18 13:32:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-01-19 21:47:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-01-18 13:32:29.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 05:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 08:52:50.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-01-18 13:32:20.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-01-19 21:47:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d958a0172855ddc71349f08f35f0c0d01ca49b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-21 02:18:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"401d7dc6cc15d1f52fbf9004a3a081a8b96b76b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 05:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8aecee615fbf4a6b87172435327b4dc7 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ebb45c82a9e4e99886991d1a2637c51 : SUCCESS in 14m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce7705a8e28d4036905a0940c8839ace : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aacc07f9e354c79abad326ca57de83a : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c11cfed0a35b48759826f75083cf827a : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed53d6c5a316412d8cb8a14a20638596 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0041cd4ca39b4766984a379993552f52 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcbaa7e673c46e5a42c338c6753a0dd : SUCCESS in 14m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6facef5bd4384406a4c2835bd1fcf250 : SUCCESS in 8m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/381f0030d34e43f0a8bbf354d8b2d8a9 : SUCCESS in 10m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b1238113403c40fba2c72c4efbf1c277 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9014eb832cd145319a06c2ca6706764d : SUCCESS in 8m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/815e25268df24011b1d3eb7a18ca1693 : SUCCESS in 1h 53m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f15ded8d10d46ba981146c101ede656 : SUCCESS in 1h 29m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e757a7706281454db604842a586509c5 : SUCCESS in 38m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9bfa62d378f7444686aef7ec8b94e766 : SUCCESS in 33m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/702b88721bd142828abe282814a271fe : SUCCESS in 1h 24m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4b05799e1ad449abf8739db95276225 : SUCCESS in 1h 52m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e1f46ba2343d4ed4a6da78bf2090119a : SUCCESS in 1h 31m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/30b9441b5c6a470f9d1662ad0503586d : SUCCESS in 1h 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dcafb07d30b34dbf8f980554ff10f9a0 : SUCCESS in 1h 28m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/993b0e57ab824cda99716a44e6e04c19 : SUCCESS in 2h 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7bef30d3bf92d1af1446dcc3e1aa0620cc2a1e0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-18 08:52:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be4d02a95d636cd62dcfc73f2f53fa7b32f12656","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-18 13:32:20.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ac4ae86427cc3f4af56629131793bc6ab6945ae","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-18 13:32:29.000000000","message":"Patch Set 1: -Backport-Candidate","accounts_in_message":[],"_revision_number":1},{"id":"17556ca1b20929aa54a26132f8049de2a0d195b9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-19 21:47:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4635974caa8bab4fc166699fec2b638a3bef310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 21:48:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8561bb179de1d47b87ee9d19b5fb357bdf3451ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:53:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce45f5a418f0455d8bccb8db7a156535 : SUCCESS in 2m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/736949d799fe48789c919fe78d32f03f : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be080187cd864d0387a21cb8d0adde4b : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed174b9fe72d405eb60deda690b0650e : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a81d726c16974680ba39e3fc26ce4f84 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2d4b201a0643a0a2c1b9bf88976ddb : SUCCESS in 12m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7c8b00b45b24ca4aebfa4a9c77db169 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fb0a0f72ddc4e0a8e3066afa8cd6edd : SUCCESS in 11m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f82cf584bc4443faf254ce6265340ce : SUCCESS in 2h 01m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08ea6a76609f4f2281494ea518b47470 : SUCCESS in 1h 24m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72827c9deee944889a03ded135e6090f : SUCCESS in 35m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1a18b7c6f3d14683a6cc7495bf7e3ad1 : SUCCESS in 46m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2adc1a396ee90e334b4ec0fc6fa8832315a64e88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:53:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de6d93b94426b532458857301f0e2fc7d7ee1b2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:57:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c342faec1cea4811bf892bd13ea1928a : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e0ee81ec141c73043f76c8f238a05ea81740762","revisions":{"0e0ee81ec141c73043f76c8f238a05ea81740762":{"kind":"REWORK","_number":1,"created":"2021-12-21 02:18:01.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/75/822375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/822375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/822375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/822375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/822375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/822375/1"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-12-21 02:17:32.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-12-21 02:17:32.000000000","tz":480},"subject":"Updating python testing classifier as per Yoga testing runtime","message":"Updating python testing classifier as per Yoga testing runtime\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are handled by the\njob template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit updates the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I4fe9c551b421fb78a23c1bc00c35dd0badf709c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e0ee81ec141c73043f76c8f238a05ea81740762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e0ee81ec141c73043f76c8f238a05ea81740762"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
