)]}'
{"id":"openstack%2Foctavia~822599","triplet_id":"openstack%2Foctavia~master~Id81ac8dfdd969c7b250d231123d392eff4a86269","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-03-28 11:13:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id81ac8dfdd969c7b250d231123d392eff4a86269","subject":"Fix compile_amphora_details when using UDP listeners","status":"MERGED","created":"2021-12-22 07:28:18.000000000","updated":"2022-03-28 11:13:52.000000000","submitted":"2022-03-28 11:13:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822599","meta_rev_id":"b4d9433651c383c159f4985a778d3212b429ade2","_number":822599,"virtual_id_number":822599,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-28 11:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-03-28 08:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-03-08 09:25:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2022-02-22 23:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-03-28 08:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 10:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 23:41:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-03-08 09:25:07.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-03-28 08:32:53.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"864517da8197a758edf7deab116bbafdbe04b0da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-22 07:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76bd79bd01843aa99a4c4415766b513845d53d2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 10:00:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20195b6522b4405b9fe9debc7fc9a469 : FAILURE in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7be4cfb4347c4da79ee00ef390ee06b9 : FAILURE in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42245b73b6a45e2b6fa85f89c0156a9 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8c091c9047f44049af2993ac53665ca : FAILURE in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0f189981ae443d8a630236c05b448da : FAILURE in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31f171e4db18486f9d30229108d98645 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410eabd5c46f47239a9c276382e9ad9f : SUCCESS in 11m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d5c912fca8bf46d782b16eecb1b269c1 : FAILURE in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9f38e56abfeb4d8eabdb3d70bd2ed216 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1beafd1206434212b3f0009d432ab5af : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3347d4cede54bdebb60380ec45e1819 : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07437edf7ce644cbb6f3f4227baba4ee : SUCCESS in 2h 26m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec5a1c23393743b49f6036018e70b0ce : SUCCESS in 2h 03m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a798913f2f96459c8298c9d58fce07f1 : SUCCESS in 44m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95879afbcd8f48929c69e6176578d2f3 : SUCCESS in 1h 03m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/291804d4c270448cb0dc54e6b0fdb659 : SUCCESS in 41m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/41d11709cebe4d8a8038f55f01b45006 : SUCCESS in 1h 57m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ba27a223bcc4bdbaca74bf07859e956 : SUCCESS in 2h 07m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3535ab3f993b48268a83cd30399fe3ae : SUCCESS in 1h 59m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/790bbe806e6448f1b5e7e337e1e52b0e : SUCCESS in 2h 17m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e387df9e1881494389ee0a8a638dbd4b : SUCCESS in 2h 02m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4502a6ff676345dca2c542de7e972f74 : SUCCESS in 2h 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97f54062140c244553402fb3ca472838a1df2c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-22 10:25:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cdf84d01d7a78e84b5a4ae02f3967e4ecd4496e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 12:37:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4e58595e6824035988a9ee1867895c3 : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bc357643cc74cbe93b88d08f676f95e : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de04652714c74c04b7b1d8dca096c965 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b14c77b135c410cad0042a280eeb629 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38d561b89986458faffd3676c7b970a2 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35a397b45cd24c06a583401d752cb0af : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0647a63bdb4d44b8787d7d573b4ec6 : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1c14d4a85d8d45808dac19ae65191e2d : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/118c49931cf44014941b64a449fa7506 : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3d60c200547455289ebe34d36c56eee : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bebc3ac84a247e7bd9fd9083c1f1b3a : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3d122cf07a14e8ca5944a329aea9ef9 : SUCCESS in 2h 04m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a7690cd94984e34b4c4d5307d0e3fb8 : FAILURE in 1h 17m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e4e414dd84da4447956f1daeb0ef0579 : SUCCESS in 34m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0e9998dd9c5a44a6b3a5a007db3a62f7 : SUCCESS in 46m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3c6cfa22a8d4ba1a064b8643da9083d : FAILURE in 1h 15m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e7f7161d8cbc45ec86370508ab217c93 : FAILURE in 1h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e2eeb9a528314682828668acf741ea3a : SUCCESS in 1h 59m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c63416ba08ab487ba414ecdfb0723f7b : SUCCESS in 1h 30m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5dfaee37950345f68ef662928564497e : SUCCESS in 1h 37m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/db0cba2085864e26b29d759104067843 : SUCCESS in 1h 59m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/c739e35819e84c1793f63d4214a16280 : SUCCESS in 2h 09m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abe549f082387c1227b07655c140d450058e4aea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-22 23:41:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a218aec619fb60369e1d8178863efaee5ebf628e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 01:58:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4e6fc209e684103a4d2d94cfc944002 : SUCCESS in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a2e0ab3c7f84b30a1051b77646d07fa : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c201651bd407479190117401064c8f10 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1087c21c239427abe18f1877f315f0b : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/960605d1dbaa47d79c788e3e0bd08220 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d43de6aac56a4236bda8e3cd119c4c5a : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d29a34c4c2c4683bbf441bcca4023c4 : SUCCESS in 12m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b1ae902fdb454879bde3331e63a78f5f : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e50364f0574b497c824ca2923964e4fd : SUCCESS in 9m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/81cece77df924d0eb33640793ba5bcce : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/122e744c5f334e1ca2e925ae89d5e779 : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e223085b2bb44931b67a82890228d81e : SUCCESS in 2h 04m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b78d83f05bd4df6a444a917c584a206 : SUCCESS in 2h 09m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c6e672f241424552aafba91b92b13071 : SUCCESS in 32m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/17effa2a30ee491c97be693cfc6fd797 : SUCCESS in 44m 08s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/6d53951652f04153b578c90f7cd8c402 : SUCCESS in 2h 15m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/93b2919d81174c73b7c6031c0e4a9bc5 : SUCCESS in 54m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf66d9c91d744bceb91d16ac8aade1bf : SUCCESS in 34m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/126abe654e594bdb806517a6c0a4f69a : SUCCESS in 2h 07m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0c4ba0d0704483fbd2974ca23b74de3 : SUCCESS in 1h 47m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5b8baa52439b46f08a34bc21970d6c14 : SUCCESS in 1h 39m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b7f7e243b345496ba26091dd8f6d1f10 : SUCCESS in 1h 41m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ba4eec67d5814a31bb0f2de609fd38c0 : SUCCESS in 1h 25m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/870deb3fa3354d55ae8488c3bcdd4f83 : SUCCESS in 1h 56m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c3b6509f4aa8e3643644048aef6820a114065c2","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-03-08 09:25:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cf435a450c833912eb6f1eaf23c963caf86f7e6e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-03-28 08:32:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"980623bdfd53789dc70fc2d0cbe56ced641dde64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 08:35:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9221c2663c8b80253442f53f12490c32710ee35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:13:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43efeefa823a492591bc2d828b2da402 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db85840b5592419fa4f1a35580961252 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01336a50b155484a84b8d51b412d49e7 : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddde07b735aa4084a6ae5abe23b3531d : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e1ee749a0d42198beba10a2ea2837d : SUCCESS in 15m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4603bba3a5cb40aaba3644e534f3dc02 : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f99153486b6f4c53afc632536bf9c8ff : SUCCESS in 11m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f10d09e67a574b3cba719ec84009a644 : SUCCESS in 2h 16m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4c92a9614cd411d9af6dae3f57a191f : SUCCESS in 1h 31m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bfc7b7fa7f02431394e389551777eda6 : SUCCESS in 40m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a96f1b3db8f492598a61a2e2493db3b : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b4d9433651c383c159f4985a778d3212b429ade2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:13:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"655fdf8c1b5786c3aa24c926dd95537a51a7c0cc","revisions":{"8609f30e8ca18a6b12a8865d2709360fc30d5d53":{"kind":"REWORK","_number":1,"created":"2021-12-22 07:28:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/99/822599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/822599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/822599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/822599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/822599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/822599/1"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-21 17:02:15.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-22 07:25:03.000000000","tz":60},"subject":"Fix compile_amphora_details when using UDP listeners","message":"Fix compile_amphora_details when using UDP listeners\n\nWhen requesting the amphora details of a LB that has a UDP listener, it\nfails with an exception:\n\n  AttributeError: \u0027KeepalivedLvs\u0027 object has no attribute\n    \u0027is_listener_running\u0027\n\nThere\u0027s a typo in the _count_lvs_listener_processes function which calls\nan inexisting method.\n\nChange-Id: Id81ac8dfdd969c7b250d231123d392eff4a86269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8609f30e8ca18a6b12a8865d2709360fc30d5d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8609f30e8ca18a6b12a8865d2709360fc30d5d53"}]},"branch":"refs/heads/master"},"655fdf8c1b5786c3aa24c926dd95537a51a7c0cc":{"kind":"REWORK","_number":2,"created":"2021-12-22 10:25:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/99/822599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/822599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/822599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/822599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/822599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/822599/2"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-21 17:02:15.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-22 10:25:15.000000000","tz":60},"subject":"Fix compile_amphora_details when using UDP listeners","message":"Fix compile_amphora_details when using UDP listeners\n\nWhen requesting the amphora details of a LB that has a UDP listener, it\nfails with an exception:\n\n  AttributeError: \u0027KeepalivedLvs\u0027 object has no attribute\n    \u0027is_listener_running\u0027\n\nThere\u0027s a typo in the _count_lvs_listener_processes function which calls\nan inexisting method.\n\nChange-Id: Id81ac8dfdd969c7b250d231123d392eff4a86269\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/655fdf8c1b5786c3aa24c926dd95537a51a7c0cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/655fdf8c1b5786c3aa24c926dd95537a51a7c0cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
