)]}'
{"id":"openstack%2Foctavia~824941","triplet_id":"openstack%2Foctavia~master~I5fdfa440eac8b42cff20875e245d52c1c73181bc","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I5fdfa440eac8b42cff20875e245d52c1c73181bc","subject":"Rename oslo_common deprecated parameter","status":"ABANDONED","created":"2022-01-17 14:02:32.000000000","updated":"2022-01-18 08:49:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88a0651eab6b220a21415d11a2021f9c8b81b045","_number":824941,"virtual_id_number":824941,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-17 16:35:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 16:35:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5676024ed344e3a484c698c071ae6cfce357acef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-17 14:02:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbf6059fced31522a3d2dea9f62f2aaa4930fc5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 16:35:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/026a5dcc8a384b67bca97e5dfa4d94d6 : SUCCESS in 12m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca8dbbdf619a4b9f9d157bbd0b5e33ae : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dcbaa904a704ad28c8c0211597e22cf : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e215728bab7a475ebe535db426b38d14 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6108d66f76a14a22a782bf46535b34f9 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63ecb22c664b4e5d975d6592a1b4bee2 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b8f8578dce4d969231c4f8ee9c87dd : SUCCESS in 13m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d932a235ee2549519dd5f1b19c7e0e3e : SUCCESS in 7m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a82ca5d717e44c4eac6286c237e74108 : SUCCESS in 10m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6c6f8ea30dc742f3a341fd1125a38f06 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72dbfd12bbdb47098d26b92a8fe72d5d : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b881168a77034699b929008bddcaa954 : SUCCESS in 2h 25m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4fb000090a3424980986d009773addb : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c8cf5a2ae0e4f59a8c5b94a55de7721 : SUCCESS in 36m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/910d251f89ec4ce6b167fe621c229f53 : SUCCESS in 46m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a62c452225834caebe1121ebc181d1b0 : SUCCESS in 31m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/37bb0d0c84104756b2c4748bc0930387 : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a598286ec34647519c661e9e57541dca : SUCCESS in 1h 41m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6ee78917cc1f4e8ca180d6f854e19d87 : SUCCESS in 1h 40m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e8190d6f07324302ab4cc7ce90cc7418 : SUCCESS in 1h 29m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5d6379eb28a84a59a322d7bc163ec8ac : SUCCESS in 1h 20m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0608d03465fa46d2ab9211f07de5524f : SUCCESS in 2h 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88a0651eab6b220a21415d11a2021f9c8b81b045","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-18 08:49:58.000000000","message":"Abandoned\n\nduplicated","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b60fcd2809fb769ee4730320eaf06a5d5fdf9aa","revisions":{"4b60fcd2809fb769ee4730320eaf06a5d5fdf9aa":{"kind":"REWORK","_number":1,"created":"2022-01-17 14:02:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/41/824941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/824941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/824941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/824941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/824941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/824941/1"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-17 14:01:00.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-17 14:01:04.000000000","tz":60},"subject":"Rename oslo_common deprecated parameter","message":"Rename oslo_common deprecated parameter\n\nThe tenant parameter in RequestContext was deprecated log time ago and\nis now replaced by the \u0027project_id\u0027 argument (see [0]).\n\n[0] https://review.opendev.org/c/openstack/oslo.context/+/815938\n\nChange-Id: I5fdfa440eac8b42cff20875e245d52c1c73181bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b60fcd2809fb769ee4730320eaf06a5d5fdf9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b60fcd2809fb769ee4730320eaf06a5d5fdf9aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
