)]}'
{"id":"openstack%2Foctavia~825944","triplet_id":"openstack%2Foctavia~stable%2Fxena~I6619ba117f7051fe44086389789bc6e2810fd23d","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-12 14:27:08.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I6619ba117f7051fe44086389789bc6e2810fd23d","subject":"Add keepalive for redis-based taskflow boards","status":"MERGED","created":"2022-01-24 08:34:15.000000000","updated":"2022-04-12 19:34:53.000000000","submitted":"2022-04-12 19:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825944","cherry_pick_of_change":823415,"cherry_pick_of_patch_set":2,"meta_rev_id":"8e4a8944eb77df671b886b3cc379fbc8375c2553","_number":825944,"virtual_id_number":825944,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 19:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-08 12:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-02-09 17:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-05 14:08:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-08 12:20:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 11:12:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 17:21:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-04-05 14:08:48.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-08 12:20:49.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"3faf87a635b7599422e2954bc59c7a3e81c5b7f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-24 08:34:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da954b8d5330dddb4b6893a4d621458a61319650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:12:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f5ec9bfa62f43e690f7b63d4d449bc1 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d277f7064994d36b7d38ed8bbe7ff66 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588dce660049414f925302ba41dcc9b5 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4e8effe08cf4d17a1eed5190c9daa60 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c74637c56fe4bef963d5f2eeebfc029 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9067b21504db4cca856986c4d0bae638 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68182e640ff04a699037f97201b910a1 : SUCCESS in 15m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e759b37de3be4914b64da0e8452efe32 : SUCCESS in 6m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f3bfb24490db48a886682638c1a6b339 : SUCCESS in 6m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/86221b9808e3423390b8ee4bda4769b0 : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f6315a96ffef460db675dbef1dea9677 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d36ceb0c47b84aa4b2c73a9431f10dd4 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b65476fc13604b6d906fd5e255c50254 : SUCCESS in 2h 35m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1982a3e3ff443128a0d305137e9da9b : SUCCESS in 1h 27m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0af10535d954a3ba3f9ca82501ec529 : SUCCESS in 35m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd3ec3bebef9482bb75e77fc52274602 : SUCCESS in 47m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03bd872ea3de4904a34af5b1e151a853 : SUCCESS in 30m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c825a095d6164d07a4ff0f5ffad70fbb : SUCCESS in 1h 15m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6fe2ce196a93457cbd141587563b665f : SUCCESS in 2h 05m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db5cfb265119457bb3b007b801628bfe : FAILURE in 1h 37m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1025cd814ca64d3e9e4342d37b2a56bc : SUCCESS in 1h 22m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0d532f94cf6c4a2e98e657267f3395d6 : SUCCESS in 1h 18m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/eefb9bb356884a6fa3d6242fce317140 : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd60db9eb68b87d0713ffcfed53aa1f8da57565e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:21:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11422406a581d19d263f3d3bafbfa865077cfb4f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-05 14:08:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"16b91d85ba725b61bf1c1be8d3a8214d68678815","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-08 12:20:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56324a2f49fa10f5676821e7b408273a81931c6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:21:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2553100dce67e90ad7096742a441251938e11483","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:45:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5883f3339e174511abe207d57b4d4818 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88c3f5e4c6e476dbdbc55fcb415d6d3 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa918df6e6c54b43acb31f46d9248541 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cb9b3d966c1427f90f7d88ad4ec4df0 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad2d0ed8b5d4cb38abbfdb730acf104 : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6663034fa18c410da5c77787959b3d1f : SUCCESS in 5m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aaab19f6532a4bcdaaeb1c2c0e737801 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7171c916c354580a71e00453fdb80d8 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f9f6e039dfa548e9b6c221621161318e : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6ef642ce13049f1a14bda21ce8558f9 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3958cf79e334e9388f98fe405a49973 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28232758218e48adab91951771aa02fe : FAILURE in 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"39aa6ec4f9e7671a0c2a2a15907dc534e28a8564","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-12 14:27:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"007bd33b0b696b11e9bb761bda3ef372d41fc779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 17:08:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25d8e3333cc54b1e92570f35caad2f42 : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26cf33487321410993d55ea0f095385a : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d734af896f894a34bc458c440e5e30be : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15b1c5d8bee24888b4f8c6715668e9a1 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa9eed1aca94f9786e33c62ce990eb0 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e61b50dd00c9471ca297657dfbdc5498 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90708d1da58e45b582f7dfcb69fe5715 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/892cf6e2c53b4408b463d16ba60b9e50 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0e385718a2f54fa2b08aacdd7dad9998 : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/114fe7263e744ab284d30e37fafbce4f : SUCCESS in 11m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f8555b69856a4093b10afe9fa2f0da48 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49002a2efe544d99a0d515ea20da2934 : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d78ab8699b9a4fc9a7f012c7d7cb3942 : SUCCESS in 2h 27m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a8d4c5c55724c5b897c2c45f141b080 : SUCCESS in 1h 45m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab67690394b14cb89f9730afc072d81a : SUCCESS in 36m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b0bce04048c4644854cccba253d7643 : SUCCESS in 50m 15s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/d9a209d4afc547e9b63c3df709f5e70b : SUCCESS in 2h 15m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/94bc12ca699d46ae8c4d3309f6bf12bc : SUCCESS in 49m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eec629d7128f4bd596c6ea442272cf6b : SUCCESS in 37m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1de76dca16634e67be19df401bd0976e : SUCCESS in 1h 29m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87a0b5b0471a4ff99df62cd0510fa122 : SUCCESS in 1h 49m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/41faeac732d6490c87c267072339b17d : SUCCESS in 1h 46m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4d854928a9094b48b76d559afd89baf6 : SUCCESS in 2h 02m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d000ce241f32490ca80943df8c823586 : SUCCESS in 1h 22m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/83e09b4343a340cc80e5543c187dda1e : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fab7f6044c27988565ccb6fe86950e843b1ab0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 17:09:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca8549b15187a573591292afd8538ca7e4878895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d2341bc2fc4f00a4c284911e1882f5 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a34353cbf444ac9ecdca9649d8df90 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b546cbeae12478db6cd0d357042b1e2 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eabbc9c7e605413297ae7a5cabf8c491 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7cd64d72fa40f9a2dbd7183e363e3f : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bbe54fdbbf444f08a62ed6bfd7e9286 : SUCCESS in 4m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5eb3e56217ee44ac9bef9dd4bf4d7a0b : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3fac795fdf574f6bb4646e7c8302c724 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc4cd2715bc04ce89d1e860a9501173a : SUCCESS in 2h 10m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/985d8fd147ed49a19f44fbe845478a20 : SUCCESS in 1h 37m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ddc17928b6542d282f339bead9240fb : SUCCESS in 36m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e1a5167d82a44bb58f520ee59f021c9f : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6ae119167044c1e900dd501a615cfae96772bc64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:31:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8e4a8944eb77df671b886b3cc379fbc8375c2553","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:34:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4248929d612e46e5b80821cc401e1902 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0931af4c7fcd499d882ab6bf39c21a5d : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"987d6a346f9a7df5ec9bb3d341fc4f90423608ed","revisions":{"987d6a346f9a7df5ec9bb3d341fc4f90423608ed":{"kind":"REWORK","_number":1,"created":"2022-01-24 08:34:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/44/825944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/825944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/825944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/825944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/825944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/825944/1"}}},"commit":{"parents":[{"commit":"4039d35ce221b58490d4aa5b3497e29dbf7664ac","subject":"Fix LB set in ERROR too early in the revert flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4039d35ce221b58490d4aa5b3497e29dbf7664ac"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-24 08:34:15.000000000","tz":0},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/987d6a346f9a7df5ec9bb3d341fc4f90423608ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/987d6a346f9a7df5ec9bb3d341fc4f90423608ed"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
