)]}'
{"id":"openstack%2Foctavia~826028","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I6619ba117f7051fe44086389789bc6e2810fd23d","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"5572":{"account":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"last_update":"2022-02-21 16:18:07.000000000","reason":"Gregory Thiemonge replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-27 13:05:27.000000000","reason":"Gregory Thiemonge replied on the change"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-02-21 16:21:37.000000000","reason":"Tom Weininger replied on the change"}},"hashtags":[],"change_id":"I6619ba117f7051fe44086389789bc6e2810fd23d","subject":"Add keepalive for redis-based taskflow boards","status":"MERGED","created":"2022-01-24 08:35:32.000000000","updated":"2022-04-27 17:42:05.000000000","submitted":"2022-04-27 17:40:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826028","cherry_pick_of_change":825944,"cherry_pick_of_patch_set":1,"meta_rev_id":"b4986ed7d8a505293f81f4f6febdc147406b2951","_number":826028,"virtual_id_number":826028,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 17:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-15 11:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-14 22:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-02-21 16:21:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-15 11:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 11:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 17:23:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-17 11:12:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2022-02-17 11:45:35.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-15 11:43:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"58eeabdebcbb9752a63d4ad939635c0a66ccfcf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-24 08:35:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a76163ac0cd5d41be1451f8a1db17b9611fa52b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:07:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/589c0d2f29244e689524787a98d79ea4 : SUCCESS in 11m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7fed09632da4869b6b3674a85cb9c33 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe68ec67b1534eb2adbdac01a30000e0 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1229583dc094df4b92d6825335efbea : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac66fd22e7be4ba48a4d46a906b3a884 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee1c88e961794de6b10a90feb0e040f2 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26cb258b91445b39f0335a9dbcc4df8 : SUCCESS in 15m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fed8f5ba450f42e48042e8156811c3c1 : SUCCESS in 6m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1164c09c0ca440e39a596888b0ac503b : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/032b50fbdd8044b39b91a5246c9a11b3 : SUCCESS in 9m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af0ab554db91405fbd9e12b0206c4765 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d8d962021764d31b2cb426f46f7ef92 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e181905e9dcb4851835a4f280e8acb13 : SUCCESS in 2h 30m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06e08eb0d419471ea2503e3ecadd1312 : SUCCESS in 1h 11m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ae2e851984c442eac0faac7da65dec3 : SUCCESS in 33m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e32e93c31e04010b867a8e493eea67a : SUCCESS in 30m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b22c54f95e6c4ea6a10cc7ddbd61aeff : SUCCESS in 32m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0bdd227c9bf46bca00edcd8ef65cd80 : SUCCESS in 48m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/62ceb42bb1f942458172df57a9003b9b : SUCCESS in 1h 22m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9dcdd149d6c8404eb1495b3e1f3fe309 : FAILURE in 29m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b6c21cab5d0547a1b4ca64f40c529cdc : SUCCESS in 1h 31m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a5138e5085684308ab2f2793228c8ab0 : SUCCESS in 1h 26m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/32839a53a05142a5b4a3b75356c64ade : FAILURE in 1h 31m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d15fc06468cb48e4bf7321eca4b2a38b : FAILURE in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"956664dddb8ff8868a1a31c7cc29767987fb5242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-02-01 18:48:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a06e8e971d269305c188ad1c3332947d940be83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 21:00:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a562df313fd4c3fadad336eadd15048 : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe59ac5d5f554260af78980c47a93d81 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bd2b183f4446b99a0626ccbd65f4da : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/970b98c1b70b4384a9d9e7e31ffd3d11 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85f87dfc3c5640f6aba57eae28234bf2 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31b2e026858148cc8369368c0538333d : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e178d9a95e1d4eb29fdd0fe8b5143eb1 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d30eafd36ab48afacebfbcfe2801584 : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b0016cc2bcbd498c80e660360084abcb : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cd211240c3f64aaca4749dee364d9402 : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/688163e00bb448ae87ceb9d8d60aad5c : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9099951772bd40fdb28923e3b73b37a5 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/124d754ed6ec445093398ac48425d417 : SUCCESS in 1h 35m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8acd1999297b4f0b87ca8c72b1822e01 : SUCCESS in 1h 15m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a800be306d9649dab7b79a4c5e6064a7 : SUCCESS in 46m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/211662fac063488f982833de7065d7c7 : SUCCESS in 46m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad266d4a4fbb48cc8852fb3d130152db : SUCCESS in 46m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cd1504e546b54087bdef4f5e92064b59 : FAILURE in 12m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c2513dd17c54c96a796f65446940315 : SUCCESS in 1h 51m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/863ca9cd50184a9081772da2ac263920 : FAILURE in 42m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/52c45b2d8a5940d29e802306ab8f9d20 : SUCCESS in 1h 24m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72a84bba6e0443ef95ccc87b6dc0ecd4 : SUCCESS in 2h 08m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/76708e03278b4d9d9683221ff7134000 : SUCCESS in 1h 58m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b3f1617a6c794a79a62696898f481147 : RETRY_LIMIT in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79ceadd023e32dd057bf1114759e6106cc6e4575","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:23:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5546e3a2e2682c82b300c1c6a495084edba3009c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:28:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3b0864af8cf26e197424dfe1b0c811b1c2f97f3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-17 11:12:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9786aad5abe3491dfa41c9f19e06099c4ab0f822","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-17 11:44:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6671b967cb0cc24d7f3d05f1fde3461606216e45","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-17 11:45:35.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"ffda74cf8b10290bf7e27b54945310dddb96f08f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-17 12:13:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04b326d1d31b3cfa03179f4563dc0f4343c516c8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-17 15:59:08.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"1d562f7ca711cd6b687dbfcb17ff046fd40a12f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 16:17:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7a0387488ef9ea8ac5e0745123f6c2b4153b5e5a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 16:18:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4eddc12c24057dc3f87be81740f0f74b50657a57","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-21 16:21:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e508b535d22581c34d72a0deead29af11d798cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:25:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7996e2b302c4e58afe937f52cdfaba0 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/285a428ca98045a1a126b940d1d4c61d : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77deca82d6fe4792abe5fe8a24ccce9c : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79a2295ac2f6441dbe3ca0468611df99 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c2c19a8c3be405f96665b48bc7686e7 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cde03e60acfd4b14932bd62107f14c7a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72826fa034df460e8969b3527ae7257c : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc4a5d20fa8a4a6f8075eba10b913aa5 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fce465e960aa4a86a03e146d7c9bee3c : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4bf87286d7d94b5f99a7b5fb45b27ef7 : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6eb16398a5b545f49256a1ac6053f0ad : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/417c35ce7a9d46e5ab283e7f8ca75b7f : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2614a12db069440baa1a92d5436596ea : SUCCESS in 1h 30m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ef813cbfed749cd87f3f2176ac5d02e : SUCCESS in 2h 00m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4461236f2a4b45718f33f2760b0ed0a1 : SUCCESS in 38m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ea75e551fd2a42da839c41ebdc37f445 : SUCCESS in 43m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a112a00b4254c8c89fd6a70f74f3c93 : SUCCESS in 46m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9eb9dfb6c36a40e382663c716474d0ff : SUCCESS in 52m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/39985bb2d0b94624aa71a0a1b8d86668 : SUCCESS in 1h 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d9edf4cd43694554a33f5c71b74e9e0a : FAILURE in 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/154717640f47475a8c15163550c5e90f : SUCCESS in 1h 55m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/377416cf8cf745ba82c66906b4de72c6 : SUCCESS in 2h 02m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e40c8264afa24e48961d816f92e39a0c : SUCCESS in 1h 50m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fbcced12e72948beb1d8ae2364eb8407 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"98d3f4c7e0c4ddde646ba469f7fbfac3d017e74a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:52:51.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dfcecd69ea0dbe7f95df1a3e217dc6450ae4e475","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-15 11:43:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5b42a1507bea71c51c745d95df52bd5ae69a111f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:44:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8c281070bfce469288ec7b48af266d50681c9c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:57:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5259610e09014b1d8ac2fc3463b0fdc6 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095a163d700f43659086f6d367a0f38b : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1a02bd6876f415fa92dd3778bae7109 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2d93cddf01e4244b0d6d32642111c0c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df389873e584a8ca0cf73cdb680f065 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c24f5470394d4ab19265b54e75d63655 : SUCCESS in 5m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd1c499ae5bb42f491f24a1ac6d553fb : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/edcfd16bb8874d86b3ab14f88fc4b7b0 : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4bfe8bd16f3b4401ab9dce362648ceba : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1770b90d86814f3fb18c494239dea965 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b52ff46118704379ace52e1ef9117d4b : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/92a24353a2ea488e99f0c374b096d41e : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e6d6eae3d99410bb831a53699593195 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a4f8684ea33c4bbbad5ea1a98fcd4d12 : FAILURE in 11m 54s","accounts_in_message":[],"_revision_number":3},{"id":"983ecd11e4e1fe4e68d3279a82aac7f2b6abd598","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 13:05:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c49cb1ebe5fb8c030b99f735e2fa3e529c9c6df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:47:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccc6a9f7548940acbe68c52304801e01 : SUCCESS in 10m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36faac70707943cba963c0a65d95e5e2 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155c9fc59c8d435687ae11a05dfec6a2 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/430a2058616845a0879cf8a885ed1825 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5eb24a4ba014fab8094b666e8265057 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f60426511177420faaad1355fed3c8a7 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b398e2483f2d490b8fd670e0e2bfe814 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc553cd6402845c7b4673c7a555d8280 : SUCCESS in 5m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e6a91eef55ac4aafb6fce622a4a1caef : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c507599fcd394393a34553478cbb99ec : SUCCESS in 8m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e451c3e782f745b6b8745f46e6320f19 : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a4d9392d906467699c1b10f42e741f0 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/844976b1e58640feb62f92c56d3859d1 : SUCCESS in 1h 29m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c891c01a1ae4c0faa7c465dacebf090 : SUCCESS in 1h 20m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/629821430b69400ca0603561560fe09f : SUCCESS in 35m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0dd4b253fe63475d95c6c981290ef122 : SUCCESS in 37m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44c36430ac304ab3b00d03f0559f9414 : SUCCESS in 33m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/42bc740763d34d88862fc2d097a88374 : FAILURE in 39m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4a2d253dd44345a88f89495ceceab2e6 : SUCCESS in 1h 49m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ecf27a9ec5ad414aa8498002a252ed81 : FAILURE in 31m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/88f0c9d0015a49379722eee94f9d5552 : SUCCESS in 2h 13m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b8bb2bf7ae2445df8f85f3303bfea3e8 : SUCCESS in 1h 45m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/860e757c617c469795ecf6458400c7a6 : SUCCESS in 1h 36m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ef33e8627d2241f186270d239f34486a : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd055fac77394bfe616829a03d9ef8d6e96a8518","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:48:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd62e38623c21720e0bbb4ed5715c54f0a0d491c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:40:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5574d8be0254b36ac867d86ce8eddd6 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2071239d4c83414bb7a38247f2a1e4e7 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f70ad18e8cf4ca89b82faa1d1291820 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ef3e589561462ab75332853b311757 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c21f9146a34d369244ae81eab4c2c2 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d29dc9eb8894b39b018fbf5d626f366 : SUCCESS in 6m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5edacb30e72c429790d363e8f27e4813 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0fb74558c1b6476dad4985a0c6f61757 : SUCCESS in 8m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e4a7a607de643f4af8f7035a8739870 : SUCCESS in 1h 47m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8887e2bf00e64ebc8d4a325cd33ce681 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef68e24eec7b436495884c6c7341fa08 : SUCCESS in 40m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8c58061e43e6464284a10b7345f496d8 : SUCCESS in 33m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb1a56d62b6c4dff8ec15b3c87701a30 : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":3},{"id":"1da02f60a79d53aac99c59e817fcd36410dd0ed3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:40:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b4986ed7d8a505293f81f4f6febdc147406b2951","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:42:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a17d0e06e0d4bbbb27129b4e26657a9 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6516498151c74e2caa469f894e4e3d3a : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b4806a83b3fa142d1fbd522add11f453199f2a77","revisions":{"f83120485c82f8713c1d3fd32fd808c9ea5b4654":{"kind":"REWORK","_number":1,"created":"2022-01-24 08:35:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/28/826028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/826028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/826028/1"}}},"commit":{"parents":[{"commit":"a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8","subject":"Merge \"Add generic network interface management in the amphora\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-24 08:35:05.000000000","tz":60},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n(cherry picked from commit 987d6a346f9a7df5ec9bb3d341fc4f90423608ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f83120485c82f8713c1d3fd32fd808c9ea5b4654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f83120485c82f8713c1d3fd32fd808c9ea5b4654"}]},"branch":"refs/heads/stable/wallaby"},"727cf365b5bbceddebc18a10af3cfe883408137b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-01 18:48:36.000000000","uploader":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"ref":"refs/changes/28/826028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/826028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/826028/2"}}},"commit":{"parents":[{"commit":"a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8","subject":"Merge \"Add generic network interface management in the amphora\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Don Kehn","email":"dekehn@gmail.com","date":"2022-02-01 18:48:36.000000000","tz":0},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/727cf365b5bbceddebc18a10af3cfe883408137b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/727cf365b5bbceddebc18a10af3cfe883408137b"}]},"branch":"refs/heads/stable/wallaby"},"b4806a83b3fa142d1fbd522add11f453199f2a77":{"kind":"REWORK","_number":3,"created":"2022-02-21 16:17:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/28/826028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/826028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/826028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/826028/3"}}},"commit":{"parents":[{"commit":"a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8","subject":"Merge \"Add generic network interface management in the amphora\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-21 15:53:28.000000000","tz":60},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nSpecific to stable branches \u003c\u003dwallaby: additional patch in\ncreate_amphora in v2/controller_worker.py (spare pool feature)\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4806a83b3fa142d1fbd522add11f453199f2a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4806a83b3fa142d1fbd522add11f453199f2a77"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
