)]}'
{"id":"openstack%2Foctavia~826029","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I6619ba117f7051fe44086389789bc6e2810fd23d","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-27 13:05:22.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I6619ba117f7051fe44086389789bc6e2810fd23d","subject":"Add keepalive for redis-based taskflow boards","status":"MERGED","created":"2022-01-24 08:36:12.000000000","updated":"2022-04-27 16:57:42.000000000","submitted":"2022-04-27 16:55:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826029","meta_rev_id":"1b6d16e00428b429d3c708c45f8c6ef843287350","_number":826029,"virtual_id_number":826029,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 16:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-15 11:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-14 22:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-02-21 16:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-15 11:44:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 10:53:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 17:23:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-17 11:50:59.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-15 11:44:07.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"fb050e255369c2bb72a44598e6b9714115c2f478","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-24 08:36:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46898080f958fc6a95757f1e383f013b5a8c35ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 10:53:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea8e8a6f05149adb36cce708454e799 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848f4d31c85a4429bbe0138503710931 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0444e6ee34bf4ab9962b447c02663702 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a0edf354f684d71954f41eb77b2fda5 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3439b2a4cb2c4c0db1159584c997a945 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f86c1eb86fd54cbb86a92864ef849017 : SUCCESS in 6m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/09868de870d4434aa9f01330a008c227 : SUCCESS in 7m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0856a5a8751b476e99dd93e25758db52 : SUCCESS in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea46bbf197a940208eed918a8119c104 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34cce28a258242e78fb9778ae8a048df : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/08ed324ee852400f9955aad7e84e0bdf : SUCCESS in 1h 37m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07b633fd4c504752a7d9b199aa33e941 : SUCCESS in 1h 27m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ec10526736c4bdba2f7264b32c0948d : SUCCESS in 50m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aa69faf1b4844869b79e4f1e173192ae : SUCCESS in 30m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/069d0f62b78d47e6a2ad5702249c0eb6 : SUCCESS in 43m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b141c8db9f7c4f0685a82955b30d1138 : FAILURE in 41m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4d6d39477f949e0a971f5c42a45e090 : SUCCESS in 1h 30m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d1fb3cb26014e7e9ab04c035c99c3a1 : FAILURE in 37m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d444803752a24d6db1ad1c36ed7dc1ae : SUCCESS in 2h 14m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d4cd47a63a9f47f6beb7270f359ea8b3 : SUCCESS in 1h 31m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/90efda5f75ea4642b4f0f4e704a9b323 : SUCCESS in 2h 01m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a339acba1ecd4d9297a1a747a6bc13ff : SUCCESS in 1h 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76245042c2075a4f8631c3a123b8c0abafd0d788","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 07:50:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b754d7d86e3acf3f723cd8d2a001d5a6ec5f7a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:07:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea6b287a214e49cd88a4b4045d06be06 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11bf0646ef8f449e9c867a45941045ab : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d5e8ceb8cc4daca5fff26194d53978 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/397cef51ccdc4d96a2a3722d4458ac35 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8914393b6d80492ebd1199244137e333 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dae9ca0fe00d42418d3fb25631d9940c : SUCCESS in 6m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0960f982ab654915872d81839b3626d4 : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ce577720128d426bb0e4ff4512bbecef : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ad7fe2171ef4b24bb5ccbd1f1a723bb : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f03efb19cc54529b1a84b9c9d82d871 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b44f14c34ab43228f0d11485b975183 : SUCCESS in 1h 25m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2fbd9dec85b40abbac48c98aebf1e57 : SUCCESS in 1h 54m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a767fb189fb94606b3fa8e8e95dbd193 : SUCCESS in 48m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cc268b84b6134a26b1ab5de89ca18aee : SUCCESS in 34m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50118a5d1f5944f1a0cadb9f5870f6ef : SUCCESS in 43m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87d949d9c50f4c69a6cb077b933ad5dd : SUCCESS in 1h 17m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8975038a980e4566aa62ccb0b53e2059 : SUCCESS in 2h 08m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e926c0c89cc84b508459aa1f4334b5a9 : FAILURE in 42m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2bfcf48ad4954b13847b03b4afa400be : SUCCESS in 1h 42m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/89eb1ee986134e7a80030c1a531f3764 : FAILURE in 1h 44m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4f97be88f7ba4420a17b3c3cecc2d661 : FAILURE in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d6c4abda805c4503af4c5dfc601f2fa7 : SUCCESS in 1h 59m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1ee5b48e076e8a01470b1961fabbfa1b3047a36","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:23:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49099b1d7cc98b103030fa36edc3ab7f4b91c1d9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-17 11:50:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5979e6767c48f9f256213c21c88c2eff2d709407","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-17 15:59:15.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d8bb96f2a335dcbaa01a4c2272ab08274ead07fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 16:19:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b91f00658bbb7a253b751071ce7bc2d7aacf6672","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-21 16:20:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"635b720bc61d136938618ceda3aafb8a751a325f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:38:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1b5634fcdb44766bdde8403e469f956 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a332e4f91a284b8bb96c912a48763f3e : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/302dc463a926407a8d930aaddcd4bb7c : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f61d96bfa9c44e16a6ef07df39e0793e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed464a4962d4bbab894f7a7ee031455 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d29a041f594f969017d333436ba5be : SUCCESS in 5m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e4124ff020644c1ba618a27057d91517 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/58c4e7c3fe6a414ca034c1761c9e2219 : SUCCESS in 9m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/718c1a3332a640b4bccdf66fbf6c5989 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5371e8f9dd143f7b0441208bf1ec3e9 : SUCCESS in 9m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd64196c734648c0aafccc0d8b699abf : SUCCESS in 1h 16m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4781fe2cce0547c4baff45f07f57431d : SUCCESS in 1h 56m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ffd1692af7245ab8b3b952d3038fdb4 : SUCCESS in 39m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4adfed5a8ba74df1996b14efd8eab7c1 : SUCCESS in 40m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19c6ea017b5143809a9a8f906d056860 : SUCCESS in 38m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c926783420394ec1a63f4dfc6294578b : SUCCESS in 1h 09m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0df721429fb549b1bcbb629d5360e813 : SUCCESS in 2h 14m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/033fdc3dfa9e4f06a27e5e0d39ae79bf : FAILURE in 28m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a49b6b81d81444e0a9cc7721595fc109 : SUCCESS in 1h 26m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8ad48e136c2d4a27a9b94bc0aa160da0 : FAILURE in 1h 32m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/42767d9ed5f54d6b8b0c994736721d92 : FAILURE in 1h 44m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/08bad84e52b74189b91abb0ad86a265f : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a666e1a7cb410d2b1c5c4abcec9e596a8177a569","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:52:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"942b3f74a936f925ac79eef0954bddc56c04be5d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-15 11:44:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df05ea2c1535aa1e27e568e06e884ea25e228710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:44:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ae7b78107322b26fc15aa7f4b9522cdf54a386cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 12:10:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c322e433a8241df8a027327772804ec : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f520378f6e9645ce9e3f89f74eee57c6 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5373c98014a04a08a21d9c44b027ad37 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1152f3292a4afb89a0071e35bb7725 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8051ea478cdd4a08af8d384c744e1e61 : SUCCESS in 5m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/edaf19beb0164be0a0f426f49a259868 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20fc93235ae2441b94fe21c1c487beb9 : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/264c83ec158e4d219400b6060c8ffb0d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f93cd32d0c0546efbebfa21f596ebb79 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8590eeb4548d4c6aa8579ec60fb2be83 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8326b6e07bac43f9a2eedde011a55e5d : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1fe8958d05e43939a6f3c85075e7b18 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fe945d339b1c4e0090372c636ffb10c7 : FAILURE in 9m 41s","accounts_in_message":[],"_revision_number":2},{"id":"76afa58138cd8a135ad5b3c79f37568f584057f0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 13:05:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b0c14c63f404ed8c807dba01b64cc79af15cf0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:14:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50571ce53ec0481c9f0bc2cbee9eb38f : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/059172a83ad2426fb377186ad92401f1 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66039da205064544b7f7ea2e1b2e5c42 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/485092d58b1c4fd08c551a757f1b5eba : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2474adc6a744f081cb57cb4cc1d404 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e04ab8b86725455989d0c0518e2f8f04 : SUCCESS in 4m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7b061b9c36604d9588c0ce56cef7f052 : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fe3cbea1a3b846a9932a17e8573e18a9 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e1528f39d3594e618454e68f7fa1e38b : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc45b62809cc40c69b90b6265e036912 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c84baf9d36cb448bbb0ec9f07cade07d : SUCCESS in 1h 28m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/095814db56e44f5290e6ceced97f102e : SUCCESS in 1h 32m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/61cc5e2bdc114c5faa5093e3b377ad44 : SUCCESS in 40m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f98923939318415b84e533fe7693337c : SUCCESS in 33m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dc6a4bc545b4e38948508774d319dae : SUCCESS in 33m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07057223c8a146cdb75ed509c063688d : SUCCESS in 1h 42m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6425f4bed774bc1bc207c1a97f54ddb : FAILURE in 31m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5c6c498a4d4746eea5cc719eb250b62e : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a49d5fc92ac94544881a3a16352ea17b : FAILURE in 1h 31m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/287417321f2e4da89d63c60a3a9d6423 : FAILURE in 1h 30m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3cc92b87bc444e168a76ac1fbd8659fb : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19e971f5be7dc9d37c7812b403f8b42daf7aac0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:15:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2025236f224881f681e9eadec56fb329b1c10d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:55:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efcd98014a1942b092245f87d19f088a : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cf41e69b4640ff9e29c2e4d1573f0e : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4191d7b29fbc462284bd8a6f5f63600d : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b568d6bb004f4c45ad497eb9c7756615 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/341caada8d3f4657b5bcceee861c515f : SUCCESS in 5m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c0eb64120ee4c228054d5e8b22f37b4 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8fac29f6d6e473eb39b4790411b810f : SUCCESS in 8m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4fa7efa567c94f20ac83107dcbd5ca21 : SUCCESS in 1h 36m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/850829d2583448268302d31a75af23a4 : SUCCESS in 1h 33m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a38e2724639461bbd34460b9d7413e9 : SUCCESS in 37m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4ef022a5bb7a4d7e876d5268c604e7e9 : SUCCESS in 33m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd5ad544492846d5ab3dcffc1424450f : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f6b020ed6e676fd7bd99b86f733d6be1a7b95d1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:55:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1b6d16e00428b429d3c708c45f8c6ef843287350","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:57:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582b701ef57b45b3adaa97df669a9daf : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42203a94dc3a43db896dfa3c6db81412 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40256308f640a8a117cf164251003e221db5d099","revisions":{"2fa1680b723a2cf30d500934e4907b924bbcdd79":{"kind":"REWORK","_number":1,"created":"2022-01-24 08:36:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/29/826029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/826029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/826029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/826029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/826029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/826029/1"}}},"commit":{"parents":[{"commit":"de20868ab74e5d0b3b24e254b8c4cefae7b78fb6","subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de20868ab74e5d0b3b24e254b8c4cefae7b78fb6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-24 08:35:57.000000000","tz":60},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n(cherry picked from commit 987d6a346f9a7df5ec9bb3d341fc4f90423608ed)\n(cherry picked from commit f83120485c82f8713c1d3fd32fd808c9ea5b4654)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fa1680b723a2cf30d500934e4907b924bbcdd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fa1680b723a2cf30d500934e4907b924bbcdd79"}]},"branch":"refs/heads/stable/victoria"},"40256308f640a8a117cf164251003e221db5d099":{"kind":"REWORK","_number":2,"created":"2022-02-21 16:19:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/29/826029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/826029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/826029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/826029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/826029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/826029/2"}}},"commit":{"parents":[{"commit":"64c792c225a510f8c27a5aa27ea5dae261e0b76c","subject":"Merge \"Fix nft command line with negative priority values\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64c792c225a510f8c27a5aa27ea5dae261e0b76c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-21 16:18:36.000000000","tz":60},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nSpecific to stable branches \u003c\u003dwallaby: additional patch in\ncreate_amphora in v2/controller_worker.py (spare pool feature)\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n(cherry picked from commit b4806a83b3fa142d1fbd522add11f453199f2a77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40256308f640a8a117cf164251003e221db5d099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40256308f640a8a117cf164251003e221db5d099"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
