)]}'
{"id":"openstack%2Foctavia~826030","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I6619ba117f7051fe44086389789bc6e2810fd23d","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-27 13:05:19.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I6619ba117f7051fe44086389789bc6e2810fd23d","subject":"Add keepalive for redis-based taskflow boards","status":"MERGED","created":"2022-01-24 08:36:52.000000000","updated":"2022-04-27 16:57:54.000000000","submitted":"2022-04-27 16:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826030","meta_rev_id":"1cd600e308cd8533dd1923faa64117810b942d78","_number":826030,"virtual_id_number":826030,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 16:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-15 11:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-14 22:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-02-21 16:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-15 11:44:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 10:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 17:24:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-17 11:51:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-15 11:44:16.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"b31d1aba2e46879d85066585ec8b503d2149af5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-24 08:36:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57faa498c679eee3deb5863e57cbcf98ab24b5ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 10:35:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/474700d0355648c1b4b94aea81b9ebf8 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc590ced73c49118930a0ca396219b1 : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9df32efcb5a44cdb9eff0cec24b87cc : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/167b6c23b40d47cb96148ede674d4642 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0906534c79b340e99a9e3b9141e323f9 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22b91c8036ca4d07a923c3c0e19a1daf : SUCCESS in 4m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/34735a63464e45cab3f88a95348659de : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b521bccd637542c385246fbadb8a12ed : SUCCESS in 8m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd637abef5614deaa41044ea1e661828 : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/458ec932a3514f1891ee44a7c1fef147 : SUCCESS in 1h 02m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecfab697510e41119c0dce5784dc69a0 : SUCCESS in 1h 48m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed98aa8a9f7a45c183255f1eddbafc55 : SUCCESS in 38m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1a249383e3794486837adb6ee1899c61 : SUCCESS in 33m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42ba06efefb14f42be5f675afa43b368 : SUCCESS in 34m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a2f9a77ca7cf436b827dc1fc9276ddd4 : SUCCESS in 57m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b93b09186b74f049715a777f0c9233e : SUCCESS in 1h 35m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c65c4f23c34488995fa481d7286f9c6 : RETRY_LIMIT in 5m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/84ed0fcf7ced4fa989949072f7b484af : SUCCESS in 1h 53m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bb19fe8b81dc316df9d12af0b379f0ca45093a7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:24:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5abb65c92bf33c23e3a15f1970c23b5524a0db4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-17 11:51:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83f3fb722005442810725dc15a903f399786e361","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-17 15:59:22.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"35887f757516625948ba59594b685d5394fd5f34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 16:40:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a96348620d612b365d5808fa8a2561d2fa4403a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-21 16:44:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9981e40c02b7790bc4b8514eb7eecf92e9df161e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:55:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f99b283c80d44fa997d702fe84fe4dc : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1f5bb8615f4617803c55fc0874bfa2 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6cdeea9dee640bcae01b55ba2a016fd : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9f0db466bf94fba9f31ee06ca9c65c5 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5ba205c6b5471cae358706098648f6 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18afddc3156843a4aab5e32f7301135c : SUCCESS in 6m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b36eae83694a4bffbd9c8940a2b22aa1 : SUCCESS in 6m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/16721572b7884a4bac1f08048474dc91 : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26cd257e825e4b38a30fa122ca56d215 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a520969396fc4794a345e1835ebe479d : SUCCESS in 1h 43m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d93b052c7c6f43f38d9461b42b735f41 : SUCCESS in 1h 35m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f82247b0852e479fbae96efadd93ecfe : SUCCESS in 34m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/024988f8a7ed4652819426f1a6a6a8d6 : SUCCESS in 33m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a15b2fae338748de8b51c774aa503ea0 : SUCCESS in 35m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dcc4e39f0b6c4e37bbad4a936e1fccf8 : SUCCESS in 51m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6cd754a77044ebbaba62329b3955f09 : SUCCESS in 1h 27m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7bb733064a74863bb00a763d7a59e45 : RETRY_LIMIT in 6m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bf699c5ca518486a8d68bff0d2c9a9d8 : SUCCESS in 2h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0615b78d681383168edafa5d9871dd0bd5daa74d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:52:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae52c4c9a98f17decbe9555f878d92443911a9d0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-15 11:44:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"347a83343cc5b6f16c3b1e42fc73bcd324be682e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:45:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f7a691863d8384aaaf7257b69bf1bfe870d394c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 12:22:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ada5ee050148f180d8eb7be728beb9 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6760866f4a74d23b18deb46969e77e0 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54ff1c3bf1114a918f1db053fbd6b1d0 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2335464b0e042d49ff4e284342b6a11 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b296f1715db4bfc9ba8ac8dd6e25db0 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2d576e4cc49443de95ca4fa974b78a72 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3053fff130654f178ece042e1c5f5949 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6944341901654a62aeaa147e52c493e9 : FAILURE in 9m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d1c7989914d448b29b0760515bcd3e65 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0dbfb68b123b45dfb1701db0e0d8e1a8 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5ce68e0fbae4d58aa54bf4e0b1d039e : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/220fa9380ca7436d802041e324388a33 : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"70123683eef2523440520e551f0cb66bd6f10633","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-27 13:05:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00720b0e4a329de74e6134cfedc39838bcfd1579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:16:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56fd9bb63fac4f56a5618af88e666564 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c564d337354179919b6575c9aad2a3 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b72257900946399d32c5fca6b8ca8e : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9438c5fa1a2944b1a18904c98704a5f3 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b64f8b3797d41feb953ae2c9f198c75 : SUCCESS in 13m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f80620b8b8034e5a89546d72953cb54f : SUCCESS in 4m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e16b68df4e714f7bb5c5cc51f43fa05c : SUCCESS in 8m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2af92911eba14c84a66bf27bbe5af8af : SUCCESS in 8m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27a00a22110243d1a59a88dd59b2beaf : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f90808aad3643979e08cb3e69ad131b : SUCCESS in 1h 23m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1bfca20c38b48b0a2992e1894247e98 : SUCCESS in 1h 44m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/708723beabee43539f174fe2fac34553 : SUCCESS in 41m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/501f8f3eb5624facbdb826aa0073b423 : SUCCESS in 38m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/111cd1c6528f46929fbd0ae036df479c : SUCCESS in 39m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/afe14211c31d45c79f1bf5feaf73ec51 : SUCCESS in 1h 24m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e076881c2314f3c81907fef3d10b3d2 : RETRY_LIMIT in 6m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cca9ea72bb244062bf23692524f78285 : SUCCESS in 1h 49m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"083a4335575b8fbc7bfdfeb873ca889a861b5228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:17:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e75a57b2d92414828b2cafec10721158b2c24a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:56:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9255d2d230b34a35a3b6c254cdbe24c5 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43b809870f3e43f1ad8957170808b23c : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3121e3da500944eeb50dbfb0889b91ca : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef00fd2b15f44838ccb623c10c8e7a2 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6b1090f95a546f8b7c4be594278e32c : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5dc830633ff24dcd85bf7bf5af25c66b : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/881363bd2a344db09298560d94fc814a : SUCCESS in 1h 03m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a5792b848204e47954a57009db6a7db : SUCCESS in 1h 37m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76347a60dbf84a26b75a3fd6fba72ff0 : SUCCESS in 38m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d8280cdfe3874620afd8d38db61675d3 : SUCCESS in 38m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1f26b3713eb4c8e97a9305d7db23616 : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":2},{"id":"99d3cb5c0ab5798d6b3cd2aee1b5c8d4b0a0c3b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:56:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1cd600e308cd8533dd1923faa64117810b942d78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:57:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8362647063354fb4a041c59c37f9f846 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca48d55f49a54bfaa131e7624c120918 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bd125d73c54f0edb25095ddeef4c25161a5377b","revisions":{"d5f0892fe377b9366b2e9afa91cca7700d34f3f3":{"kind":"REWORK","_number":1,"created":"2022-01-24 08:36:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/30/826030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/826030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/826030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/826030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/826030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/826030/1"}}},"commit":{"parents":[{"commit":"0d25fe5fff0ab4c2815e8ec2fc9da7c8fe70171b","subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d25fe5fff0ab4c2815e8ec2fc9da7c8fe70171b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-24 08:36:38.000000000","tz":60},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n(cherry picked from commit 987d6a346f9a7df5ec9bb3d341fc4f90423608ed)\n(cherry picked from commit f83120485c82f8713c1d3fd32fd808c9ea5b4654)\n(cherry picked from commit 2fa1680b723a2cf30d500934e4907b924bbcdd79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5f0892fe377b9366b2e9afa91cca7700d34f3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5f0892fe377b9366b2e9afa91cca7700d34f3f3"}]},"branch":"refs/heads/stable/ussuri"},"7bd125d73c54f0edb25095ddeef4c25161a5377b":{"kind":"REWORK","_number":2,"created":"2022-02-21 16:40:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/30/826030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/826030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/826030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/826030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/826030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/826030/2"}}},"commit":{"parents":[{"commit":"1d8ac7e2c13a1ba5370eb8708f5eb83f72d15a5c","subject":"Preserve haproxy server states during reloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d8ac7e2c13a1ba5370eb8708f5eb83f72d15a5c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-04 08:31:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-21 16:20:30.000000000","tz":60},"subject":"Add keepalive for redis-based taskflow boards","message":"Add keepalive for redis-based taskflow boards\n\nWhen using amphorav2+persistence, taskflow maintains the ownership of a\ntask using a TTL in redis for each job.\n\nWhen a task is longer than expected, it is released in redis and another\ncontroller can take it over while it is still running on the first\ncontroller.\n\nAdd a keepalive mechanism in the worker to periodically reset the TTL\nwhile the worker is still processing a job\n\nSpecific to stable branches \u003c\u003dwallaby: additional patch in\ncreate_amphora in v2/controller_worker.py (spare pool feature)\n\nStory 2009761\nTask 44242\n\nChange-Id: I6619ba117f7051fe44086389789bc6e2810fd23d\n(cherry picked from commit f13c3ca6b8480792e2ab86a5413843b56eb453c6)\n(cherry picked from commit b4806a83b3fa142d1fbd522add11f453199f2a77)\n(cherry picked from commit 40256308f640a8a117cf164251003e221db5d099)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bd125d73c54f0edb25095ddeef4c25161a5377b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bd125d73c54f0edb25095ddeef4c25161a5377b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
