)]}'
{"id":"openstack%2Foctavia~826769","triplet_id":"openstack%2Foctavia~master~I550c7d2563a2d3ada07e2162383f59f104bcbbd5","project":"openstack/octavia","branch":"master","topic":"c9","attention_set":{},"removed_from_attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-02-03 23:12:37.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-03 18:25:05.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I550c7d2563a2d3ada07e2162383f59f104bcbbd5","subject":"Restart rsyslog from cloud-init","status":"MERGED","created":"2022-01-27 19:12:29.000000000","updated":"2022-02-03 23:14:23.000000000","submitted":"2022-02-03 23:12:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"826769-c9","meta_rev_id":"bbd8adbbc3dff434ea2772458879f4b89f46390e","_number":826769,"virtual_id_number":826769,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 23:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2022-02-03 12:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2022-01-28 01:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2022-02-03 12:39:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 22:11:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 01:13:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-03 12:39:55.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"4e4dc0edb9eabc280d06a42c810691dd474319fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 19:12:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad48c9f11551193ba13b4b8e43033bf6db4f6c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 22:11:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/261393d567f94fbba77a9073c92edb25 : SUCCESS in 12m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05819504c8f1493c9f8c9a0f07104536 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab6e34cdc6f48a1b073cbed3100507e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d10a65800e94a548e09ab33da7548e0 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/312e05fbfd1542bd8791e8de53766966 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f7266bea199433eb562dfdb6b9971bc : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a0daab80d14018b4e2806b703f084a : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/304e65f33163401db7b108593e81535a : SUCCESS in 6m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/551f599264c146b08f78c21fb899fdcc : SUCCESS in 6m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/76cc285af276499784aef7414ec5453a : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/81e5933927384aecb2f9b73a3c06bded : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/827477835d9348a9b3c64ed8c0532249 : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e04e2df2e96f4d4a8df717d4d1fc03cf : SUCCESS in 2h 04m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a44595cdda94adfb22d7eab5dd47b7d : SUCCESS in 1h 36m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1506ef509b049ed8bad310764e79dbf : SUCCESS in 37m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f80521d09bdb479b8f6fb684acb81fac : SUCCESS in 46m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c61d6e2f22aa42f69bd0d5607fe6e29e : SUCCESS in 36m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5bd932fd75344400a149f9df311e9159 : SUCCESS in 1h 25m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42ae68eda8d04cf99c7011b11d06f0f9 : SUCCESS in 1h 51m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eab1d0b9697d4f0f8cd39711d1bf7b14 : SUCCESS in 1h 39m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/af959dd01920443ead319cefd164b40b : SUCCESS in 1h 22m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f39b7f0a860a4c7988ea535366a20f88 : SUCCESS in 1h 22m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d60b4779801840b7a30e3080acc81f14 : SUCCESS in 2h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"586ed88fe830070309717cfdf893eaf4f88f3446","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-28 01:13:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67bfdd52ebacde5932ead8b5f6f026ba1f249d95","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-03 12:39:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"109d1a75eafa95cc47f5ab0a9d54b9539c9f1657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 12:40:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8d9a3557789134eec174d7f3deaa995d1832140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:28:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f269fe76d57c44c2a6ba4e1241fae8ad : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdaf525fde6c43ab8beb63394cf89efd : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1023b30a09614626ac5f523c7f9c49c8 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10dea6a6ccc5496682695a605c29a1a9 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0137c6b5e6124f4eaf4a324d939d9faa : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59cea28c50b24ed7a8df603befae97d4 : SUCCESS in 6m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f8cb1d833bd46a4adf5a76ce4869bb7 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5e868bae56147708388d530cad31cf1 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56b40f70a1bd4c5abf19ca08401af93f : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/977f14b1048f48698db98a066d9df5c1 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3e9da64a69e47e5a7675a877fbb04e3 : SUCCESS in 45m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5ac187b093c0445da8947792da6a989a : FAILURE in 46m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2ecef5c0cef8b49c76c68b430b74165728a60687","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-03 13:48:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"878eb06246f5598b215fa56c1c875871d0d1b925","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-03 13:53:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3090274b685d5d60dd69eb1fed8a8a22ee7740dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:15:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/056a1d3a9074440dab318ca5e3815d7a : SUCCESS in 13m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55d3c612cf5a4b89b312ec00020e0973 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5236532e1b54e81ac2d9e9560b5b509 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c21b93c59744f6d83b68e09acdc787c : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df129d3f1a654e6681c88f1242bd9bcb : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fd9e3e8f11840d5aa34691e8d9672ab : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864adb3c9d094b97bc17fdc2bc97bf3b : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5cdc91c5c74d108d5266c080a615e4 : SUCCESS in 5m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ed46adcc51ed4bce852b7071a1744f68 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b801fca256a440abbde0374511a1b5e7 : SUCCESS in 10m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/daafded2fa0b45dd959a81be14863e1e : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/51ce716e618b49d991d759993a530de0 : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10347bc3118e488da154e377e04fe091 : SUCCESS in 1h 59m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0f44fed654f4eefafb7867453c49321 : SUCCESS in 1h 31m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f24a11a55604ac19bb618064f84c84d : SUCCESS in 36m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa86c614b5874a4e9b49eed465241e4d : SUCCESS in 44m 33s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/4373b6b87b6e41cca0b8bc067b800e4e : SUCCESS in 2h 21m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/47e8c4b63c064ab1ba3d4476b01806f6 : SUCCESS in 50m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/916971d7eefc44738759bf1781e7b5de : SUCCESS in 34m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f7431c20d4b4b44a9d21266fed5534d : FAILURE in 1h 42m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3390c8f086e4ba2b480c8388f87649b : SUCCESS in 1h 39m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/91f70010ef424be5a04cbea67e9f5788 : SUCCESS in 1h 42m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5101e3bbf3b54fa6a7a7090ef2d215bc : SUCCESS in 1h 23m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1e29ce6178b84036b780d8b1953fa6e3 : SUCCESS in 1h 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/33171ed4ed884b5b9d791289f276e421 : FAILURE in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b0b1ba4643622127fae69c6ec6a451fbce28e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:17:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b9f87d4c8dfc980b73312cf41b8ecfe47a0ffeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:46:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31fd7aa9b3e04a5dbce5c5a1ad7fcde8 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ef84da43114d918406d72d4376f507 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/978f96cddb0c471baf5fbe1db4b76d98 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35cd3b7adcf74097abb22782fb1953eb : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9370cd9330e942488c1d69d4a6ad4aeb : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae0af99fe9f458da7fdb4202c0b3ff4 : SUCCESS in 5m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cc8d9fc4cc2e4158abbb16da153ca1ce : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f97b5c0b5a54f9687c5ebd40fde9efb : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e973c0680414831bb21f9c779e7cf5c : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1469a9ab0fbd4f89ae2b9d94f2a16e44 : FAILURE in 1h 28m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9af96b8bcb7e4d1fafd7cc94c6838ca5 : SUCCESS in 33m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/457777140f664c3fb288cae571f25a10 : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":1},{"id":"dc23308d72d0007f9f780cb8f1dbb58771d0a229","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-03 18:25:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04f35893269922dd715a75d60de2095cdaee953f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/210f9dc1594e48d498ce75edfdd2b995 : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea26c99f4a8c4c8faa4d1c37455a007c : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a3f730c7824456b0110269302b4607 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26a1fa70241c4ae5b71bbfc10cd35090 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/643f46a0ac6942e4af361d13b773dea6 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37273d163de14f1c87f91a0f11a9bb3f : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace4d29eeb4c4f3ead61df1302d81956 : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccf00c2f69bc43d2bfac1119d91ecca4 : SUCCESS in 8m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/52d9f41d3b95424eb83ac80c54e241fd : SUCCESS in 9m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8a971318c61d454b96f7e533875b1fbb : SUCCESS in 12m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e075282da4af42bba8a065d7d6794742 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5915843408824f8fa4c340e80327ab30 : SUCCESS in 9m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1952b522a661484a823f6190ff6d3862 : SUCCESS in 2h 12m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e2bd70b141941c19412b625da40c2d2 : SUCCESS in 1h 23m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f4fad70058dd46d88d1f8150e4af21e5 : SUCCESS in 50m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e224fbd05c44eb1a688ed32703246d3 : SUCCESS in 53m 26s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/5ff626ff4678433eb8460cde79b6ea93 : SUCCESS in 2h 24m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f54840ede080473f96352395195f6a04 : SUCCESS in 1h 06m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbf87f9569844f7a8173cddbd3ffcd9e : SUCCESS in 48m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7dc55eed0da04c5fb68c0205b946a6f3 : SUCCESS in 2h 03m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa43ae8ec51c400e8dca70e0af23e222 : FAILURE in 2h 15m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/75a1539eb4d74ab799c68bf1219d7217 : SUCCESS in 2h 06m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a6acf2d847a45cfa8c57d7d308e34dc : SUCCESS in 1h 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3b82d04a162344f28722879e597224b2 : FAILURE in 1h 26m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc39989a2cff44c895dbaa1626e98eb6 : FAILURE in 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df14c4260342b7c1c3e96848456d423a157ddc9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:59:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb52a8b57daa8e6b2786b445a1403b0746be9c80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 23:12:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95ba6d621d7848f6b65ea0abe03952a7 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f09717fa634700b8273b5686ea1814 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fae57cdb3df64112bed07483273686c7 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f06fb3a74534d9aad59e81701a242d8 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6211f08e0e8436a8c83a1b8874526c8 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567a4751bbc14c6e84690fd08ddb0132 : SUCCESS in 5m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/74ef54b92aa4419ea275bcaf310444b6 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c8e0061e8bd4e3485854b0ded6100a5 : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de18aa1332de485e824c0117d96f8eaa : SUCCESS in 2h 06m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9c0deeb1a8e4abc85fbc63c8bbf45fd : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed91acea1c2a4be6aa9704fe2ec9a572 : SUCCESS in 36m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c08d11d79cdf4921b3342c5b604989e1 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":1},{"id":"149973719eca8d4d1c69c31225682b4e8f2e94dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 23:12:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bbd8adbbc3dff434ea2772458879f4b89f46390e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 23:14:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05455d4c085e47d5ae792e1b560681cd : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e67b620fb6a340de9f33a3b928b7bd8f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbffae6eb684629a19844b5f2f4b7523b77a725e","revisions":{"cbffae6eb684629a19844b5f2f4b7523b77a725e":{"kind":"REWORK","_number":1,"created":"2022-01-27 19:12:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/826769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/826769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/826769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/826769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/826769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/826769/1"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-27 14:19:00.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-27 19:11:24.000000000","tz":60},"subject":"Restart rsyslog from cloud-init","message":"Restart rsyslog from cloud-init\n\nOctavia uploads a rsyslog configuration file using cloud-init but\ndoesn\u0027t ensure that rsyslog is restarted. In some cases (CentOS 9\nStream), rsyslog is started before cloud-init and doesn\u0027t use the\nprovided configuration file.\nThis commit ensures that rsyslog is restarted by cloud-init.\n\nChange-Id: I550c7d2563a2d3ada07e2162383f59f104bcbbd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbffae6eb684629a19844b5f2f4b7523b77a725e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbffae6eb684629a19844b5f2f4b7523b77a725e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
