)]}'
{"id":"openstack%2Foctavia~827330","triplet_id":"openstack%2Foctavia~master~Ief23b3fb6f4cedd2866e34c6de184e643d6789ec","project":"openstack/octavia","branch":"master","topic":"c9s","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-07 16:07:18.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-05-04 17:21:03.000000000","reason":"Tom Weininger replied on the change"}},"hashtags":[],"change_id":"Ief23b3fb6f4cedd2866e34c6de184e643d6789ec","subject":"Switch centos-8-stream jobs to centos-9-stream jobs","status":"MERGED","created":"2022-02-01 17:30:19.000000000","updated":"2022-07-07 16:09:37.000000000","submitted":"2022-07-07 16:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827330-c9s","meta_rev_id":"9679305d685bc006b0af017018630833f882f3e3","_number":827330,"virtual_id_number":827330,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 16:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-24 21:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-07 13:40:28.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-07-07 13:40:28.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-01 17:34:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-03 09:45:43.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-24 21:57:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c4a3e2bc04d4338451649fe9762cb3f9d4fb9f01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-01 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89804da3161cd1644dac732344fac27484a8859d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 17:34:49.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job octavia-v2-dsvm-scenario-centos-9-stream not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python3-xena-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n      check:\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api:\n              irrelevant-files: \u0026irrelevant-files\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-scenario:\n              irrelevant-files: *irrelevant-files\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files: *irrelevant-files\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n          - octavia-v2-act-stdby-dsvm-scenario:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-cinder-amphora:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-two-node:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-ipv6-only:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-amphora-v2:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-amphora-v2-no-jobboard:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-centos-8-stream:\n              irrelevant-files: *irrelevant-files\n              voting: false\n          - octavia-v2-dsvm-scenario-centos-9-stream:\n              irrelevant-files: *irrelevant-files\n              voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-grenade\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-8-stream:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-9-stream:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n          - octavia-grenade-ffu\n      experimental:\n        jobs:\n          - octavia-v2-dsvm-scenario-nftables\n          - octavia-v2-dsvm-noop-api-amphora-v2\n          - octavia-v2-dsvm-tls-barbican-amphora-v2\n          - octavia-v2-act-stdby-dsvm-scenario-amphora-v2\n          - octavia-grenade-amphora-v2\n          - octavia-v2-dsvm-cinder-amphora-v2\n          - octavia-v2-dsvm-scenario-two-node-amphora-v2\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":1},{"id":"d62cb2b969a467637378e76f260b78e845267602","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-02 05:50:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81b17a82308d1d3df7e7ec6ba4af79aa2479a81c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 18:03:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3467b36608c188947d84b1a265b114dbd6df9db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 22:40:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d81bdf51028146729287f37b0df07f4e : SUCCESS in 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6bce4412dc24b619a58c0e3ef27b4e3 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/561b81ced036431d8f98b5a71d289bea : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d05d9b28d6b145288c5eacb1183b18c6 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dcf2900051648db89e81f106272ce63 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f74b4cde2da74dc788a84b55369cce8e : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f48f41261154b65a9023e06e62df2dc : SUCCESS in 12m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b2b9c489463e4fa68a5d430ea9fc6bbe : SUCCESS in 8m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/72b4a31f2fd147b7965711e9b6536fc5 : SUCCESS in 10m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ef206532fe04e91b7aacbede58ab672 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/657aeb0f5c1f412186dc863556dc0c31 : SUCCESS in 9m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e084a76ba317487da75b18ae03a09b74 : SUCCESS in 2h 03m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa89fd6371ff4a48a72a1ec9f6928c62 : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5039f75c21bb4882b3f05f30906f8610 : SUCCESS in 37m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da06f42bcdae4144aa7cd0b10000887e : SUCCESS in 1h 03m 09s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/d05cc8183d1947e789c496c653490377 : SUCCESS in 2h 15m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/64911e21fe34482eb35ec4de0587df66 : SUCCESS in 53m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6cd7a89a4a84d7bbd93bccfd19c7c4c : SUCCESS in 55m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ca454f8e976249a08c6434f8b1ad2630 : SUCCESS in 1h 58m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ef7f63592514459a5d473e060fe7325 : SUCCESS in 2h 07m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d30d8c69d913405b82eb606c9b203ec5 : SUCCESS in 2h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/61da6fdf8c3144c6849721fb69aca3f9 : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0329b70d3588411dad7520de2bb5a91e : FAILURE in 1h 47m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/b92165b529ce4c4eaae9cc4aee5c8482 : SUCCESS in 1h 49m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e99e2e66b991475c95fc5b8a3336d709 : SUCCESS in 2h 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83c19300e195d47c0bb43c94e81b9a59118e332a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-03 09:45:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8660aee2bb2bcc5a44fde91d965b9aa6286a1179","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-12 13:09:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eeecb8911c10609c2b92d19bd047fa9cca94e357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 15:54:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67ce34c4fb53434493f903f16096ebe3 : SUCCESS in 11m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb9efec8132245e18eb5029523a0ba5d : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6149bdfb9a34709a1233e22f0b0d23a : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/401756db25a945a5a8bf4f6100ae86b1 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1edb2e2cec1c40c5b02d57264993c021 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c770b52ea38456c903f7c1f2273d2bd : RETRY_LIMIT in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d8e46bbfec42858b056b5d23ea5a30 : SUCCESS in 13m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d963c820fbc486fb9749a6bc840fcd7 : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9314dcfdb5a74903999b561fc2601e2a : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ca66c16392a04760bfba27322ead6528 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe0cc359ee614709b609c3d0abc25ffb : SUCCESS in 9m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42520e34e0af4420b593e1caf28794b7 : SUCCESS in 2h 10m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24886ef1fc3e41acbf5401edb0e38745 : SUCCESS in 2h 24m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51d3a3964aee4abcbfcc87e8a33d6c6f : SUCCESS in 34m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1d8cff4eae2483386a62e9b149a9aec : SUCCESS in 53m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/23054f15d5604effb228da20adfa3537 : SUCCESS in 1h 03m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21214cdf76b840df8150bb2224350bb8 : SUCCESS in 34m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61848f24bd3441b08d056623882610d5 : SUCCESS in 2h 04m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f387186415274455bef1a18f0c94315f : SUCCESS in 1h 50m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/42d271eb8e7847ed8ef453ad301fdbf2 : SUCCESS in 2h 11m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/363b461a948144a1b236ff2a618ebfab : SUCCESS in 1h 47m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2da453e0c15f4aa18ec4f11fdbb8ef86 : SUCCESS in 1h 20m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ece404094d184811b28e3f11639a4953 : TIMED_OUT in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6c3de764bb14c4d9c39da22eb3c8ebc : FAILURE in 1h 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7089c2524a00b73eb920c618bdc4bfdf5ce6eca","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-25 09:28:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85cff663a324442ddd2b211ff6c5bc203f8c258a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 11:47:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d170e77490542f695d6c536c2c23366 : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f009172fd5a24e8eb536fd5e64bb259e : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c39c62ba60b4d00b3e3080d0775cfad : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52392ffcc20143aa93f702f7adc316d6 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13fba8ce753b46659b1fd6bf383a1dc3 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b1b8062b0a749ab907f306320d8fb6f : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5585dea96c3549769f55a14f742e8f30 : SUCCESS in 12m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4000f8df6a434d93b9e0c5d126e8df4d : SUCCESS in 7m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b9c073d1cb8246ceb4f0975cc0d8d16c : SUCCESS in 10m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2fdf1edfe0d8485c83148d0e7bd36b3d : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb5572f5fcdf4bfdabc2a169253da75e : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e31d374b8a8b43e7bc2c42d3b91c16b7 : SUCCESS in 2h 13m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7103f4f4da04658a34bcf87cc1c31a8 : SUCCESS in 1h 17m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01d3e88f0c9a4412b8c6994b35cfa3bd : SUCCESS in 40m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bab7afaa20674f71ac00049e51cd8abb : SUCCESS in 1h 04m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1f58f654ed084c1f8d16168a91f58a8a : POST_FAILURE in 17m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cb86cdd926442e8b1071274706f4373 : SUCCESS in 36m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/136c002908364224b9a60e3e2406630f : SUCCESS in 2h 00m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/33b842df528847e5ba62f2715979225b : SUCCESS in 1h 54m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c542a310a1b94189b0aa0d2db0d27c6a : SUCCESS in 1h 41m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c9f803e6d75f41a39ec5047976331bc9 : SUCCESS in 1h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/145a41020239420d915d9b33ea1c3f5a : SUCCESS in 1h 16m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3defc671c00248a6b95862de93eb1c6f : POST_FAILURE in 15m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ca0546eb56d410fb4c7cc3aeecaf7cf : FAILURE in 1h 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b99028c6a330ee1597ec38aa9ff474b0570e7a1f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-02 06:19:33.000000000","message":"Topic c9 removed","accounts_in_message":[],"_revision_number":2},{"id":"84508223e808cf94b4b15abe29ef561c269fc5ac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-02 06:19:37.000000000","message":"Topic set to c9s","accounts_in_message":[],"_revision_number":2},{"id":"1a7155d18681fa56c12e5ddcff097e1acfa03536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-02 06:37:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"229ae2990cdf8a33e43b1fc2d87cf00ed20fe378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 09:15:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4afb57b19f7f47d59457ec9e8363d50f : SUCCESS in 11m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b7dbafb34b6498288850bd7f252a22c : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf293d96937c442eac4518ec994ab4e9 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b992b35b49f488bb81fe1fc28af2082 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4762b2c4bca34a9fbdf5d4b0082eaa67 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/448d1fa101324bb0a8cd25919fee7568 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1efd54855924454a761ffa275ec2997 : SUCCESS in 11m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7eae29a4a91342e99e746e38e8cbda2b : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f75e1919d884fa9b51e4822b17968d8 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/84057c4ad0124e01a8fa21e9dbcc4410 : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81269050aff84703b59782940e74ec6f : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d4d077cb694448c89326abc24be4927 : SUCCESS in 2h 30m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d23de4889004aea9652179a65d5e127 : SUCCESS in 1h 40m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a73fdd1f43dd4a1c900d20ac6d396dcc : SUCCESS in 36m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/22a9e2b0f642467faa25b447c6810cb4 : SUCCESS in 53m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/76fbecc0ea584d8f8867fc24dc223bd5 : FAILURE in 15m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5797cd2b92414501bf4574ae17d7ee11 : SUCCESS in 35m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34ee604df1874f029aa49ac34205642c : SUCCESS in 1h 28m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43d442354a9143edbcc70a24d0ef5ad7 : SUCCESS in 1h 41m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e4794d2a61204f0588b89c55b7b554d1 : SUCCESS in 1h 43m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b352099c967547749054b33cfb19544c : SUCCESS in 1h 50m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2755c7c52d654b3b8d60b1923be4a9b0 : SUCCESS in 1h 31m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1de679711e24f228356680f7630bc8b : FAILURE in 1h 04m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e88c4a698f7a74be389b50a7193c37ec7e08cf4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-04 17:21:03.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d0c2bdd5ba66e561ba16b75aa7c092bbce0f3037","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-06-24 21:57:22.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c1a5554f50c2b0bd80ba56ae52b1deca71e0451","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-07 13:40:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f927154317c4d7479f0a54d52a7dd09d753eb877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 13:41:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c33841454ae431d7e2da37e476abb70f422ad3e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:07:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e191d0ebfa0c42bba2dc4694bca9da8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3f5854739943df81d127ea41f328c5 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2c9f89cd05f48baba8a2c88455d260b : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25f247485e5140e1af685f442b26e6e1 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a32a1e575864958a137c8af99641a3b : SUCCESS in 12m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/343fa679c622419f98baf6347f9c6f07 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c984a677aaba4076804917d78611c7e5 : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dae1da067a344b628b26ba2acfc61733 : SUCCESS in 2h 21m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cf438fadd5445918eadaf091af8b0b2 : SUCCESS in 1h 43m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4865a7d1205a42ce92a633ec0d0275a7 : SUCCESS in 35m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/34580936ee2d49fd9875999263338d35 : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":3},{"id":"09015a099f6b75a149cb55f12ef7901fc6cbbdd6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:07:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9679305d685bc006b0af017018630833f882f3e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:09:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d84fe4c9f34c80a33b62b3548b5c9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af02d4b1508405aa307635ba5767501 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cff2cfa6237ca9680548452f4ee1b18ac2c15c69","revisions":{"fa2cbc98f662f28e84072cfdb23f0ad84b6bead2":{"kind":"REWORK","_number":1,"created":"2022-02-01 17:30:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/30/827330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/827330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/827330/1"}}},"commit":{"parents":[{"commit":"00a3546e13b7433ead53a89aba502385dcf9731f","subject":"Add support for Centos 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00a3546e13b7433ead53a89aba502385dcf9731f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-01 17:29:09.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-01 17:29:19.000000000","tz":60},"subject":"Add centos-9-stream jobs","message":"Add centos-9-stream jobs\n\nAdd octavia-v2-dsvm-scenario-centos-9-stream to the check pipeline.\nAdd a periodic job that builds and publishes c9s amphora images.\n\nChange-Id: Ief23b3fb6f4cedd2866e34c6de184e643d6789ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa2cbc98f662f28e84072cfdb23f0ad84b6bead2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa2cbc98f662f28e84072cfdb23f0ad84b6bead2"}]},"branch":"refs/heads/master"},"6f53443d88b8104d20a0487cc557182952be09d5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-09 18:03:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/30/827330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/827330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/827330/2"}}},"commit":{"parents":[{"commit":"76e731ad7390ea2f934bccf040e66809dbc0a294","subject":"Merge \"Fix devstack on centos-9-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76e731ad7390ea2f934bccf040e66809dbc0a294"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-01 17:29:09.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-09 18:03:41.000000000","tz":0},"subject":"Add centos-9-stream jobs","message":"Add centos-9-stream jobs\n\nAdd octavia-v2-dsvm-scenario-centos-9-stream to the check pipeline.\nAdd a periodic job that builds and publishes c9s amphora images.\n\nChange-Id: Ief23b3fb6f4cedd2866e34c6de184e643d6789ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f53443d88b8104d20a0487cc557182952be09d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f53443d88b8104d20a0487cc557182952be09d5"}]},"branch":"refs/heads/master","description":"Rebase"},"cff2cfa6237ca9680548452f4ee1b18ac2c15c69":{"kind":"REWORK","_number":3,"created":"2022-05-02 06:37:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/30/827330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/30/827330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/30/827330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/30/827330/3"}}},"commit":{"parents":[{"commit":"76e731ad7390ea2f934bccf040e66809dbc0a294","subject":"Merge \"Fix devstack on centos-9-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76e731ad7390ea2f934bccf040e66809dbc0a294"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-01 17:29:09.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-02 06:36:02.000000000","tz":120},"subject":"Switch centos-8-stream jobs to centos-9-stream jobs","message":"Switch centos-8-stream jobs to centos-9-stream jobs\n\nAdd octavia-v2-dsvm-scenario-centos-9-stream to the check pipeline.\nAdd a periodic job that builds and publishes c9s amphora images.\n\ncentos-8-stream jobs are removed because its default python version is\ntoo old for installing projects from master branches.\n\nChange-Id: Ief23b3fb6f4cedd2866e34c6de184e643d6789ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cff2cfa6237ca9680548452f4ee1b18ac2c15c69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cff2cfa6237ca9680548452f4ee1b18ac2c15c69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
