)]}'
{"id":"openstack%2Foctavia~827447","triplet_id":"openstack%2Foctavia~master~Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82","project":"openstack/octavia","branch":"master","topic":"octavia/castellan","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-01-11 15:29:54.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-10 13:27:19.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82","subject":"Rename Context to RequestContext","status":"MERGED","created":"2022-02-02 09:18:44.000000000","updated":"2023-01-11 15:31:08.000000000","submitted":"2023-01-11 15:29:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827447-octavia/castellan","meta_rev_id":"792e918e0be10c6cc0372ddd8f7e630eda27f67e","_number":827447,"virtual_id_number":827447,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 15:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-10 13:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-10-04 08:21:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-10 13:27:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-10 13:27:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 13:25:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-03 09:59:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-09-29 14:05:39.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-12-12 13:32:58.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"19a131f6d0f0f8cf99456f1c035b9245383bbdac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-02 09:18:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb931f360099897126e3fee5bdc719b2c7311d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-02 10:18:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9cd03813e1797d4c1a16c9a63a0efa45cb322c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:25:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70a4fa0cb5784dfdb2cb579830d69ecc : SUCCESS in 12m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f78cbc47a16a486e843698270af364ae : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c48f20217614beca09e120a5f3e5bd6 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0380fe353b5f404eb4d0ae1ffe09d713 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d06acd5bd2874baf88df88b37f83fa7e : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8807bccc05654deb89957ed87950e21e : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5298e4d3c04aeb9fdec12a2d211235 : SUCCESS in 11m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/082df2457ec14444a6a604ae7b226f28 : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c772944f49cf4d21a04a4e33c3150523 : SUCCESS in 10m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6836c6462cb246b2a907e16d43614186 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bcbf3a02c995448ea132c81df09310f3 : SUCCESS in 8m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d554f4e9de954433bbee09fdb7ec6288 : SUCCESS in 2h 11m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eee8a17ac33c4d978720ee47160906c3 : SUCCESS in 2h 17m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/765422a97cf7406caf400b58e031d8bd : SUCCESS in 47m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ebeb61c49e1347fdb470d95045832181 : FAILURE in 1h 47m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/467f7767184a44f7b9c00748b69d998b : SUCCESS in 33m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30c18bb1eeb14f1dafa3430468f9d3ae : SUCCESS in 1h 27m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0b54d14b6a5c4c75a873747860ae27e1 : TIMED_OUT in 3h 04m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71c8ffac709046c29bdf952f86cb5579 : SUCCESS in 1h 36m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce0895e2921044a2bcb815fbe111fb2a : SUCCESS in 1h 32m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/84de8b35727e4e2cafce8c0367f5781d : SUCCESS in 1h 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ebb4ffa05344e19a36300ef5f2a8994 : FAILURE in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a1967ddb73219fdd83fc97f28bb07057affc5e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-02-02 19:23:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e16246818fb08cfa34cf294130647a8b502b7aa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:54:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c45a1c9116e446b8c378c372e1a6367 : SUCCESS in 12m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5559bd50a4d44fe58bb87729432e8800 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e7c07fc39b4fb5993a8560ce466e21 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fcc3d0af9114baf9ec9dc97a51354a5 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0994a3794944e99a2bfd51b8c2be1ef : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d39364cc574d49d59e40a44aa2d3307b : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8bcb24c20f54dfba95052b27a7e14d4 : SUCCESS in 11m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dc9e2328904c4167bb13b21ea27a0479 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4b8465d22a194debb59874a1d16bd667 : SUCCESS in 10m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c789772d18374df08a9ad28d9f2c5f8b : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cfafb8dd0965453a83c260c1988e2a1f : SUCCESS in 9m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18ab007009ff4d57b556887ed4c555fd : SUCCESS in 2h 04m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75243ab23ad4463aa7459f17e84021c9 : SUCCESS in 1h 32m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/431fe45d0b95458fa53e1798c6268ef6 : SUCCESS in 32m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/34899305a824440b98d4da1e316efec5 : SUCCESS in 57m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5321478195bb432ea1be230cb5dcc29c : SUCCESS in 35m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/975cdbb00c47427e90057ea981c0e592 : FAILURE in 1h 58m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82324b809e604e96923242bcff024686 : FAILURE in 28m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6c3501f52bb64a12abdae54b808ba68d : SUCCESS in 2h 21m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0fbbbdfc29434f289ef87d119ce2b36c : SUCCESS in 1h 16m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ff749fd7fb9a4bd2a14cb16969c200aa : SUCCESS in 1h 43m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e8b9d5fdf274d26aee0e571746101d2 : FAILURE in 27m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a389abcab75b3a8860230e121f4c5d8792bb559","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-03 09:59:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e80c1c392ff507876336b4c02621e0e2d477e681","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-29 14:05:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf544cd3851eb10f0e2e7073a9617643622b6a2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-30 07:15:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"812131b46b2e7098a70c4b0927e10279e250531a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-30 09:41:38.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0a2178be3fc364d129884aa1a20310a81a18c839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 09:49:26.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04edea78c6ba4b3cbb1b7266178542b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/767e6f9adf2d47dcbfa7dbb4ded83a01 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b873ef564a34b99bb2b42ea38b3ec3b : FAILURE in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4587413a828344ec8be8005572caf1e9 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/614e7447eb034d9482f99db9a766d95d : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ff2d30713940cc8df971341e5ffb68 : SUCCESS in 10m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5731f08ea6c1436682bb4b613c961a28 : SUCCESS in 5m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ff4b375275ca41e681a764e2f0389e64 : SUCCESS in 9m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8195292115d24ed89bba6965a4a81b1d : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/81ae819654264e95b67966bb39602621 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ad70c639578464ea6cc9da260bb982b : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4702e9cd4eca46ad8016afbcf7a7ed6f : SUCCESS in 6m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f15837b911b4875b65eaf6def6d9a30 : SUCCESS in 1h 19m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a780d1e6b864fb8827a6786957ebc5f : SUCCESS in 1h 40m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45f3222c525a446c833d6debb5aa1c9a : SUCCESS in 49m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/412e155ae1bf4898b5af0924fc9199ac : SUCCESS in 58m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0c80ee77846442bd9d806c495b8723c9 : SUCCESS in 1h 14m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d20d8ebefef4e8b87704e450f1daf0d : SUCCESS in 35m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e95898c2435b430dbff4494c489cae42 : SUCCESS in 1h 43m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/404969e116944e84b1390253893c958b : SUCCESS in 2h 24m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4dc77f06b82349c19044110b443a3d8d : SUCCESS in 2h 05m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f7d0beee1764d5eaed278fdfc1db58d : SUCCESS in 1h 43m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1297c712afc24c65b92adcfb576f2e4b : SUCCESS in 1h 41m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b1f0ca9383d46de8f47a725b7ba600d : SUCCESS in 2h 31m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eefebcdc07cb54c190332cd5a10809f733fae7b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-03 08:58:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dc0129a068198dbdcaccbc024b52899b4b89e2bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 11:34:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4d7c7ac74b4becbb519472b9a61b81\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e20f30cb053493ba3ff693a4b3f2bb8 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/896b0081a39f497db24686e4d1c47dd5 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5949f56700fb46b8a2b443ff4296bfed : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f3aeac1b86840b5b3ac2d3ab536b3ad : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb2ef3ef210542f29a717742e5831ffc : SUCCESS in 10m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f43ac9e260ef498fa9432fba0985fe69 : SUCCESS in 6m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ae7d6ff07abe47dc9d7aff8cbadeefe0 : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/87910ef465114a1da974e15ea9412c09 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5d13e5b7ddff46b9ac4e201174d2636c : SUCCESS in 7m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d3f244ff933492da6fbb2bbc0bf1916 : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/71c83d0aebdb4b7c93bd5d573ffc0f22 : SUCCESS in 7m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c6334a7a9574263879422f0dcd7caf4 : SUCCESS in 1h 40m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70088ab885554504a3150e900f0bac04 : SUCCESS in 1h 54m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25bcfc7d1614402f9cfe4911e4bc8e9f : SUCCESS in 41m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60e653c5c7144c65a5066b2f4cbc66e1 : SUCCESS in 1h 00m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/db7008bcb7d945b1bed27a0dcb0c7ea7 : SUCCESS in 1h 19m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/929b89d32ec14f158baef7c46cc9d2e3 : SUCCESS in 36m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0053246edd9e4bbd9dfd0961268864f0 : SUCCESS in 2h 08m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0ef5a93492940f39af8c88d5c5b0323 : SUCCESS in 2h 14m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5f1141ae994742558b89db76c1a7dfe7 : SUCCESS in 1h 57m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/04ce4e6ec3b644adb7f2bd4754e251d6 : SUCCESS in 2h 06m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6409c7c9dd314d9c853639518cec7baf : SUCCESS in 2h 19m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae701a1fd9e54aab8b0264bbf0785a69 : SUCCESS in 2h 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fcc4f3d9dfdbcbf2588a7defa4486403f13a9e93","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-04 08:21:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"eba6ce040fa8a66d5da7c3ab980036d115b34b81","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-12-12 13:32:58.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"69cc24a3ee02e2cbece67c0eac79ec4ed79609ef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-12 14:41:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be7f2695e90ac4b9801f4841905693fb9d95f046","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-13 12:40:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"065651f608e7175fcdc89b8c7ed68318589c386e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-13 12:48:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed2c0eae1af261e2bc2b77f534e0b0eef107b4ef","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 13:27:19.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"385c25dc84735fea7fe0db0dd3612f58c5471a74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 13:28:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6d7e2e2389474ee40a33db5df6ec230c351f2f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 14:08:47.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0b29b2fec2545fc9572100a9d1a4c48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e9ade6dba444498118a7a6e4dacb44 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/426887376f7c44afa844752654cb9f4c : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/760f48cb837d4e039edc68c9943664c2 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046b5724fad740c1b925e469a8f4d7cf : SUCCESS in 10m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/facc9dad31f641f5b661c1833f5958ad : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dd50efb7b60d47fbb0cfca339c9e49db : FAILURE in 3m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/52e517ee6a3f43beb251e4e924a46166 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e6bd5fa968045f8b98484d3e097f6bd : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/005f6ea0853b4a8cb68071a5b4213fc7 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db0935f433b44f11850c1c24616b49d6 : CANCELED","accounts_in_message":[],"_revision_number":4},{"id":"a85da0727ab74265912add3e12f6c7b91dc5490c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-11 10:26:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bd351932aa072935d9432ce2c248ae68cbb75904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 13:04:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b01fce134224ea99eb92b3592d62d17\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c928b9ed7f54f2bbd76a5ddf66cf95e : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a34d255758548089c52463945d3c58e : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5834b1b30964aad8a6c911a5a14dd0f : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cf8ebd6dd9b4c25a63e467545f18f35 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64526a3d8fa7427c8f273f4698b0ee52 : SUCCESS in 11m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0143801e6c2143f1a7f653bfa22e8ada : SUCCESS in 4m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cb2b8885741e4c23b6026cbc97c988d9 : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d1ddc80606d341a6bfe52e18d03dbdca : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62f86a3ee8b54c04b9645d6a883eef17 : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39c3d1a7b4f441dfb05838ccbe7c0ecd : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/350cce99a3ac4ee8ae1619952520f96f : SUCCESS in 7m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4742a31e9794c589da1c6a052fff72e : SUCCESS in 1h 54m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a72d12124c494d3a80aa6749fcc98d5a : SUCCESS in 1h 49m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43b4a4d113bf486bb0100057b30a9562 : SUCCESS in 33m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b318cfff8c7d4480a344046ab52d0a79 : SUCCESS in 50m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/054a7155544345089b837986c71ce64a : SUCCESS in 1h 22m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bf4fa54f9c545e1b9ab368237b0bc0e : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/33a385cbb8a44050897b34426aaf5b8e : SUCCESS in 1h 51m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3371f59940c345c0a4671583c0b3e198 : SUCCESS in 2h 21m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e20f7ec223dc4cafb5d15f35e3b19999 : TIMED_OUT in 2h 34m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a352d6bce05a493cb74a76c1b407c699 : SUCCESS in 1h 43m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5db0ee7924d349deb6ecf0c455b794c5 : SUCCESS in 1h 36m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/be41fe29bdf442009628cf0e8eecc305 : SUCCESS in 2h 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9f0adfb883f78b9b1d33acd1bad515a834c849b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 13:04:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f7a8c64e57b58bcfd57866b4ebc44a3bdd441465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 15:29:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e99355db94c49c58e87f2c5da6ad06e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99564c7810245c383ee6e0d642c6665 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/908e59212b7643d99ba49f80d9153148 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1935c225ce7476189bcfb711c4f4587 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8faa6646943f4f50a07a1408a61cbbdd : SUCCESS in 10m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a05bc03c3b5441f18a31b52c3874933e : SUCCESS in 4m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b468699ab14445ab2f2cdba27aa9a00 : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9961fe6b7c4a4db7a42e959294c006dd : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a10d99f4a4124f06bfb287856c64e71d : SUCCESS in 2h 23m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ded4d806942845ad972b95394dd50993 : SUCCESS in 39m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0e3233d0a906409d954c59742664fabd : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":4},{"id":"82e81325400d58c1398af99e21f934bf47812dbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 15:29:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"792e918e0be10c6cc0372ddd8f7e630eda27f67e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 15:31:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74ebd1d541ac461db93f88df47dd0a29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9810c6712b1411eb2b2ae4a4abd4f48 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"62e1d871cbbbeefc4919c0c12a3cfff606c53557","revisions":{"9009fb13655ec0df13f2a8d0f540469a4281590e":{"kind":"REWORK","_number":1,"created":"2022-02-02 09:18:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/827447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/827447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/827447/1"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:18:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:18:01.000000000","tz":120},"subject":"Rename Context to RequestContext","message":"Rename Context to RequestContext\n\nIf castellan is used, it requires one of the following context types to\nbe provided during init: KeystonePassword, KeystoneToken or\nRequestContext [1]\n\nAlso it makes sense to be consistent across projects and with\noslo_context regarding context class naming.\n\n[1] https://opendev.org/openstack/castellan/src/commit/86712360f345866e108e12eda1075101635dd1ec/castellan/key_manager/barbican_key_manager.py#L208-L210\n\nChange-Id: Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9009fb13655ec0df13f2a8d0f540469a4281590e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9009fb13655ec0df13f2a8d0f540469a4281590e"}]},"branch":"refs/heads/master"},"9a8e2bf358ae26f7aaadc52d3b05ceafb5c22454":{"kind":"REWORK","_number":2,"created":"2022-02-02 10:18:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/827447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/827447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/827447/2"}}},"commit":{"parents":[{"commit":"c766d891e512d28ab6056126a977d467ce15e317","subject":"Merge \"Updating python testing classifier as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c766d891e512d28ab6056126a977d467ce15e317"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:18:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 10:17:51.000000000","tz":120},"subject":"Rename Context to RequestContext","message":"Rename Context to RequestContext\n\nIf castellan is used, it requires one of the following context types to\nbe provided during init: KeystonePassword, KeystoneToken or\nRequestContext [1]\n\nAlso it makes sense to be consistent across projects and with\noslo_context regarding context class naming.\n\n[1] https://opendev.org/openstack/castellan/src/commit/86712360f345866e108e12eda1075101635dd1ec/castellan/key_manager/barbican_key_manager.py#L208-L210\n\nChange-Id: Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a8e2bf358ae26f7aaadc52d3b05ceafb5c22454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a8e2bf358ae26f7aaadc52d3b05ceafb5c22454"}]},"branch":"refs/heads/master"},"469512591cde2ad221552ff65b256d9e350ee298":{"kind":"REWORK","_number":3,"created":"2022-09-30 07:15:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/827447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/827447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/827447/3"}}},"commit":{"parents":[{"commit":"95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:18:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-30 07:14:58.000000000","tz":120},"subject":"Rename Context to RequestContext","message":"Rename Context to RequestContext\n\nIf castellan is used, it requires one of the following context types to\nbe provided during init: KeystonePassword, KeystoneToken or\nRequestContext [1]\n\nAlso it makes sense to be consistent across projects and with\noslo_context regarding context class naming.\n\n[1] https://opendev.org/openstack/castellan/src/commit/86712360f345866e108e12eda1075101635dd1ec/castellan/key_manager/barbican_key_manager.py#L208-L210\n\nChange-Id: Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/469512591cde2ad221552ff65b256d9e350ee298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/469512591cde2ad221552ff65b256d9e350ee298"}]},"branch":"refs/heads/master"},"62e1d871cbbbeefc4919c0c12a3cfff606c53557":{"kind":"REWORK","_number":4,"created":"2022-10-03 08:58:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/827447/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/827447/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/827447/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/827447/4"}}},"commit":{"parents":[{"commit":"95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c","subject":"Merge \"Add *.orig to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95adbbe618a2fb5ffe0f0e833dcfce5a9d85870c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:18:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-03 08:58:06.000000000","tz":120},"subject":"Rename Context to RequestContext","message":"Rename Context to RequestContext\n\nIf castellan is used, it requires one of the following context types to\nbe provided during init: KeystonePassword, KeystoneToken or\nRequestContext [1]\n\nAlso it makes sense to be consistent across projects and with\noslo_context regarding context class naming.\n\n[1] https://opendev.org/openstack/castellan/src/commit/86712360f345866e108e12eda1075101635dd1ec/castellan/key_manager/barbican_key_manager.py#L208-L210\n\nChange-Id: Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62e1d871cbbbeefc4919c0c12a3cfff606c53557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62e1d871cbbbeefc4919c0c12a3cfff606c53557"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
