)]}'
{"id":"openstack%2Foctavia~827722","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I414c6a3cb978a2960a23cf6dcbacfdb42eb636d0","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-04 13:55:02.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I414c6a3cb978a2960a23cf6dcbacfdb42eb636d0","subject":"DNM Testing stable/train scenario test","status":"ABANDONED","created":"2022-02-03 17:05:47.000000000","updated":"2022-04-12 14:22:45.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cf68d5681628e4d2e9edf010b72275cb2790fb45","_number":827722,"virtual_id_number":827722,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-04 15:55:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 19:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d44d97a815aa18f9074bc67a98772890b4b9fbf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-03 17:05:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f8bf684f980223c7f3178744921edfb60ff1165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:21:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f8285483d8d4718b2fbe41119b6eac4 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c55e11f4f924c7bb2e0339e4d2c4462 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e562d499b8864115b852064e71491f49 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd299832c47649d5805022e812e63076 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/906af253703c444b90d809925dd6b1cd : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b4502d2a9e48bcb373eac89369c03d : SUCCESS in 13m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30dd50d9d8324c2da996d526de8d1c2a : SUCCESS in 8m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/59eb16424b974fdc97ad2cc235bf307b : SUCCESS in 10m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7055b3f2e1284e26915129944c7a15a2 : FAILURE in 1h 21m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6f85b9ed8af91bf0d82a42f4ad04fe300d859999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 08:00:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d8ccb9d8463baa7c7472be92ecb2fbb4af00d754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:45:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7e94c296b4344cbb8521c5fd4fc6d71 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad6a5f40a164b6e8eebba669c3872b4 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b6601a5794d4872b40e06b9bbb2b205 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/655acfae747a4bde9aa5d5e3b9c23bbd : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff57074362cf42ff9ebdafeaf1253763 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6609d2cced4be4879d0077789c07ac : SUCCESS in 11m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/43c8783bed024482b7e001b9c477611d : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/062c25be39304efe92df053eeb15e19a : SUCCESS in 8m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0354c855d82140da8ee82655bf4cda85 : FAILURE in 1h 40m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f072ee2ab3f7ee2e0c54b8f1bbcd0712c7ce8cbd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 09:59:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4da1a4787f9f30c0ceee5a416f5311d820ff24d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:47:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c7ddf91145f478482f74f4b851401b6 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f204276028be437d8052de0475b10ffb : SUCCESS in 8m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/245d5c3e61e3449587c1ad4894428f5a : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6007775577841e8a913b53ee7b81352 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbb0bd25d52e4da2b593861c2a0fe743 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0af61cf17c241d7a1cea04b4bc09551 : SUCCESS in 13m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f38b0272c05c41aabff06c5fb98811ed : SUCCESS in 7m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9c997ad3de744301be4c336345f45aa9 : SUCCESS in 9m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb26e7d03ad94875ad5b743c7e9e7e3f : FAILURE in 1h 36m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1e3dc5bbc3f977e477843735c0c1bde532dedc18","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-04 13:55:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b9c2d21ba7513b27a642b0571743cbfe4b450cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 15:55:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24b40dc430624a5882be1670c7853712 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1cf1956d964d8c8d647104f8901ce4 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/247ee5287501444d935aac59870e3f33 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3b5a7c946a04220970436d252a0acea : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4f16868bcf94f739b35052b6fb2fab1 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ae39c25e8442febfc7ab35adeeef23 : SUCCESS in 11m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/375c51af8a384e8e866bba7458188f3c : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c24b326cfe764bba803aa8cf1ef91ad2 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5987b033881b4533a42eba9891de7d4d : SUCCESS in 1h 48m 45s","accounts_in_message":[],"_revision_number":2},{"id":"cf68d5681628e4d2e9edf010b72275cb2790fb45","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-12 14:22:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b8a62b60d1e00eaf76ad2c41594f2970f7c1e7a","revisions":{"13920965d2aae3f038f27de399f9578c106bd423":{"kind":"REWORK","_number":1,"created":"2022-02-03 17:05:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/827722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/827722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/827722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/827722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/827722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/827722/1"}}},"commit":{"parents":[{"commit":"51c6d8bd9006041c782bb115ec10627e86b87327","subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51c6d8bd9006041c782bb115ec10627e86b87327"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-03 17:05:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-03 17:05:36.000000000","tz":60},"subject":"DNM Testing stable/train scenario test","message":"DNM Testing stable/train scenario test\n\nChange-Id: I414c6a3cb978a2960a23cf6dcbacfdb42eb636d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13920965d2aae3f038f27de399f9578c106bd423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13920965d2aae3f038f27de399f9578c106bd423"}]},"branch":"refs/heads/stable/train"},"0b8a62b60d1e00eaf76ad2c41594f2970f7c1e7a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-04 08:00:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/827722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/827722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/827722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/827722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/827722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/827722/2"}}},"commit":{"parents":[{"commit":"51c6d8bd9006041c782bb115ec10627e86b87327","subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51c6d8bd9006041c782bb115ec10627e86b87327"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-03 17:05:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-04 08:00:30.000000000","tz":60},"subject":"DNM Testing stable/train scenario test","message":"DNM Testing stable/train scenario test\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/827811\n\nChange-Id: I414c6a3cb978a2960a23cf6dcbacfdb42eb636d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b8a62b60d1e00eaf76ad2c41594f2970f7c1e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b8a62b60d1e00eaf76ad2c41594f2970f7c1e7a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
