)]}'
{"id":"openstack%2Foctavia~827898","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I1aab9faa8afe845e445e620d1800785d2e19ad1e","project":"openstack/octavia","branch":"stable/wallaby","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2022-02-08 07:21:10.000000000","reason":"Ann Taraday replied on the change"},"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2022-02-08 09:18:00.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-02-08 09:18:00.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-08 09:18:00.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-02-07 21:54:51.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I1aab9faa8afe845e445e620d1800785d2e19ad1e","subject":"Update scripts to use fips allowed algorithms","status":"MERGED","created":"2022-02-04 20:05:43.000000000","updated":"2022-02-08 09:19:59.000000000","submitted":"2022-02-08 09:18:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827898-fips-compatibility","cherry_pick_of_change":827897,"cherry_pick_of_patch_set":1,"meta_rev_id":"ecebcaf12e28bd1ecb4d820e661170c55857800e","_number":827898,"virtual_id_number":827898,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 09:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-02-08 07:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-02-07 21:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-02-08 07:21:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2022-02-08 07:26:21.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-05 07:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 19:45:47.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2022-02-07 19:45:47.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-02-07 19:45:47.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-02-07 19:45:47.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b2b37ed171d33763ec83b9aa067ac1921bd120a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 20:05:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"b818aaf816e4970a1d735f55cc5910e20464449b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 22:08:02.000000000","message":"Topic fips-compatibility-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"404ae9947ba117240c1588e51455d82a0fbbf728","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 22:08:10.000000000","message":"Topic set to fips-compatibility","accounts_in_message":[],"_revision_number":1},{"id":"7515d526a8b1756cf49624f43cfdd0e9adece09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 07:27:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e107f578cbab4f5dbc08caf5905024fa : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4d63150ba384a28abec2d12e88d376c : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e03920143b48539430527814c6f3f3 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/439eae6819444920aab1b5a407644591 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/457f3e44f9784c30a0f95aa727242111 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e154680a909452ea22f889fdd69e583 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4fee6aafb7044de9106ea0ae7d529f4 : SUCCESS in 11m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dbad43be673142238e9f72feb879cee7 : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d18ad943423542e38c0a5876812f5ef6 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b460164bbb847829a20d8c9b255e0de : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/730143df9e334b298812c1272a0fa515 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19bc36d65c354782b76cc8296c030bf1 : SUCCESS in 1h 30m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/363c3c4dfc8148028a22128e732c4773 : SUCCESS in 1h 24m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/596bc47e7b8a4620b67fd56fd0b084d9 : SUCCESS in 38m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b429a9209d848558a739669724ff685 : SUCCESS in 38m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c46486fcc9964491ba2e69f8aafbd35d : SUCCESS in 36m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea2424ffd0c948de8f5d3c63f4324147 : SUCCESS in 1h 06m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/566a2ef51d2f40c7b7ff4a81ee74a650 : SUCCESS in 1h 15m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d3edd83f2b14faf88d1761fc68f9d40 : FAILURE in 31m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/adbf1ed492524c138430a32b6d02f77d : SUCCESS in 1h 55m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c19bc4cc21154e71a2fbc7b63ff7790a : SUCCESS in 1h 55m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6873b1d896514f509e708a12f1f3304b : SUCCESS in 1h 51m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5df626eef2364a1eaa61bd8e242f4437 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2f87c7788ea4a8b7e7eee8126d67f32656650ac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-07 21:54:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e171ed47b826e10f24a7f8baeaafed633a19e7dc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-02-08 07:21:10.000000000","message":"Patch Set 1: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e3141f6a0bd5b2b48d901f14e68e38fe797fd6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 07:22:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e17ce96c912706b925a6d4ddb8bdbe3e82015f62","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-02-08 07:26:21.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5be962e919b44c718f4b98f0ff83ad3cce947446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 09:18:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05cdc6aafdc04ea599b57e6afc4f0878 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6dc1552b9344a0bbd07600784d527b : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db9a266d9be94e3ba51466410b9e54cc : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc6e16645d8249d88e3fbe136fa4e8e4 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ad594a40404ddbbc37732fbf9450b7 : SUCCESS in 12m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/999bb1a8a5794726bd7abed8151bda6f : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d9f62129dbd4d5480c7f96f6562b67c : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a31a2bbbfefd44df8e5229c6bcf9803e : SUCCESS in 1h 29m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04fe19fae37541198251b2f41b79a3ef : SUCCESS in 1h 29m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/498e7d281b5a4f11a9fefe01ecf48621 : SUCCESS in 47m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/23b75cb6bc28453b802b5c65328837d1 : SUCCESS in 35m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6da0a07014b0413e9c0f4bd440cd2f5a : SUCCESS in 33m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e29a8d718b44b3da03025fa5a72c128 : SUCCESS in 47m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8d0cc25d2935a155fd538f04a9b1bdb7a1f92224","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 09:18:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ecebcaf12e28bd1ecb4d820e661170c55857800e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 09:19:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b19ae00c9b54a62951062d38b2d8571 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a20cf46efac04a97ab64043ae6ba7850160eb2","revisions":{"d7a20cf46efac04a97ab64043ae6ba7850160eb2":{"kind":"REWORK","_number":1,"created":"2022-02-04 20:05:43.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/98/827898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/827898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/827898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/827898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/827898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/827898/1"}}},"commit":{"parents":[{"commit":"a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8","subject":"Merge \"Add generic network interface management in the amphora\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5d94247d15b028c57d8f6f9dbbd2daf0adb34d8"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:51:07.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-02-04 20:05:43.000000000","tz":0},"subject":"Update scripts to use fips allowed algorithms","message":"Update scripts to use fips allowed algorithms\n\nOpenssl genrsa is deprecated in favor of genpkey, and fails in FIPS mode.\nUpdate the relevant calls to use genpkey instead.\n\nChange-Id: I1aab9faa8afe845e445e620d1800785d2e19ad1e\n(cherry picked from commit 36a642d9d0b95b6337558144450bbc7802784c23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a20cf46efac04a97ab64043ae6ba7850160eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7a20cf46efac04a97ab64043ae6ba7850160eb2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
