)]}'
{"id":"openstack%2Foctavia~827899","triplet_id":"openstack%2Foctavia~stable%2Fxena~I2fd562c3b2ff13be11bdaed126c47b1fb9f5ef91","project":"openstack/octavia","branch":"stable/xena","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"7249":{"account":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"last_update":"2022-02-28 11:36:33.000000000","reason":"Change was submitted"},"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2022-02-28 11:36:33.000000000","reason":"Change was submitted"},"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-02-28 11:36:33.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-02-28 11:36:33.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-28 09:33:44.000000000","reason":"Gregory Thiemonge replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-02-25 19:39:47.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I2fd562c3b2ff13be11bdaed126c47b1fb9f5ef91","subject":"Add fips jobs","status":"MERGED","created":"2022-02-04 21:31:44.000000000","updated":"2022-02-28 11:41:47.000000000","submitted":"2022-02-28 11:36:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827899-fips-compatibility","cherry_pick_of_change":798151,"cherry_pick_of_patch_set":24,"meta_rev_id":"ead735ebb40cefb8bf97157a682ef97fa7ffd95c","_number":827899,"virtual_id_number":827899,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-28 11:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-02-25 19:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-02-28 09:33:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-17 10:30:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2022-02-26 06:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-02-28 09:33:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-05 07:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 19:38:39.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2022-02-07 19:38:39.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-02-07 19:38:39.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-07 19:39:18.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-02-17 10:30:33.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-02-26 06:29:53.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"97692d860962dd0b1ff28fd4b4a58dc5a8a08533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 21:31:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"813184716de07776c9bac3b731c7b3aec05917fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 21:32:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3d9938c0e9a739981d2ae2faedc2f764b83cc668","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 22:08:14.000000000","message":"Topic fips-compatibility-stable/xena removed","accounts_in_message":[],"_revision_number":2},{"id":"2370ba948515c2ff50eff8af15001422859a0831","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 22:08:21.000000000","message":"Topic set to fips-compatibility","accounts_in_message":[],"_revision_number":2},{"id":"992f13c29850d970f7fb04cad2819fab99be05f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 07:27:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405d56e405914821a0d67878a7f15637 : SUCCESS in 10m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/651eab8f174e4a13bb1f0a4a00496bbc : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7ad7a04165452eb6ec4c50224f270d : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8014f321dc542bcb6a6b37aa68c2512 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4aff7211b97476ba1e0a5ce854ffdca : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27ef809bfe4945fba806a5735cea2ca8 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c5155c954f4bc485c078cf00797e12 : SUCCESS in 12m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b57dcfd2a5114375b017af7b33c87829 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d18c49c3c43e4b32a5e52345a42bf0a2 : SUCCESS in 9m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/25478ce9bf774cc8abc6c04dcccd476f : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe5f19d37b614b3e87bd82b5533d0f0c : SUCCESS in 9m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4e872f29a0234813961c6edf2a9d859c : SUCCESS in 1h 51m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62d6760d7a3c4bdabea468a59f8884a0 : SUCCESS in 1h 28m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82a9be80a3ca43218f19a237c3e49c5d : SUCCESS in 34m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/588ef516b803473db255d64505c70dee : TIMED_OUT in 3h 04m 26s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/cec4901a136941e0a52f66c3344a9fd1 : SUCCESS in 2h 09m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d083ad3fb2cd42abb9ba7b799bfac88a : SUCCESS in 1h 05m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c7b80bb7b484b1fa1a82f10924fc038 : SUCCESS in 46m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e5e0e469c0ce4ed8b725ac97459014ce : SUCCESS in 1h 49m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/06f625da1338407bac6831c33a41e55a : SUCCESS in 1h 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5eabee90e0624ed6a7006ddbbeffe5c2 : SUCCESS in 2h 22m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19702fc0d8384ffcb15d5d98b571a153 : SUCCESS in 1h 34m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fea797d9306d436a894697d22a01fb60 : SUCCESS in 2h 10m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cdd0b6ab3584456ea5c979db020559ee : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec897aeb24cde3fb03e22be40282b588857aa5d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-07 16:27:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ee022f9fade5584066bf4c897375cfa9226ecb05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-07 16:27:55.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d5686fad281419e2e541934452cf843356e39747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 18:57:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5d7d48af3b54ed2b67c33d460c39198 : SUCCESS in 11m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4202b233e9bb4c86979a0cad9086014c : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4162679a474203b45375b370a95d60 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65e758515c154558b8eeb48696a254c8 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ae3b09cc4704af1a0ff2ca08e3189ea : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f950813a4ca4ac09d9cfe8d4c9736b1 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b465960086354e64a58446b91eed8404 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c4d38431bed340bebf6220fb706b99e4 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2e475399e7ed4bf3b7bff2840f854959 : SUCCESS in 10m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/192c9cadce6f48749d43b28494fa8d2c : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9374ef5cbe8b4e9596932c71927a41ba : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f532643ada04c8ca91a2011d9fc1a88 : SUCCESS in 2h 14m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/716cf3a9d5904ca7a028bfdb66ac9a12 : FAILURE in 1h 51m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a7ad419000b49d1b6eb52237ee9f8f2 : SUCCESS in 35m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b7e242ed3d3947ea94cffa07ee983451 : SUCCESS in 48m 13s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/646afc8e2f1b4f309071ea9fc65a52cc : SUCCESS in 1h 53m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b8bd375ca8a641e8b00f47b7935dedf2 : SUCCESS in 43m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a18da0c7f46d4672aaa5d8891d856af6 : SUCCESS in 33m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c86669d93b6b47889e73c77fdd6179d2 : SUCCESS in 1h 36m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a01f4ec1835416cb90f40880bf69155 : SUCCESS in 2h 26m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3a49db5a9c964a5bb2a0388aebcab151 : SUCCESS in 1h 26m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a6e1ed47aec643599178fb6199422056 : SUCCESS in 1h 54m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6d6c9fc6c8334dd9996054b934426f30 : SUCCESS in 1h 17m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/56fa8b52379f4fdda9f2ca90f6c8ab91 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3642193d1d96d9c71769d235ee522eb6ff4ce2fc","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-07 19:31:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea2b57f1181969d7bc2e2f01293a843afad01baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 22:37:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1d1735acac247bf8f89a22d441b7184 : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faefe5e2b5064f42b1c2eae0ba2ef288 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19808737276a453582d59f0a32393352 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/415a95fbf3ca4ccdb5e6f3602bdfa024 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b0ef73925d46e9b63a78a47e907cb4 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e74508fa396f434f88851b176d757a08 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e6154ac9b34e2fa3b79dfaf9015b6a : SUCCESS in 13m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5e8e84f6d03f4c8b8a54c27b0b049479 : SUCCESS in 8m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/58f306dcd8684cd7b816695c8e001542 : SUCCESS in 9m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/839950b0337b4c9f98e876bf735415b1 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e359a7668bdb4002bd4d4014a809a8b3 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58c533d0f2b6407da6a5213d26b7e241 : SUCCESS in 1h 54m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5afa2f37d2564edabf3696141984273c : SUCCESS in 1h 24m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a784b2416c74513a33dd5c0d517aa94 : SUCCESS in 37m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3387d0b1f0da46e9a400641352eb643e : TIMED_OUT in 3h 03m 57s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/ca197ba4c9ed47bdb20137282bf9ed74 : SUCCESS in 2h 08m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/819efa9cf27f422584f1dd2cd3b17acc : SUCCESS in 55m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/790375004e1e4beb94de6d5987fd62ab : SUCCESS in 33m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/afea810bd2704630a68a4b9384d4a53b : SUCCESS in 1h 18m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4763b0f5853a4f82b84d5276e2af7240 : SUCCESS in 1h 55m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f52b432029384782a5f1fd2bdcd2c0a8 : SUCCESS in 1h 35m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e5757f3acb14c66ad4ebb6a4d207ce2 : SUCCESS in 1h 54m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1a8e717e35a240d7912af9af898b8e94 : SUCCESS in 1h 40m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a60429bc13554d03ac0af3606de30764 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a36b5d417704981924ae7f3921a91bc9c3a4cd3f","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-09 16:56:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bfc2dcbc44ca88fb280fb4b11c9fe0c06ec58a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:49:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c4e60328a264a76bebd9ab366efad07 : SUCCESS in 11m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60eb3699283b4121ae7b8b6265a564fd : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ad6374b0014324aa507e5e0577035f : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3ad7fb47f84e479045bc3bf3365d86 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9f1d1105d39402f9b9c340412fe2b2b : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63be041aeaa7456daf0df778d16c02a3 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5dd84dbfe54e6eb9f7babe95f3a019 : SUCCESS in 12m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06ac5f0ba5f14fafbe84d3e4c8811979 : SUCCESS in 9m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6958ad245ff64d0881aaf5726c42cc12 : SUCCESS in 12m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11ed7509cf9948a29e24dfadca6b9991 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/648ef07c3fb64c3b86f382b914bfc9ce : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f07b2101b16f4c17b45b47687e2a7dc9 : SUCCESS in 2h 26m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1f7ba46a3a345d58354fb8d05633695 : FAILURE in 1h 36m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bbc09520bbc24327832564207fc21171 : SUCCESS in 33m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1264956e63a34566be2798067649c722 : SUCCESS in 49m 55s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/b982d6c79c864dc1bdbdf292379f9978 : SUCCESS in 2h 11m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6c17f10b2ec942198b264cbdc5ce15a8 : SUCCESS in 55m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96a2113a712d419d80fea124111dc838 : SUCCESS in 30m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6b62a48d48b24c02acde50d4e622e624 : SUCCESS in 1h 31m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/459c4fb7c1684fd391a3b348986e37d6 : SUCCESS in 1h 45m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/830e73106df247ecb4d60dd36890a663 : SUCCESS in 1h 30m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/74e2c6e3cbb94be7a5f247f1a3cb60fb : SUCCESS in 1h 25m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b856bf23cb46405284f6de44d205b045 : SUCCESS in 1h 52m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7517f115557f477fa989223b82ce335d : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00158e946deb9b6b4db00809befade4ecf7ee7bd","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-14 20:11:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"75cb1c62784e5758fd2a3d08df16d00bea800ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 01:18:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1e5ede8174f45dfa130d2985ce7d981 : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/416df0df4166441bbac711a0feb50093 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ffe48281bcd4fc19e306be3292ef940 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55680258267e4ffeac039878086d04da : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/682d78b32470486992320e35bad40968 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c2aa2c3c22e4099ab8452ec89c950b3 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580fad28eae044c5985425a0f349d7c7 : SUCCESS in 12m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5d411f891fe84822bf9dd2a84e0174c3 : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/095fa224305f4a11829fbcfadee028ca : SUCCESS in 10m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d8f761b8aacd441393f8739cff80e13b : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86e2710bd7494a14a09e68290488c421 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b838739644604ee7a5af9ab1a31d1eba : SUCCESS in 2h 33m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2cea1a00e7a4400baf923d463c33d25 : SUCCESS in 1h 19m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4952ad687ee2441586a761fd63e165f7 : SUCCESS in 35m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9cd38adb1a6547af81b896170e81fb90 : SUCCESS in 46m 57s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/2e57306e6fc64ba3bd3d749ccc905572 : SUCCESS in 2h 05m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c813cd581a424a7bb593331f376522a9 : SUCCESS in 53m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e73504c013f34d9fb43d72e8e3d735b1 : SUCCESS in 32m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1a2397e27e514b0f995376347476d49a : SUCCESS in 1h 39m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4007e8066c2843a1a53e9eb4e6394cbc : SUCCESS in 1h 33m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0f5bb863ebf54f528d6d3f54ee30e4cb : SUCCESS in 1h 25m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0733d836b7194e6bb1d234f51c9347ef : FAILURE in 1h 26m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0c5452b2e0864a49832833869e2195da : SUCCESS in 1h 35m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/00b3805eca4f4f189ee7142b66f5cb47 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6644fc699841974ebf0b702097e8f9a69deb25a8","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-17 10:30:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9cce62fbd1711c45c286ad9af67b8fc02e24ab4a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-25 19:39:47.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"084a908293517020a79902849601c0343d20c265","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-02-26 06:29:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0628d3394fada0f6aaff16337d399dfe3acbe429","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-28 09:33:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c02438a2f8d72034383e323162d850e5de8ba168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 09:34:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a40cd011715c29a019d15cbdfc8b9c16d0ba6a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 11:36:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55318cdf94c9491f9cdce53cb43f68ae : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/590ac00408ca483788c407bcd53a0821 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe0949ea312342509c2c710c51076f51 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/023f76bd3ed748189914f3de8941e11e : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a492bea88f5442738f62e73b8f678b10 : SUCCESS in 11m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00cabd8365354d2c935363bfed1203b1 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d327b6c27694144827cb29783be99b3 : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d08d3a4187684196b3358323ab990e00 : SUCCESS in 1h 59m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef39f02631c440bbb8192a4e2182a5f7 : SUCCESS in 1h 48m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a161097edf44c6d99e53c253720f10c : SUCCESS in 36m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37059b6d91a54acca1a75619e362fe35 : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":4},{"id":"c6bc23da6b60ce20551b1a65b296d376bff539f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 11:36:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ead735ebb40cefb8bf97157a682ef97fa7ffd95c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 11:41:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2386b12926b643869e03d82e05b7651d : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f78d7b0df51faf0695ec5d67c9b46b50df1b02dd","revisions":{"8a04cda18d265796b1c9bb3c67f2042a2a2d23df":{"kind":"REWORK","_number":1,"created":"2022-02-04 21:31:44.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/99/827899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/827899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/827899/1"}}},"commit":{"parents":[{"commit":"4039d35ce221b58490d4aa5b3497e29dbf7664ac","subject":"Fix LB set in ERROR too early in the revert flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4039d35ce221b58490d4aa5b3497e29dbf7664ac"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:53:30.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-02-04 21:31:44.000000000","tz":0},"subject":"Add fips jobs","message":"Add fips jobs\n\nAdded jobs to test to see if things still work when fips is\nenabled.  Initially set to non-voting until the job proves to be stable.\n\nChange-Id: I2fd562c3b2ff13be11bdaed126c47b1fb9f5ef91\n(cherry picked from commit 14f66646fc08584317fd2279836e7a598cfc27cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a04cda18d265796b1c9bb3c67f2042a2a2d23df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a04cda18d265796b1c9bb3c67f2042a2a2d23df"}]},"branch":"refs/heads/stable/xena"},"a3c6bc654cce09f4a07032d7ad2fe5aeb3aec669":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-04 21:32:52.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/99/827899/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/827899/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/827899/2"}}},"commit":{"parents":[{"commit":"4039d35ce221b58490d4aa5b3497e29dbf7664ac","subject":"Fix LB set in ERROR too early in the revert flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4039d35ce221b58490d4aa5b3497e29dbf7664ac"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:53:30.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-02-04 21:32:52.000000000","tz":0},"subject":"Add fips jobs","message":"Add fips jobs\n\nAdded jobs to test to see if things still work when fips is\nenabled.  Initially set to non-voting until the job proves to be stable.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/827897\nChange-Id: I2fd562c3b2ff13be11bdaed126c47b1fb9f5ef91\n(cherry picked from commit 14f66646fc08584317fd2279836e7a598cfc27cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3c6bc654cce09f4a07032d7ad2fe5aeb3aec669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a3c6bc654cce09f4a07032d7ad2fe5aeb3aec669"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"e60c04b257dd1ceece1d622259ce5c69aceeb184":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-07 16:27:38.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/99/827899/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/827899/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/827899/3"}}},"commit":{"parents":[{"commit":"58eac3d91f47ee18a3cd282c20ad28e3b5a0a111","subject":"Update scripts to use fips allowed algorithms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58eac3d91f47ee18a3cd282c20ad28e3b5a0a111"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:53:30.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-07 16:27:38.000000000","tz":0},"subject":"Add fips jobs","message":"Add fips jobs\n\nAdded jobs to test to see if things still work when fips is\nenabled.  Initially set to non-voting until the job proves to be stable.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/827897\nChange-Id: I2fd562c3b2ff13be11bdaed126c47b1fb9f5ef91\n(cherry picked from commit 14f66646fc08584317fd2279836e7a598cfc27cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e60c04b257dd1ceece1d622259ce5c69aceeb184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e60c04b257dd1ceece1d622259ce5c69aceeb184"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"f78d7b0df51faf0695ec5d67c9b46b50df1b02dd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-02-07 16:27:55.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/99/827899/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/827899/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/827899/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/827899/4"}}},"commit":{"parents":[{"commit":"58eac3d91f47ee18a3cd282c20ad28e3b5a0a111","subject":"Update scripts to use fips allowed algorithms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58eac3d91f47ee18a3cd282c20ad28e3b5a0a111"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:53:30.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-07 16:27:54.000000000","tz":0},"subject":"Add fips jobs","message":"Add fips jobs\n\nAdded jobs to test to see if things still work when fips is\nenabled.  Initially set to non-voting until the job proves to be stable.\n\nChange-Id: I2fd562c3b2ff13be11bdaed126c47b1fb9f5ef91\n(cherry picked from commit 14f66646fc08584317fd2279836e7a598cfc27cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f78d7b0df51faf0695ec5d67c9b46b50df1b02dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f78d7b0df51faf0695ec5d67c9b46b50df1b02dd"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
