)]}'
{"id":"openstack%2Foctavia~828270","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-09 20:30:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","subject":"Preserve haproxy server states during reloads","status":"MERGED","created":"2022-02-08 10:33:55.000000000","updated":"2022-02-09 20:34:20.000000000","submitted":"2022-02-09 20:30:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828270","meta_rev_id":"d67ab4d68502c2bbca0274f0cb5ec9843f4f2579","_number":828270,"virtual_id_number":828270,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 20:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2022-02-09 17:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2022-02-09 17:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"date":"2022-02-09 17:10:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2022-02-09 17:29:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 13:43:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 00:46:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-02-09 17:10:13.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2022-02-09 17:29:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6342044c64b3c9e22112cea91928d93f7f9e7f48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-08 10:33:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df3de302c0becb3f53b8dfe8bf2f277eb36cf147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:43:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca118e2c0066410cb991f66b931329d3 : SUCCESS in 11m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e24b96c988244485a60d3c201cda8c8d : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cae587993ed4e7192a2e7fbb7a7aa23 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcae3647e5a34d0cbece5b6400dfb81e : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/210caa366dcc4460862392b939b4d589 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0d6594a931d453c84893146eaa65d52 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0370117c601240c5be9efbfc8ee0b3e3 : SUCCESS in 15m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4d72d0b59154c9d9f87266854eac388 : SUCCESS in 5m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c0d212d7eff240e6bbf19b228f41445a : SUCCESS in 7m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2b675dc0f0c9486095c53a9d7c479926 : SUCCESS in 9m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/31c2e147234a448795e9d60a2754451b : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/596a063806bb45f18df061bb16f86791 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cfad690bf3ef45b5ab535066c84e0140 : SUCCESS in 2h 33m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aab76ff556cf430a9c01d271af493bdb : SUCCESS in 1h 42m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6eefbd0254fe43adb39dbc1efe8dfffc : SUCCESS in 43m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/848fa0b2b09341b38eca76d1697251e5 : TIMED_OUT in 3h 05m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dfc2cb3720b4434b9210839ff0cb7e26 : SUCCESS in 33m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1bd1afeaaf934efda981ddf6153ab6cf : SUCCESS in 1h 34m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4bb01eb751d4f338032f999eb3a51cd : SUCCESS in 1h 53m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1284401aee7441d1a9c83d05ad842c6a : SUCCESS in 2h 17m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2cec18a7366945bc9a91414976ea5fb8 : SUCCESS in 2h 25m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/42286b2eacff4445b929a1a0020706c5 : SUCCESS in 1h 28m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/840e3a456f984b8f8fe8697449584723 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4610cd5c3cc1d8110f3693f2f081cdec16458ff1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 00:46:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46897410f241babfb6670d4518397b51cedcf28e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 03:14:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20710f1ea4234fa58cfda0b554090dd3 : SUCCESS in 11m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20d56debf6c7431fbb6f42c227703677 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3851cbb921f0417e91d8e0efaa85c996 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/554fdffd360b4769bb722ebabe3279ae : SUCCESS in 8m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0d659e8435141f8ae38f60e1a490e7f : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feb10d8a32ac4719afad264852c8d423 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657d3b5c306349f08e593c2eaf7f352f : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9e275738b744074a9cc86fe4c641d6a : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c602a03fad3d4d3e8c9b2213ecf3ffbf : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/37c96dd0b15b47b9b246e258853b0cd6 : SUCCESS in 9m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/431fddf540b64b8fa230270d65982668 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c72c4829a5634df297c8714a4111f225 : SUCCESS in 9m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b534922fa3c543dda38d4be98e1b773e : SUCCESS in 2h 22m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eaaec03d77fb4d868cbc05c9a3ff7762 : SUCCESS in 1h 28m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/369cdb696b8f4df8ab902575ef2d95ae : SUCCESS in 33m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/27005d87b016494e91fe4259b30803ba : SUCCESS in 48m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e9ffcd8e3cf4c558a7baeece5ff57be : SUCCESS in 30m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f8b14002b7844ce78f61fdb06292c318 : SUCCESS in 1h 33m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88a48c9e647a4c28b625c649274ff787 : SUCCESS in 1h 54m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4885d08e67f0418991619f71a080cf8d : SUCCESS in 1h 43m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e352c22e11714d5bbb6572e26574317a : SUCCESS in 1h 22m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6b051ee1191f43a7b78608ddfb1e9e0e : FAILURE in 1h 45m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/adc9b8dfae1f4ea9a405dafebdf2d8c9 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e13c1d95122e2ab16798f25de4dcf6458a8d818","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-09 17:10:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50249bc0de240e33bed5c240867c8f9a4c1f3275","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-09 17:10:41.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ba54372b3985d4ad61c2d0c42239732934b78325","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:29:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6a0c44aee32b185fe18d3f6887e6c6172935016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c3ba2db99d83c808eae53d3520fe97bcfbb1129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:30:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a526e3eac15a42c384606c6d9ba75f92 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326c738feb74478bb2f94b0f32a9b72a : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee77dedc50234345bd575b9dee1d3358 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be994ca3dc5e4ba78a7caf285ea21804 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6461addf1ce45189cec40730dd5ab08 : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a79325cbd3b4d9fbb62dff4072a9574 : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3e4fa55722354a098836e59ead503c78 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a33d8c9efa441c39a8d9504be68a58b : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e76788a219254533bf0496c4274d5e8a : SUCCESS in 2h 03m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e6496bb97df4a4da532df81db63a520 : SUCCESS in 1h 24m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef859e8490ba41579a2403df94668ec2 : SUCCESS in 34m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ed3b5ee60ff7410e988aedd73342c7db : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":1},{"id":"27637a59b820583a4c6a40774b9af99a40f3ebbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:30:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d67ab4d68502c2bbca0274f0cb5ec9843f4f2579","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:34:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ebb4d75828449b86a2eb8ebfae1675 : SUCCESS in 2m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2550886e333c41eab72ace1c05302192 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24ebc652e84a832ff9baad4b4e27ee6a89e816c3","revisions":{"24ebc652e84a832ff9baad4b4e27ee6a89e816c3":{"kind":"REWORK","_number":1,"created":"2022-02-08 10:33:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/70/828270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/828270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/828270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/828270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/828270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/828270/1"}}},"commit":{"parents":[{"commit":"58eac3d91f47ee18a3cd282c20ad28e3b5a0a111","subject":"Update scripts to use fips allowed algorithms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58eac3d91f47ee18a3cd282c20ad28e3b5a0a111"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 17:53:28.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-08 10:33:33.000000000","tz":60},"subject":"Preserve haproxy server states during reloads","message":"Preserve haproxy server states during reloads\n\nhaproxy doesn\u0027t keep the state of the servers (UP or DOWN) during\nreloads, so a member with an ERROR operating_status may be updated\nto ONLINE after updating a load balancer.\n\nThis commit adds the load-server-state-from-file option in haproxy\nconfiguration files. It also adds an extra step in the haproxy systemd\nservice file to dump the current server-state in a file when reloading\nthe service.\n\nStory: 2009142\nTask: 43084\n\nChange-Id: Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6\n(cherry picked from commit 1b15e1ef22963324e556e072f4bc58021f96bfec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ebc652e84a832ff9baad4b4e27ee6a89e816c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ebc652e84a832ff9baad4b4e27ee6a89e816c3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
