)]}'
{"id":"openstack%2Foctavia~828272","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-09 20:29:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","subject":"Preserve haproxy server states during reloads","status":"MERGED","created":"2022-02-08 10:41:09.000000000","updated":"2022-02-09 20:34:19.000000000","submitted":"2022-02-09 20:29:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828272","meta_rev_id":"1fda9cbf53421a852ed300d38ee9a480950fb5c8","_number":828272,"virtual_id_number":828272,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 20:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2022-02-09 17:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2022-02-09 00:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2022-02-09 17:08:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 13:58:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 00:45:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-09 17:08:01.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"152a12182029639c41d65f91ab9012c68a54ec4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-08 10:41:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c98ae596d1dfa817223bf11d92de3c682130e65b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:58:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04d2d001502a4dea918400c4dd068c91 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455cc7f877f2499bb313305ab4703f96 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/348d17713c2e491e9315fc03b8d98249 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e91e0c6f1f6d4c77a7c537295e23ca07 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/397e9818dca34d15ae4d6b9967aff009 : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0fa83d3fc364d0fa654fa4f5908fee6 : SUCCESS in 5m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb1f5ca1eedf44a4a64950c130896ca1 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e155d64fbd62459d90185328a58f8447 : SUCCESS in 8m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9d0a51630c71457eac651ec2da3efc4c : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8d8d065855a41359461ade4dd8432d4 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4796d6c82b0f4a83ae962ce90e63abc6 : SUCCESS in 1h 25m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60693015169e4246960b0f04e1ed8fd1 : SUCCESS in 2h 11m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f8dda4c6438439f8bd2760d42ba3482 : SUCCESS in 34m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ccc05a9574fb46cda3c5e11f61385f00 : SUCCESS in 31m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/237a27d5cc40474cad9418905106e054 : SUCCESS in 32m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6e30a58cb39f40c7a4387105bed7e862 : SUCCESS in 48m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae5db5dfcc4549d6a5b2c5b6fa015480 : SUCCESS in 1h 25m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/721927711d7a4124bb39fe758f3cc9de : FAILURE in 28m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4dcf13ede3845bfa51ed8332e8bb27e : SUCCESS in 1h 28m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2f11246703a9439486cdee1f31bc1d76 : FAILURE in 1h 20m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c267a277090c46e6b971af6a063734ce : FAILURE in 1h 24m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/21a2d294226440a390c9625bc77936bd : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd33543255b14e876d04ff5fc2ba763fc2c681bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 00:45:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4620d6eb9dc96eccb1f2c899904366606cd26c6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-09 17:08:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"956a9c3788c83c6bf9eedc925cce73c67e797bd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:15:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3875fa4ac3238e9c22d147d3895f571224302bd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:29:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf255cdaafe477aa645f78e104a23f6 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72daaeff43c049d59e023fac07f4eccd : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36ae9a650af348e4969c949152675291 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4568f3f5853e41b6b80f754358c81155 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/202a66e3c1934495a8199056e9f9f0ec : SUCCESS in 6m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ca41be54e5a478bb107a878b3ca7df1 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a79526b85f14198a57486f4f0187859 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/206d4a9094ee43449109ba06fdbe2f07 : SUCCESS in 2h 08m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f87e74409d6495db31d3c0c3ef27382 : SUCCESS in 1h 31m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10660a0ca97d4b29b29a3920078450f4 : SUCCESS in 34m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7e3511c84b4c4922acd00fdf75ca743f : SUCCESS in 32m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d25983cab9e4f61baec4c46cff0b883 : SUCCESS in 33m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/52df7f83cf6f4d3cb1eac08f0e2e8492 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6a2a49d5c24d1910621ef49e27d4e6fe63cd9235","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:29:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1fda9cbf53421a852ed300d38ee9a480950fb5c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:34:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f26c586d35b42369ac0db164204756d : SUCCESS in 2m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5738fc4e2214f90af2904bc6a4c4b20 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20a874e7742346e62e426ddb04e505a59cb3ae5a","revisions":{"20a874e7742346e62e426ddb04e505a59cb3ae5a":{"kind":"REWORK","_number":1,"created":"2022-02-08 10:41:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/72/828272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/828272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/828272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/828272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/828272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/828272/1"}}},"commit":{"parents":[{"commit":"de20868ab74e5d0b3b24e254b8c4cefae7b78fb6","subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de20868ab74e5d0b3b24e254b8c4cefae7b78fb6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 17:53:28.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-08 10:36:23.000000000","tz":60},"subject":"Preserve haproxy server states during reloads","message":"Preserve haproxy server states during reloads\n\nhaproxy doesn\u0027t keep the state of the servers (UP or DOWN) during\nreloads, so a member with an ERROR operating_status may be updated\nto ONLINE after updating a load balancer.\n\nThis commit adds the load-server-state-from-file option in haproxy\nconfiguration files. It also adds an extra step in the haproxy systemd\nservice file to dump the current server-state in a file when reloading\nthe service.\n\nStory: 2009142\nTask: 43084\n\nConflicts:\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nChange-Id: Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6\n(cherry picked from commit 1b15e1ef22963324e556e072f4bc58021f96bfec)\n(cherry picked from commit 24ebc652e84a832ff9baad4b4e27ee6a89e816c3)\n(cherry picked from commit 55bce250ca7e2a25b2767d60373a108ee1b804a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20a874e7742346e62e426ddb04e505a59cb3ae5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20a874e7742346e62e426ddb04e505a59cb3ae5a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
