)]}'
{"id":"openstack%2Foctavia~828274","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-22 10:47:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","subject":"Preserve haproxy server states during reloads","status":"MERGED","created":"2022-02-08 10:46:55.000000000","updated":"2022-02-22 10:49:57.000000000","submitted":"2022-02-22 10:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828274","meta_rev_id":"0fde8d1ca21e2fece01e756d4cbcd47db7c3e211","_number":828274,"virtual_id_number":828274,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 10:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2022-02-22 08:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2022-02-17 19:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-17 09:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2022-02-22 08:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 11:59:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 17:24:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2022-02-17 09:46:10.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-02-17 19:47:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-22 08:32:41.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"ccd46e5349067f1579b37ff527fb5279b63a2716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-08 10:46:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00c2432359c2d55a4c59494a410ff28d6d986116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:59:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af287e573e0746b8bab29c5d1392b821 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4a3b5b87f84190bf88d6b90118baae : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bafce3a54eb40b5816440429173a8df : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/324abb8d38a14fff8bb337f100721d8a : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5820f85b8874d9bb4fd7ce91e6b4a86 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c578d928f8c74e77a3c98164ef9ec19b : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04e646340acd449ca5f26400c246904d : SUCCESS in 5m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fa1d629b29fb44b4b3dc58778dca03f3 : SUCCESS in 7m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9ae7e35d278a48a2acbe0415e87df582 : SUCCESS in 7m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9b6109a766654f7aa599f7803b791358 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/704b3b8c53ae4d0e96f30905769d2e18 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c99adfaedf249399f2c43986c6b7486 : SUCCESS in 59m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/28e12a7b7b894b83a8909966f7590989 : SUCCESS in 1h 09m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12401ec61d204001bde5096ac27ea6ae : FAILURE in 1h 02m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d85540103bbc41b89c5bbddd85c94114 : FAILURE in 1h 02m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/989df25b3150456896f32623b9fece0f : FAILURE in 14m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29da2d7c3e6049a8862d2b2fca000404 : RETRY_LIMIT in 6m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9714268d260d47d8b9dac5c90e0e8686 : RETRY_LIMIT in 8m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b192cccdca6d452e97826d901f62fb5e : FAILURE in 30m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9920dbf3b38d47798d4340195d1f5322 : FAILURE in 1h 03m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c85532a943446cebe46655a3ae75d54 : FAILURE in 27m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fabea56ca27d4eeaa335c9fd25ba56a5 : FAILURE in 27m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8a8d110c93c54157b794874a3fcbda31 : FAILURE in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f89e829e3aa54ed5e2281138be9f21a5c2b5144","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-09 17:24:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4b4a4cdfcab8d1c4a95b90167cdb724cdcdaadb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:35:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/184c7fdc6d974e11884ac906c31753fd : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f5eb3ee74c41fbb0204750dad765bc : SUCCESS in 8m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9928f725e93a4866846f92f2ba9af62f : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cb0795908314903a534d7fc1db43e22 : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59037e1952974492ae42895bf967c436 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329261dc23e54817a4d11c804d4294ad : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96204249886a4a399a7116ca714c0990 : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/08ec8360c9494d768ae6dd7592b101bf : SUCCESS in 7m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7112fd9c6fac42898a6b3ca0c3ef66d6 : SUCCESS in 9m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e969bf5b791b48b190aa4db686f7926e : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4fc9baeca2149d1a0a3738e1e78edfd : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f1b209c95104c0980008283daf17d27 : SUCCESS in 1h 05m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a560ac75f06d430f866984531ce964cb : SUCCESS in 1h 10m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e80784563f7640de9a5f3973b2bf3628 : SUCCESS in 1h 23m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6f2c5461ed024706a6c90e273b3df076 : SUCCESS in 1h 36m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/257de8ffdb3547e88bc86b9f99390796 : FAILURE in 14m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/85c3db66884e480699231db3307563e9 : RETRY_LIMIT in 9m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/75b7832ef73343cda139abbd911700ce : RETRY_LIMIT in 11m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/debc0cf715344bc5b94c2d1103eb8993 : SUCCESS in 37m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f6d6296c6220464eb757591536d00221 : SUCCESS in 1h 39m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8e0b4a24ae44db18b2e0e5e74465f00 : FAILURE in 32m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bc8fabcd86934a148c934800d158e096 : FAILURE in 33m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cdd98a2bfce54e73a5ec75c61dcf65ea : FAILURE in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"292e2ea264b2c1c8bd47df81866c602ed27a15b0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-17 09:46:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ee25b1942b1bc4d07216910b7f6c47744a94c1c9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-17 19:47:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8637245fe934ec35c93efa387751e87c10a5f509","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-22 08:32:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3894abe55b7f61aea4c962dcc44ed8b7a50053bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 08:33:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4ed41e1cd087285270e3737380decc4b0be035e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:47:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c835d4d1eac4d7c8194aa7b82713369 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e0338b0c4724e05b6a1354143457c14 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fad54cf455245b2aa6915db52e22314 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b69c673ef8724548b0eb9da15ee1bcf7 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f2ba4e852a40aea0bd757048229bc0 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b12569dc44654d1298e92df30c3f486a : SUCCESS in 6m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e3e961ad7ca440ec88adc1046208a44f : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e0433d538d7462a903e18cf096ed29e : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21dc45db0f394031ba5f012b2cc9c0c3 : SUCCESS in 1h 27m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1411fec404d54f3e9d95e6e2936118cc : SUCCESS in 1h 40m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8640eb45987f45d88cbc73ff6a0825e2 : SUCCESS in 2h 12m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fd5dda564e2a6d29959b6a3623093184dda369ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:47:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0fde8d1ca21e2fece01e756d4cbcd47db7c3e211","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:49:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a71329bdb44c52a2262dcf0e71893d : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/392b5023322849be869d598e800ae435 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bc14778098a632c4f72d44231239720c0ab28fe","revisions":{"1bc14778098a632c4f72d44231239720c0ab28fe":{"kind":"REWORK","_number":1,"created":"2022-02-08 10:46:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/74/828274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/828274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/828274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/828274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/828274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/828274/1"}}},"commit":{"parents":[{"commit":"51c6d8bd9006041c782bb115ec10627e86b87327","subject":"Fix LB set in ERROR too early in MapLoadbalancerToAmphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51c6d8bd9006041c782bb115ec10627e86b87327"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 17:53:28.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-08 10:46:32.000000000","tz":60},"subject":"Preserve haproxy server states during reloads","message":"Preserve haproxy server states during reloads\n\nhaproxy doesn\u0027t keep the state of the servers (UP or DOWN) during\nreloads, so a member with an ERROR operating_status may be updated\nto ONLINE after updating a load balancer.\n\nThis commit adds the load-server-state-from-file option in haproxy\nconfiguration files. It also adds an extra step in the haproxy systemd\nservice file to dump the current server-state in a file when reloading\nthe service.\n\nStory: 2009142\nTask: 43084\n\nConflicts:\n\toctavia/tests/unit/common/jinja/haproxy/combined_listeners/test_jinja_cfg.py\n\nChange-Id: Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6\n(cherry picked from commit 1b15e1ef22963324e556e072f4bc58021f96bfec)\n(cherry picked from commit 24ebc652e84a832ff9baad4b4e27ee6a89e816c3)\n(cherry picked from commit 55bce250ca7e2a25b2767d60373a108ee1b804a1)\n(cherry picked from commit 20a874e7742346e62e426ddb04e505a59cb3ae5a)\n(cherry picked from commit 1d8ac7e2c13a1ba5370eb8708f5eb83f72d15a5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bc14778098a632c4f72d44231239720c0ab28fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bc14778098a632c4f72d44231239720c0ab28fe"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
