)]}'
{"id":"openstack%2Foctavia~829031","triplet_id":"openstack%2Foctavia~stable%2Fxena~I3f9d08fd4ce038208b8c2d160e6abecae8c8383d","project":"openstack/octavia","branch":"stable/xena","topic":"cs9","attention_set":{},"removed_from_attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-02-21 07:49:07.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-02-21 07:49:07.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-02-16 15:27:54.000000000","reason":"Michael Johnson replied on the change"}},"hashtags":[],"change_id":"I3f9d08fd4ce038208b8c2d160e6abecae8c8383d","subject":"Fix amphora build in CentOS Stream 9","status":"MERGED","created":"2022-02-14 11:53:36.000000000","updated":"2022-02-21 07:50:40.000000000","submitted":"2022-02-21 07:49:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829031-cs9","cherry_pick_of_change":816370,"cherry_pick_of_patch_set":12,"meta_rev_id":"71679c68811d77ce15395666a028e2a7f3955bc4","_number":829031,"virtual_id_number":829031,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 07:49:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2022-02-17 19:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-02-21 05:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-02-21 05:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-14 11:53:36.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-02-14 14:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 07:04:56.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-02-15 07:07:30.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-02-15 07:07:41.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2022-02-15 07:07:41.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-16 12:26:29.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-02-17 14:08:33.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-02-21 05:47:00.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"320dd3175b74b55b12d792195b5a384ccc7e6e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-14 11:53:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"45c906e2459383a13fda7df7199e7506e97ca456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92321f499412485488f330f806097ec4 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7839a01066db4365b7c5b3255eb4ad04 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87a6e7e9d3c4254ab7e46559bad9b8a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5849a20a2a2e4919bd5e1a917f693b1d : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fbd74e0ced243a4baddec75644d4f08 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42565a83688d459586e900797f6d5ad9 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def19f2ef8db4a02aa5ae4e29e065b72 : SUCCESS in 10m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea5d20eb5b974ba6b65f6e793e1ff16f : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/41c0e75cdbce41b8a9c0ce497e63b293 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c38473de6094dd2a124afb0572e8ad4 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71728a41648d42f29702c3650f4f87db : SUCCESS in 8m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f18881008dd142e493b70f4e4d0261df : SUCCESS in 2h 18m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95da2d670dc14c48b5bc255ba018ebc5 : SUCCESS in 2h 09m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/46fd671858d841938e9e316a2e8b271a : SUCCESS in 30m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/27cb75a2a12f4fb9b610298cfc38b772 : SUCCESS in 46m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/497a7521d602425bb51764585b84cdb4 : SUCCESS in 44m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b16f3afa762e4395a145200344b0e9bf : SUCCESS in 1h 24m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ff56c2a4332429289f95ea108a4fda3 : FAILURE in 2h 22m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/67f607f738d44c389884aea4874f9412 : FAILURE in 1h 44m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/452d56f59e7f4b899df70177177db2c7 : SUCCESS in 1h 25m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d17c9aa8103a4866977cf16b6b9b8907 : SUCCESS in 2h 13m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b2a79d833ccb4eb7b4f95c8c2281fae1 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c641f469b184b00f8d3b7646d6a03e122f97edcb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-02-15 07:04:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6aa9518f82df0e3f55b2831e414da0b7f8b04c85","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-15 07:07:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"668e4bc709372b68ff683321eddce133066d50d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-16 12:26:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4661a6d0727b8adb77c64505e19f614cae5cd968","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-16 15:27:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80b9d06ad93c9b35e8759e892b90312b0bc6b099","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-16 21:18:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bb19f15bfce84411c4b900931a92eafe681fc2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-17 14:08:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45846b54ed65d10463718e946767b0f80234adce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:40:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5030358b51d34e52a315aee3f94916bd : SUCCESS in 11m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e320d3ff2114e74817a056d48e4fef8 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ede4c014724b4a9bc1e899292a095c : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd24f723d27a4ddd855b13895633fc2c : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2745646e78274783b39d022e94d23eac : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dedea6e370d9438ba6ef09ece4f661d5 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e853aacc4524a868c3d103048202184 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30736cc858a04961b5bb1e684331bf99 : SUCCESS in 4m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c21b540bfe184bf6a7c5dda7340284be : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/26d172d6ecad435b940a81801f90f363 : SUCCESS in 11m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b5e6a430eda44723b06b21ae10a1e9ca : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73d79ed5ee2847cba33731db02dd77a3 : SUCCESS in 10m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ce7016aefc1e4738bb851c81876ffacb : SUCCESS in 2h 22m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffa392dc865345c7b917914f66d0ed27 : FAILURE in 2h 05m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/433e2eaece334994814912e7436a099b : SUCCESS in 47m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5cf73fb92474ce3b1b578a300768ab2 : SUCCESS in 1h 16m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/541bc9143a0e4a7894e783a900d0bd3f : SUCCESS in 31m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e49857d381554602806d7025b842ae0f : SUCCESS in 1h 56m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9bb93e1fac214e848fb05962e3fc5e3a : SUCCESS in 1h 42m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7cb55b092a864e6da4184ba463c6379c : SUCCESS in 1h 27m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a43e278511bf44659f9be98b523fad48 : SUCCESS in 1h 40m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/518325f8a45243ac96fb5226e34ba379 : SUCCESS in 1h 19m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/294749e160bd4770bd3027736e4c656e : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24b111cbf299d7dfc064566d508866bfa7cdc03f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-17 17:53:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"deb90ddc1e1adc750ad0406989e8fa8252311f58","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-17 19:56:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8d915cfd215e991d06175c02abe81475f00d15a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-17 19:56:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d172a3486ad30f2e1ac032cb5f7f4572bf1aa9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 20:24:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51e868ac672247dcaf6947c9a28a8181 : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3795733be64842dfab52175f2b5f814c : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480d4e5ad0bb4829be6000cab9497397 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/626a7cd0d2ed4fc7aa82ec652883090e : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/083e8eaf4ca24d55b6dff7ac0f894f81 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/515aa1e1c529469ea348ea1b7d27e7e2 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c24a9b9a1524c518ffa0336a976fc3a : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e714d41a363d43ea9f5d5983e3c34f43 : SUCCESS in 5m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5d63a0c09a94b14a9270cba6e86e1bc : SUCCESS in 6m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/414d2ef3947848f29de70ceb1c5b8f6e : SUCCESS in 11m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5675ae54bbbb42f7bbd0b49e564d0105 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7404fd14508b43e4aee0c9a8a6f922e9 : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eecae6522cc747e4a3297a8e8b321af3 : SUCCESS in 2h 27m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bece55c4f2f140e991f10d1794eda46d : SUCCESS in 1h 51m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ce9fa8f23934de1abcb948ed2dc77e4 : SUCCESS in 38m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad74e30beddc4e5c94ae8633b17cad39 : SUCCESS in 1h 01m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d6752c623e04f9f9a16963ad9e67283 : SUCCESS in 36m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b7d5efffb1db4c448fa5e4c17b009695 : SUCCESS in 1h 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c76660414c84dbaa11613198ca1a485 : SUCCESS in 2h 27m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/835d68ee23b04e0e9113b5c24a80a8fa : FAILURE in 1h 43m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/87d9bbf6efd54b2f863bff10dc811b93 : SUCCESS in 1h 47m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b72c6892a725426b90e387069634a064 : SUCCESS in 1h 31m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/febcca6f1a2f4e3b88d705e6b123da87 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7cc33c77230f1c91c29dc472bde9a0fc533fc12","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 05:47:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"75c63905a1dc7138997121d4c4187d944beaf78e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 05:47:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aad067208979a908a0d67f9c7453df756e5b27ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 07:49:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea2fc3459eea44d8a6fb2155a68051d5 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f37619b13a84fc79fa97afeb6058fee : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a83d998dcfbc48eba47f1510dd4b92b5 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73ff1de1bd5046c391ff2f7ffab88198 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502ffb22c017415e8807bf538abdbcd9 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b009fc4a9dba46519f0ac4a030b8bc14 : SUCCESS in 5m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b538fd96351f4ce98ae40fa774e024b8 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fef5b33abfb446cbe074c88c84659a8 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/929188a9c2724901af7501e7f4ff7e69 : SUCCESS in 2h 01m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1fb397b1d9746f0a840bc75117156b4 : SUCCESS in 1h 34m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/421284dd6b914ee682f9b485af37fd51 : SUCCESS in 34m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/21dfdb4c0fa1427ea36dbef576386cb8 : SUCCESS in 1h 04m 47s","accounts_in_message":[],"_revision_number":2},{"id":"f322f220b13d784146c811aee1beb433df579210","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 07:49:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"71679c68811d77ce15395666a028e2a7f3955bc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 07:50:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681e5cfece2e4096a434469daea11577 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9876e133a324ab5a9d8ad8e3e6859e3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28220096b5076ff18a6e6873d20f747d0a53d00c","revisions":{"15f099f41decfcce0aa1a3e08b18c9b38af31291":{"kind":"REWORK","_number":1,"created":"2022-02-14 11:53:36.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/31/829031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/829031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/829031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/829031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/829031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/829031/1"}}},"commit":{"parents":[{"commit":"24ebc652e84a832ff9baad4b4e27ee6a89e816c3","subject":"Preserve haproxy server states during reloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ebc652e84a832ff9baad4b4e27ee6a89e816c3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2022-02-14 11:53:36.000000000","tz":0},"subject":"Add support for Centos 9 Stream","message":"Add support for Centos 9 Stream\n\nUpdate elements for c9s\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n(cherry picked from commit 0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15f099f41decfcce0aa1a3e08b18c9b38af31291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15f099f41decfcce0aa1a3e08b18c9b38af31291"}]},"branch":"refs/heads/stable/xena"},"28220096b5076ff18a6e6873d20f747d0a53d00c":{"kind":"REWORK","_number":2,"created":"2022-02-17 14:08:33.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/31/829031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/31/829031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/829031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/829031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/31/829031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/31/829031/2"}}},"commit":{"parents":[{"commit":"24ebc652e84a832ff9baad4b4e27ee6a89e816c3","subject":"Preserve haproxy server states during reloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24ebc652e84a832ff9baad4b4e27ee6a89e816c3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2022-02-17 14:02:57.000000000","tz":-180},"subject":"Fix amphora build in CentOS Stream 9","message":"Fix amphora build in CentOS Stream 9\n\nThis patch is a combination of 2 commits:\n\nAdd missing release note for commit 0a9f3a8\n\nThis patch adds missing release note for fix added in commit\n0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0.\n\nChange-Id: Id19c8446ac3521af454d53b05c4ea7177b2e21da\n(cherry picked from commit 4dc9a7ec957e8c1b558b0a14a132c849aca6833e)\n\nAdd support for Centos 9 Stream\n\nUpdate elements for c9s\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n(cherry picked from commit 0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28220096b5076ff18a6e6873d20f747d0a53d00c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28220096b5076ff18a6e6873d20f747d0a53d00c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
