)]}'
{"id":"openstack%2Foctavia~829032","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I3f9d08fd4ce038208b8c2d160e6abecae8c8383d","project":"openstack/octavia","branch":"stable/wallaby","topic":"cs9","attention_set":{},"removed_from_attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2022-02-21 07:49:09.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-02-21 07:49:09.000000000","reason":"Change was submitted"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-02-15 07:06:09.000000000","reason":"Chandan Kumar replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-02-16 15:27:12.000000000","reason":"Michael Johnson replied on the change"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-02-15 07:07:15.000000000","reason":"Sandeep Yadav replied on the change"}},"hashtags":[],"change_id":"I3f9d08fd4ce038208b8c2d160e6abecae8c8383d","subject":"Fix amphora build in CentOS Stream 9","status":"MERGED","created":"2022-02-14 11:54:05.000000000","updated":"2022-02-21 07:50:55.000000000","submitted":"2022-02-21 07:49:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829032-cs9","cherry_pick_of_change":829031,"cherry_pick_of_patch_set":1,"meta_rev_id":"cd3af67b45fa7eb63bb9741f9aaf6902ebbf9fe2","_number":829032,"virtual_id_number":829032,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 07:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2022-02-17 19:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-02-21 05:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-02-21 05:47:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-14 11:54:05.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-02-14 14:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 06:38:24.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-02-15 06:38:24.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-02-15 07:07:08.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2022-02-15 07:07:08.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-16 12:26:36.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-02-17 14:19:55.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-02-21 05:47:08.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"2446009f6a9d6a9856aedae2ae7385febfe457ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-14 11:54:05.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"0e0d21c40745c6a2265b2b607f58a97498c735a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:03:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26aba1f4dbec44c28478fd6c6820765f : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9f8fe968d614d4d8eba2e64ba1c0a30 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f89427c37494450a0e023647dbb13af : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d7df194196b43ff973408bcc4057fd4 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c186f27960684ca6882f62870cfbf6f4 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e14061f26824535894c77edc7c3f338 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadcbdc3e14a48a79dceb566ab452054 : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fe96cd57368a4570ad1ff2927f792bfc : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0044528ca78e483d8af053f35bb9711b : SUCCESS in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/deaeb39a91454425ac7a55cd5ce44a35 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccac9f57b3bc40dc98475f7a4c4d1ee2 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98d1acf70e2849c38bb0837a323c6244 : SUCCESS in 1h 29m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cd25b70448f4f35815ec64ea6233cc4 : SUCCESS in 1h 40m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0dcc0121d794c54a9a6dd6ec110a874 : SUCCESS in 32m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ada9304117d642a1baf07cd460bcbe34 : SUCCESS in 30m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29abb86f3c2f4436be0a2208adef364b : SUCCESS in 43m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37aaed7137554bd5b847b246f75e0835 : SUCCESS in 49m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60d1e15bd5394e14adca14f25b84e4d9 : SUCCESS in 1h 18m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7558f2b7d011426aa345bc7823c7c902 : FAILURE in 27m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6394518793b54c3c992bf4fdd25c4be2 : SUCCESS in 2h 07m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3687f668b5bc4c96802aa48c5285fe46 : SUCCESS in 1h 20m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8c638e3aa871479085a37ce47146cada : FAILURE in 1h 24m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/356866952aa74ffeb128e55436bdf78f : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0cc4cc4e6becab547b24e04fb9013e495bdbfbc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-02-15 07:06:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c87612fc6521311048e33aa27e4e98d7ef3fae8b","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-02-15 07:07:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a7bac8d231e789336fdb016a16a037d17acd8fe","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-16 12:26:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dfde48549150c7637f8e0fa01070941ce96c6b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-16 15:27:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"589ca5e9cd0a605bb665d6e096cfb7db4fe15ec9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-16 15:27:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6af2f26f7f0925a6059a314133b3a0dda05ad24","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-16 21:17:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b77b0a5db233d1e4bc6833fd999dffdc621caea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-17 14:19:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6507dd53b66493a83f5e29664e47ee6a0b629b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:50:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6ec4a457b314e839e6d5d1527cbe22c : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ebe408ea6eb48f99bff879bb0356f44 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894a2ed3fa844d7285938a693ecb08de : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/476a1dd96b174f439330bd052cacde9e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f36de1ec4c19451382552cfc4325474a : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc372ecdb3ea4d70ba9c2857cdc40c25 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e489ff7315c340afbf67e49d85e7092b : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c978e87af4644d809ffc6a4ce7d5e384 : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5715661ac087436793b40eaffda40fed : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2af759522cf54398ba0a6b57b9e32f61 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9cb3235c28534987b5ad5f69d37f2c5f : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a03d33b1a6c94b269d59523ee9b8e3d7 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c02a13fe3d2f467fafa72856c65f8d3a : SUCCESS in 1h 40m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2f2eaad37d443589888178a7aea88fe : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a8b6b19e6ead461c997d80f8ce0b0368 : SUCCESS in 54m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1c14ebf469cd45b995511d4a0e88f2dd : SUCCESS in 31m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/166814967278442d85f0e7353336ecea : SUCCESS in 53m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e417f706fc74465b9321c41a77a2055a : SUCCESS in 1h 11m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/313f907a9d584d5faee99d18974829a2 : SUCCESS in 2h 26m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57862408f8ce489197e046f38946a80b : FAILURE in 30m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6938837f01764140a09a96e84b3fbefe : SUCCESS in 1h 46m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd8bd93c334f4525bcecad5c85aac3d2 : SUCCESS in 1h 29m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/353325a315e348c39160b17b7bda597c : SUCCESS in 2h 15m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e3fce40b75204a0a99331e6185dbaddc : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"146e0b0dd95c0745666d2197ced2071aa0ed7497","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-17 19:57:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3463b9af24bc4384870b8236e10bd81d6f84c93","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 05:47:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0fddf0a718c6171d0873b5ad42c0118fabecc95f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 05:48:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bde78d100a289a2f27255e96a3eda97772e935f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 07:49:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39c86f8beefd42ad86fe9a7bf8510757 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b226216c384c4cb4777c9d61eddca8 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0787e29ef1345b89bafccc7f6bbe9c5 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f090fda7db4bccb081030b61ee8324 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeb5e0c37e91457187af32131fe14646 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4491f04616844f91a4ee82bc84edeb9b : SUCCESS in 5m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6ee3c373376748c6972ddf90f7355f70 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3a009cef27747a0a495c1660684ed11 : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af222523f2124e34a4d8160a94216180 : SUCCESS in 1h 32m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75f742a94fcd41b4a34d2284152190ff : SUCCESS in 1h 17m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/366f420d6f9f445d945ad720004f15e5 : SUCCESS in 33m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f11bfa91ccf842a994925d1993ab9763 : SUCCESS in 31m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48e633fa05504ecaa5817b402eaee3c3 : SUCCESS in 43m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e727cfa67edc496eb07c6e37dba5deb1 : SUCCESS in 45m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6432d72ed3032f194b3045d88fb9e22d3251a6de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 07:49:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd3af67b45fa7eb63bb9741f9aaf6902ebbf9fe2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 07:50:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f411c54ee14c8183e18e1264d4b05a : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c109a6d59287418a9daf7f8f85caf633 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d14cc33b7bc98f1d62f40cdf9233122e4a3dfbdc","revisions":{"856bcb3a98c363c7dcbe52de67be212feb9b725f":{"kind":"REWORK","_number":1,"created":"2022-02-14 11:54:05.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/32/829032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/829032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/829032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/829032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/829032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/829032/1"}}},"commit":{"parents":[{"commit":"c28ca9171d1a98d8d31a2710cf9eabd62660de4f","subject":"Merge \"Fix nft command line with negative priority values\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c28ca9171d1a98d8d31a2710cf9eabd62660de4f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2022-02-14 11:54:05.000000000","tz":0},"subject":"Add support for Centos 9 Stream","message":"Add support for Centos 9 Stream\n\nUpdate elements for c9s\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n(cherry picked from commit 0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0)\n(cherry picked from commit 15f099f41decfcce0aa1a3e08b18c9b38af31291)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/856bcb3a98c363c7dcbe52de67be212feb9b725f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/856bcb3a98c363c7dcbe52de67be212feb9b725f"}]},"branch":"refs/heads/stable/wallaby"},"d14cc33b7bc98f1d62f40cdf9233122e4a3dfbdc":{"kind":"REWORK","_number":2,"created":"2022-02-17 14:19:55.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/32/829032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/829032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/829032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/829032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/829032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/829032/2"}}},"commit":{"parents":[{"commit":"c28ca9171d1a98d8d31a2710cf9eabd62660de4f","subject":"Merge \"Fix nft command line with negative priority values\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c28ca9171d1a98d8d31a2710cf9eabd62660de4f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-02 15:39:38.000000000","tz":60},"committer":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2022-02-17 14:18:09.000000000","tz":-180},"subject":"Fix amphora build in CentOS Stream 9","message":"Fix amphora build in CentOS Stream 9\n\nThis patch is a combination of 2 commits:\n\nAdd missing release note for commit 0a9f3a8\n\nThis patch adds missing release note for fix added in commit\n0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0.\n\nChange-Id: Id19c8446ac3521af454d53b05c4ea7177b2e21da\n(cherry picked from commit 4dc9a7ec957e8c1b558b0a14a132c849aca6833e)\n\nAdd support for Centos 9 Stream\n\nUpdate elements for c9s\n\nChange-Id: I3f9d08fd4ce038208b8c2d160e6abecae8c8383d\n(cherry picked from commit 0a9f3a8e6f728ceef4c1dc992b513dc5c0e3a6e0)\n(cherry picked from commit 28220096b5076ff18a6e6873d20f747d0a53d00c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d14cc33b7bc98f1d62f40cdf9233122e4a3dfbdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d14cc33b7bc98f1d62f40cdf9233122e4a3dfbdc"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
