)]}'
{"id":"openstack%2Foctavia~829805","triplet_id":"openstack%2Foctavia~master~Iaa8ea5255b225ef87461c1f85b80f36210b1bde4","project":"openstack/octavia","branch":"master","topic":"ipv6","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-22 10:06:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa8ea5255b225ef87461c1f85b80f36210b1bde4","subject":"Fix unplugging member ports","status":"MERGED","created":"2022-02-17 20:37:38.000000000","updated":"2022-02-22 10:08:03.000000000","submitted":"2022-02-22 10:06:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"829805-ipv6","meta_rev_id":"a4e748c9d3d61dd05962fff8f3c265442f54008d","_number":829805,"virtual_id_number":829805,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 10:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2022-02-22 07:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-21 18:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2022-02-22 09:01:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2022-02-22 07:25:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 23:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 18:40:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-02-22 07:25:55.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2022-02-22 09:01:09.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"812f8b0b68bc3b149f51098c036c672f5860bb58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-17 20:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e1df5cecbd8859732898930c25a1f9c6541b58e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:28:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa57be9edc6844189f27723d411a9a3a : FAILURE in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acd66d073d8b44a69ca00b50e1039789 : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb4619b03a9148baa54a96090311a789 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3639dd219dd47e4910543bf853d0434 : FAILURE in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/221648a09def46298cfc884cd4577b8b : FAILURE in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7acbd1b00d64df58433f9b2a0e3048e : FAILURE in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7823ee15064c440ab8d34f9119895b11 : SUCCESS in 16m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a2d1a5d25b2243d7af79cb09085b52d8 : FAILURE in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c2464016bb33454c9e4ba3217b631d88 : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b7b417e6b90a4f8390518fa17fa13881 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3289c00efd14518933ea404d63efecb : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c3237a1dc70405d9b25a36f9dea7b10 : SUCCESS in 2h 23m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d95a16ee02c4a0da8916be20384b439 : SUCCESS in 1h 54m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c7d355ef92dc48be948dcfb363df6f72 : SUCCESS in 38m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d167ee2e5acf44a9ac48d6a1f6c13d27 : SUCCESS in 1h 08m 56s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/eb651f0ba97242ef90011db44658968a : SUCCESS in 1h 59m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a118f36e4089451e8755e63ce7c7a308 : SUCCESS in 55m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c8e2172ae194c8084cd65071e28df59 : SUCCESS in 31m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0d4069745c8c40439b9da16e00cbaee2 : FAILURE in 2h 04m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ac7abf036104be68cc9eaccce58acfc : SUCCESS in 2h 44m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f7109ee4abfa411287fe0c2d43135824 : SUCCESS in 1h 50m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f304353952e54f4e845833b23bf6bc58 : SUCCESS in 1h 52m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/721680bae50f4d8e8700004f76df1465 : SUCCESS in 1h 53m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a31a7c9d797c49f19aa685da45c75684 : SUCCESS in 1h 45m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9e6e5e2718159675510c624df16d653a9113ada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-18 08:54:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"554874dc0282a816391174ce24b4979a4fd9a191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 11:30:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48de4b47cbd746999421ac840edd5423 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ddf75f064fd454cb69b1b797cd2839e : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f7f9624dc944a0c936df23dc9719d56 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1e294f5573941a0862e9a15c623e356 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47fa05ad61e4422a90b53cc3cc1481fd : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eccbd74db324b4fb3c5107f8712ae62 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2d29ae266449e4abd1377f2ab91a13 : SUCCESS in 13m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1a756bbed7864622b12849c85ec99f4e : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b03f24ce33d94aa9ad32caeb8d6126ce : SUCCESS in 10m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3032e26013bb452ab90ed1f8a1470cb4 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28130db87a2a48e081e21b486feeec4d : SUCCESS in 12m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ac1357245534a869d221f553d8492ef : SUCCESS in 2h 12m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e81eacacda144d59ab65e705d916faf6 : SUCCESS in 1h 50m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81275dced20b4c9080128c0ed347bcc6 : SUCCESS in 38m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dede6cc39a384daf89f940970a3a82c9 : SUCCESS in 57m 21s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/0a70077299014f4abf10e1a0815c0c3a : SUCCESS in 2h 04m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/094b8d1c7e6e44e29dfeb777246ebf72 : SUCCESS in 52m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed2b22d80ddd47bfbb46b71867cb1608 : SUCCESS in 37m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a26b6ce9e9c46228d44ce65f58055ab : SUCCESS in 2h 25m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a19ae22391c440949be77060c04c5ac3 : FAILURE in 2h 09m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0b7325aa0a73423aa1f24e824593dc7f : SUCCESS in 1h 54m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4e0eca7fc4c0458ebbfb0f988ad3ebd3 : SUCCESS in 1h 20m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/993a9d45bf4c416ca7ff564611f9d6a9 : SUCCESS in 1h 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/15332366d7244c0d9afb890ada65cd5c : SUCCESS in 2h 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74243371621caf0536ee897c11c973dd2c4db886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 08:37:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ef3eb6a8aed13a85e7a8a8d7f9d410acfb91aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:02:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c50d2a01ef6426299605f575645821b : SUCCESS in 13m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54447cf8c0ac4f8cb01c8a4e4d1712e0 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eee7318a67f427cbdfc62e5bd9dc7bf : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d13ace93858344dfbeab811e7dc634e5 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46a724c3f764423da6147ce813971264 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5815824977c14979af64d7561acdffe2 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f348bc789a4889a84a567748eabd97 : SUCCESS in 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f068bab5462e4a03b45c1900d3c9d172 : SUCCESS in 4m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/73fedaabb7d54efeb0b71a9b93f89b49 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bd45de70bd0045628404c966d76b80e7 : SUCCESS in 10m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/322d15da226b4f6c8a7b9b1691890a21 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/261a46aec5294ad88a22f9df6d5614bc : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e690cc76c33489793eb7a28f4f0d798 : SUCCESS in 2h 17m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/904804cf6c6145a89d840242d8ffc3ce : SUCCESS in 1h 57m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5c3e28fee144eafbe2bcd2ac392431f : SUCCESS in 47m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/abb26d699b0d470bb19762497da7d8d9 : SUCCESS in 57m 30s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/e3d17c2a8d4b45f4861fd4f69220c261 : SUCCESS in 1h 39m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b069183f588848e089a59fd197088749 : SUCCESS in 47m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e44ef1d846a240a0ad207ac2f5d7ced8 : SUCCESS in 36m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ac38c81801643c5a82397cdd30b0884 : SUCCESS in 1h 56m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fbef510772d247a5861cfadd33d9db34 : SUCCESS in 2h 00m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e6d28c66cac4dfab43b2148a6f5a9d0 : SUCCESS in 1h 56m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/31f06eb4b2c849338ae601e859c9c69c : SUCCESS in 2h 21m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6f165e177ec64cf08078b150036ff788 : SUCCESS in 1h 51m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c584a130372497db598f46609c3d272 : SUCCESS in 2h 05m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"049c7fe2270f8e729b8a5263352ba3ca1eb31c7a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-21 18:40:44.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"93544b42c1449c92f2249a4b8983cbc52e2ad975","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2022-02-22 07:25:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5742e0a9bf94bd4afbda2bfe14604c3c8a12a306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 07:27:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3ddd0d497cd310a3569f3621bd2cffb7498082dc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-22 09:01:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cbea345457c2ba9205b771c11fd7deb056bb694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:06:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf9f2c1d4f9e4247b2675dc62d248620 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f5c4033df84ba5b8984d115ac58b89 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/885cb937150745d68255291cfbe5541b : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a0b8eac318f415184e9c893170829a0 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10654378be1c48e39debce0b773b04bc : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db6f9663afd24bfc9c0b2e4023367b3d : SUCCESS in 4m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9e75ec4a9d5b4c90ae504b27eeef0fa6 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef4e6cc2e9b44fb591b8d4bfe733e3d0 : SUCCESS in 8m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65efe14c86cc465587fdd3a46fea6079 : SUCCESS in 2h 32m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21f120d1ae7c47fdb1310da2549a8caa : SUCCESS in 1h 31m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1683dbcc3f1648c5b5fd8624acdd6945 : SUCCESS in 33m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65c5c5b5ea81426e92c32bdd2e35a5d1 : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6840c3800f8803cf699acb4763082cf1bae09a1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:06:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a4e748c9d3d61dd05962fff8f3c265442f54008d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:08:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d83e186100e4064a4b87c978ee1a2af : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3edb84bcd234a0b862a3f6e68b2631f : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"17a79b7ebf1b6e2faf8266b66a46ea83411359f3","revisions":{"2bc7d5578ca65e5f2067e32f04b80f2a82bd4e54":{"kind":"REWORK","_number":1,"created":"2022-02-17 20:37:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/05/829805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/829805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/829805/1"}}},"commit":{"parents":[{"commit":"2323797a7f8eff16b8d2ce90f41adc428438b869","subject":"Fix ipv6 interface configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2323797a7f8eff16b8d2ce90f41adc428438b869"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-17 16:29:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-17 20:32:36.000000000","tz":60},"subject":"WIP Fix unplugging member ports","message":"WIP Fix unplugging member ports\n\nChange-Id: Iaa8ea5255b225ef87461c1f85b80f36210b1bde4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bc7d5578ca65e5f2067e32f04b80f2a82bd4e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bc7d5578ca65e5f2067e32f04b80f2a82bd4e54"}]},"branch":"refs/heads/master"},"b7a8a12af55c4e082c30519baeb1c8ad56467273":{"kind":"REWORK","_number":2,"created":"2022-02-18 08:54:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/05/829805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/829805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/829805/2"}}},"commit":{"parents":[{"commit":"2323797a7f8eff16b8d2ce90f41adc428438b869","subject":"Fix ipv6 interface configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2323797a7f8eff16b8d2ce90f41adc428438b869"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-17 16:29:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-18 08:54:08.000000000","tz":60},"subject":"WIP Fix unplugging member ports","message":"WIP Fix unplugging member ports\n\nChange-Id: Iaa8ea5255b225ef87461c1f85b80f36210b1bde4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7a8a12af55c4e082c30519baeb1c8ad56467273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b7a8a12af55c4e082c30519baeb1c8ad56467273"}]},"branch":"refs/heads/master"},"17a79b7ebf1b6e2faf8266b66a46ea83411359f3":{"kind":"REWORK","_number":3,"created":"2022-02-21 08:37:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/05/829805/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/829805/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/829805/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/829805/3"}}},"commit":{"parents":[{"commit":"2323797a7f8eff16b8d2ce90f41adc428438b869","subject":"Fix ipv6 interface configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2323797a7f8eff16b8d2ce90f41adc428438b869"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-17 16:29:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-21 08:36:17.000000000","tz":60},"subject":"Fix unplugging member ports","message":"Fix unplugging member ports\n\nUnplugging a port from a VM can take some time before the update is\ncommitted to the DB, 2 (quick) successive member deletions can trigger a\nrace condition when computing the delta of the ports that need to be\nremoved/added: Octavia may get a list of ports connected to an amphora\nthat is not correct.\n\nThis commit ensures that a port is also removed from the DB when\nunplugging a member port.\n\nStory 2009864\nTask 44543\n\nChange-Id: Iaa8ea5255b225ef87461c1f85b80f36210b1bde4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17a79b7ebf1b6e2faf8266b66a46ea83411359f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17a79b7ebf1b6e2faf8266b66a46ea83411359f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
