)]}'
{"id":"openstack%2Foctavia~829887","triplet_id":"openstack%2Foctavia~master~I5b50ebc5e9443f1a0edd987069d482e03d85e726","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-03-31 13:08:44.000000000","reason":"Tobias Urdin replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-04 15:39:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b50ebc5e9443f1a0edd987069d482e03d85e726","subject":"Fix centos 8 stream haproxy repository","status":"MERGED","created":"2022-02-18 13:47:56.000000000","updated":"2022-08-04 15:40:52.000000000","submitted":"2022-08-04 15:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829887","meta_rev_id":"7e8914b957d7ed179543f6760c49c124c2f2d803","_number":829887,"virtual_id_number":829887,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 15:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-02-21 10:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-03-07 11:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-02-21 17:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-04 13:39:35.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-04 13:39:35.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 16:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 10:28:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-02-21 17:45:51.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-03-02 16:11:31.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-03-07 11:12:01.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"ab1224e32f7d8000898245c6ee84ece2bebe8e0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-18 13:47:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88713f24cd89df75f997341382e133ad356c2346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 16:25:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f66aab8e5ae44ee6b3963853f1bca950 : SUCCESS in 11m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49eac0df1d2847f9b8b8cff53144cf8f : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c2ee0f4ab54baaad709c1df68b49d5 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05f20f9b1e4e4e5bb2b17db4937321f1 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62b479f57fb94b44b9d8fc50995f208a : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/805b6865553a431a9391d0488ffc3fe9 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad70461f2a644c30a6a1929e0748190b : SUCCESS in 11m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fa1b253096904c2e8f27c2ef9e6f92a7 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ea3133ac9bb84d6788779be3751b78a9 : SUCCESS in 8m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c15b0b7322b34a289477e9ce125a93e6 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d17d669543d0478086db663026b3c95c : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f9263cf062e5480783cc85719c2b285d : SUCCESS in 2h 07m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cd4647bbe4241238b8a537f8d51073d : SUCCESS in 1h 52m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f28c58a4f6d84d7abfcde21770dbfc5b : SUCCESS in 43m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9fda1dec763492eac422baea64efd30 : SUCCESS in 1h 01m 13s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/a345b283317d452db36228650ea33f07 : FAILURE in 2h 07m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/815c16a3c02c45019d7036ef17148789 : SUCCESS in 1h 00m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff964af6fd7e45c5bed6513e23859378 : SUCCESS in 42m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cd2eaaba486f4a4a90433d9ceb815a69 : SUCCESS in 2h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/844ad9d040eb4dc08d735e76f29ff538 : SUCCESS in 1h 53m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f2f42c8014824fd5b859e189dcac2bc3 : SUCCESS in 1h 37m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f260935d2d847739f2dfdeb1595d8b8 : SUCCESS in 1h 35m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/33e14e82d85b498eb98bb03ee8c63d9c : SUCCESS in 1h 37m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/5070eb4fdfae438db8c282832e53d4e0 : SUCCESS in 2h 20m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44b2ded1814e9120e8f6fb1a4239247780439bf1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-21 10:28:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0e32b10bc52e8282870369810aca1281c4f482f9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-21 10:28:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21c5bbea70ff3c5e8278abda4f834aa67c3037a7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-21 17:45:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45ef0f359e3b4a6c4ee4c67eef487989b39b8259","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-02 16:11:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"654853dd4cefddba0ef8cd668df43bbf98dad0c9","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-03-07 11:12:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7364a3e3636e055f831eac0c6728cf12f28356c6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-29 12:16:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6ee6c2a63d587430164881a35362e987c69e445","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-29 12:22:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"676509d070e9b914ae3b5728383f5763107870fb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-29 12:31:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7af62047f19832dd675d4e9efa5db1c8168c091","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-31 12:49:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19bf7caf65f590c9544ff7be610feb2223584e48","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-31 12:50:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98dc5a51ebfa6f9a7e01a878a03324ecd45ff4bd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-31 13:02:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b06eab87f56c45d608876ccee3a020a6e6d7440","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-31 13:06:22.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"85916b6a5eff79775a6962d5d88e58a65ec9d91b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-31 13:08:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b996801c74c698ca031e99fb2b8788d45e8535f3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-31 13:09:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4effdb57781478f0668825aedffd14657cb5fc2e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-04 13:39:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"affabecf2b59c8595b3a63cda7984a3a4e436704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 13:40:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3794ffbfbbf3e9cd3bd9cfe0f711c0a75d28c88b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 15:39:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a51c06c952b47c490b862ebf9cc50fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7afecdfeb4e46d49bed64bfdcc7767d : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff77fefe140c45b89cbfd32b32a4ba4e : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c5434ff605b470597151592590f7b77 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74362fbdd6484e57ba92b123936f4a38 : SUCCESS in 11m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ed654872380417e8e7970ab902028e5 : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd6b5669d01d4eebb9d974f2d338f4ed : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b5b3912486448eb84ad7ea3cf111649 : SUCCESS in 1h 56m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae75ac09a15b452fb7cddb9db499ae72 : SUCCESS in 1h 55m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5a6233ad7d84ff7b0d6cb52f63d4b3f : SUCCESS in 1h 01m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5ca80c7cb3b44b79359118e880be26b : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"168eba80500c9dc04caace1a8c5ce3156b4a8067","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 15:39:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e8914b957d7ed179543f6760c49c124c2f2d803","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 15:40:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17903ccd29ce472bbf53d4d5322a01ed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75dae9f6dbd149eb8e38b2c2be338102 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc18c1e1b513b02559cd57ce36a6aedc76a606c1","revisions":{"cc18c1e1b513b02559cd57ce36a6aedc76a606c1":{"kind":"REWORK","_number":1,"created":"2022-02-18 13:47:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/829887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/829887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/829887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/829887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/829887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/829887/1"}}},"commit":{"parents":[{"commit":"76e731ad7390ea2f934bccf040e66809dbc0a294","subject":"Merge \"Fix devstack on centos-9-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76e731ad7390ea2f934bccf040e66809dbc0a294"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-18 10:29:22.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-18 13:47:43.000000000","tz":60},"subject":"Fix centos 8 stream haproxy repository","message":"Fix centos 8 stream haproxy repository\n\nCentos 8 stream amphora images were incorrectly using haproxy from\nCentos 8 NFV network-extras repo. This repo was deleted and was replaced\nby a Centos 8 stream repo.\n\nChange-Id: I5b50ebc5e9443f1a0edd987069d482e03d85e726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc18c1e1b513b02559cd57ce36a6aedc76a606c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc18c1e1b513b02559cd57ce36a6aedc76a606c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
