)]}'
{"id":"openstack%2Foctavia~830159","triplet_id":"openstack%2Foctavia~stable%2Fussuri~If7d68c95173b8da66050659d1fb86225880dafdd","project":"openstack/octavia","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-22 08:16:53.000000000","reason":"Tom Weininger replied on the change"}},"hashtags":[],"change_id":"If7d68c95173b8da66050659d1fb86225880dafdd","subject":"Fix detection of member operating status DRAIN","status":"MERGED","created":"2022-02-21 13:11:34.000000000","updated":"2022-04-22 13:28:24.000000000","submitted":"2022-04-22 13:26:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830159","meta_rev_id":"54716c50626eba378a61c3713441864212f9bd80","_number":830159,"virtual_id_number":830159,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 13:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-02-22 13:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-04-14 22:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-04-14 22:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 15:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 13:50:51.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-04-14 22:38:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"947b3cdf68a9dba323f8635e2c64dcbd01ac8df0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-21 13:11:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e89546a8f5fb6a66b7811f343f7f49b4554ac0ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 15:08:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/531bea14c8864bf1a07200a44b588ade : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94b67c96b69c40c8b6fbfe56e390f7bc : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c2dec71efc848168c5020592a31c5a0 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf6b754e5df84a6c834b45698a139128 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749751bf62e041a3815186156a74f946 : SUCCESS in 14m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38ace4cfd19c45c2a737dca7b0246a5f : SUCCESS in 5m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/15cd92dd586a40efb9d1ed2be13a0d8b : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bdb49417dd194e32913ca904e578aa17 : SUCCESS in 9m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c1f0ff6a6ec3487691a1fcf12c38555e : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd2750e84bda41d2b62d70cf766552d2 : SUCCESS in 1h 41m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c5878103ca74ba5b509f62e743d2f15 : SUCCESS in 1h 51m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/15b2630517f1457c9d5ed81a7a4cb25e : SUCCESS in 57m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2e6740c599f842caa7a33f857a79514f : SUCCESS in 38m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1bba89d3eca343bbac2a6b079927f82f : SUCCESS in 47m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e63c68f9f8f041579a883f82175ec19f : SUCCESS in 51m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3d74cc4c17d4feb92870c28636d54ce : SUCCESS in 1h 54m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bab66ae0946b4913bbd5f67ca533e709 : RETRY_LIMIT in 6m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3ba60b7666dd450fb4c07198f11c320a : SUCCESS in 1h 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee6f00919d7affbd28dc6c432b87686395568910","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-22 13:50:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"879e30a92a82c374002a92c303b2aea221deb56e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:38:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e742957a3266e6929a44c53946d2866d8f07868","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 22:39:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"800e90876cb5feb3d8ef75f6a0306618615c97d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 22:51:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8fad175547482c9897504b8739597d : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/205c8a5fbd1d45c0b7bd56d75704f632 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fed79846c40f4f399aaf91897048d4cd : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7186447cabe040d0a3299722dc49779f : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45bb626031bf48aabbdd6ceed67d3eb1 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1baf0c78fdd9434f8efbf17ee97f0d04 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f94198f378264ba497d5fd2dc7bacf41 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1e466ab81ab4ba2904893defd0feb69 : FAILURE in 11m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/adbd543dae234a6f86ca20075379a201 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/17ef10e2684a44e4b418ec97fae13f68 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff7a31fbea0f44dbb421df94f8a37ddb : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/18fc96b998ef49caad798dec6281a603 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"15d84226e1395a19d39ede3c3e3bd04649cf54d3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-22 08:16:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9bb5d07fd043bab2831af1c39d248c9caefab213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7e3fc2f2ef741cb93e9a1490d28ee7b : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88a820667934a518a4f43c02127b43c : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbeb7d6b6f9f45d598e63d8b59867f41 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56b4e9d35a0f422da8d4c2ab7d35e559 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f64d7f8e9a64c99b4b15614a0dd76f9 : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f230a4e0865a436c97d60996b6bbb3de : SUCCESS in 5m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/93ac206553504a83b3c124073a142ea4 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/50e2466b99f8492bb8f9af564062af80 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fe417437ce244f7afaea51f387136bc : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11fce88aa0344b9bb5fa35c1c82118d9 : SUCCESS in 1h 08m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e422cc0e5bd14ca986e8b1cccbc9c121 : SUCCESS in 1h 35m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1bdcf2d34724b8b83282a8b17e9be9e : SUCCESS in 37m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/199abc81a3cb4a5aa4d1efcc96701a55 : SUCCESS in 33m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7858e707bbda455b8489ebb79b97e8a6 : SUCCESS in 35m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/062731b2d24d48828fcd75e01b7f8da9 : SUCCESS in 1h 46m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ffd216f512764f92a99bcc349da479b0 : RETRY_LIMIT in 6m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/10807d3de5dd44e6a501286672df57e8 : SUCCESS in 1h 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8bfaace2dae0dbdbf429c4d1c04fcd49f747594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:06:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"058a69a1fc4d02f228813b28783a3df9aba7b3c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:26:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f379ea281a0c4ef489ae2ce02a635568 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d112ea60b5764fd8a1bede2681f049fe : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6221eb3b578f4a4c932b6e60d65e115b : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ceb3acfbe14f6dacd6fe3a5d096a11 : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a37fcac74cfd4b0288fb6c5a9c60a3a4 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19c92bcf88bd4f2bbd5f39b68b583267 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/150cd9b05b254d0e85b8dcc8d9d30b32 : SUCCESS in 1h 06m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8bad08884d443308211d203012caefd : SUCCESS in 1h 38m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6645b3090c564fb2909c2899c38c23ce : SUCCESS in 34m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/53487a8f3a814bdf8beeefcdf3e8dab5 : SUCCESS in 38m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf9a9acb6be24e9f85ffe23dc2f84825 : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ae785860555c9b4d7ff6b087aae7d0eedcd3164e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:26:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"54716c50626eba378a61c3713441864212f9bd80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:28:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82fd74a8f6d84d26b885f31cf60dd0df : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2b05fa9cae44d779b4c541117ac16f3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebe6e74ac08e52ff51dcf0f5dcf41990f1352997","revisions":{"ebe6e74ac08e52ff51dcf0f5dcf41990f1352997":{"kind":"REWORK","_number":1,"created":"2022-02-21 13:11:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/59/830159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/830159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/830159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/830159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/830159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/830159/1"}}},"commit":{"parents":[{"commit":"1d8ac7e2c13a1ba5370eb8708f5eb83f72d15a5c","subject":"Preserve haproxy server states during reloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d8ac7e2c13a1ba5370eb8708f5eb83f72d15a5c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-28 14:58:01.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-21 13:09:07.000000000","tz":60},"subject":"Fix detection of member operating status DRAIN","message":"Fix detection of member operating status DRAIN\n\nWhen querying the status from HAProxy all values get read as\nstrings. Because the check wrongly assumed integer type values\nit never detected the DRAIN status.\n\nStory: 2009817\nTask: 44389\n\nChange-Id: If7d68c95173b8da66050659d1fb86225880dafdd\n(cherry picked from commit d350f0908f5111cf2791feed9ea83219c8b984fb)\n(cherry picked from commit 9718e036926b2423980656bd3a6c2d3dfaa3f294)\n(cherry picked from commit 0b0131583bcf0863827da0cd81029eb7f285aa8d)\n(cherry picked from commit 7592904b7b5c9b87a46665ac5bb8df2dcc294211)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebe6e74ac08e52ff51dcf0f5dcf41990f1352997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebe6e74ac08e52ff51dcf0f5dcf41990f1352997"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
