)]}'
{"id":"openstack%2Foctavia~830164","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~If7d68c95173b8da66050659d1fb86225880dafdd","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-04-22 12:58:21.000000000","reason":"Tom Weininger replied on the change"}},"hashtags":[],"change_id":"If7d68c95173b8da66050659d1fb86225880dafdd","subject":"Fix detection of member operating status DRAIN","status":"MERGED","created":"2022-02-21 13:13:57.000000000","updated":"2022-04-22 16:44:03.000000000","submitted":"2022-04-22 16:42:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830164","meta_rev_id":"8beecb799e7085a91aef0747a55effa6ee3a404c","_number":830164,"virtual_id_number":830164,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 16:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-02-22 13:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-04-14 22:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-04-14 22:38:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 15:21:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 13:50:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-04-14 22:38:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d7653024ddbbdf487d4d3032f1ba9b809c167eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-21 13:13:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a5eede766a4a2a41df9a66af7794c686d07ffc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 15:21:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/711c4ca01d0547ddbad34dd7ab37268f : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac847c73129840a6b201d360ba3c49c8 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ec08ab34c964cbb87c40790e0dac449 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb219d599ae5423593d43f5d5eb28f6a : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8a84c1a17c40058333b4ff463b81eb : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c278737ca5b455896269d667b6cdb28 : SUCCESS in 7m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/60f740f938bb4cbbb673c32d0e3af3fb : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0446009223b24e16a4139c6584e16611 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/752b8d992bb445eb86f8ebaaa77fca97 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd06a5c06ac44f4d9283abc1b7efc5f7 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9a7fa93126144bc8ef8d187cfc7437c : SUCCESS in 1h 33m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cda6976626bd463daa7f2310797f3d4e : SUCCESS in 1h 41m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31dda547b15f4808a62a2a5494b25255 : SUCCESS in 37m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b6be0439534d456bb6b5b98fe552b045 : SUCCESS in 35m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/381b8ab27ddd4ccf990e283b50e933d9 : SUCCESS in 31m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b4f30b8f4f74cd7aa4eec1bf4130b99 : SUCCESS in 58m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b769853d43d6495eb351b2fb93f1de94 : SUCCESS in 2h 03m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dbba48d0946d478aab9e9efb597c3602 : FAILURE in 30m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7885484e87a94031ac55f8a1f1820e3c : SUCCESS in 1h 17m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b80a1f39f14d4851b7d23b63c28517b8 : FAILURE in 1h 57m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9a13e74a38d94094aeb239850f41175a : FAILURE in 1h 41m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1f8a2e82022e48d4a07a3ffcafd388d0 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"645166415f38518f083c53e908272e41d4cd9e3f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-22 13:50:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59e8e479d39f43205181adf19fbca1487a5d1d5f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:38:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fb96af40df0fc5f4994a1b4c977d41c914fc99f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 22:39:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b7bf442c310d28f5932368d5ed9c55f6c3a031a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 23:03:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86a66dab6364c8abe9520f9acdcf15e : CANCELED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a67948e5da743e98a8d6da9edcbf014 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f0c5113dedf4a36b35191a25a8ccb06 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331989a8e130427d8e0a9a3ea6253c0c : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/210edd5ef9e0470cb3f7d4079f0f5e44 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/60ea5b6f9d4547cdb641601fd3234d03 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ca9339e0fba44089ee4fce86163154f : SUCCESS in 8m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47470ca458ac4732bcccfe1bde45f309 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a131a0e5c42445b4b8696eeca255c8e9 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6446d752e53e437dba41c236fc511b8a : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/70499c1af0fe4bb59c524dc731ccde30 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fa1dbe6579143b5a9dbc473bdd678b2 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e36725ed1b6346e08869a3bfc8aa07a8 : FAILURE in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e9310a3d4f569cf3692320de99980f804e80ebee","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-22 08:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ce33a13d0588acf262a34572efcdfa0a553f1a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 09:59:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47bf7e1c53bc4cf0a59894afaf0b3808 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791fcb81c39d43e8afa12b0e30369bfb : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1b2aa75e64e42dc9b8d8c41fb9ae978 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a355f0f63ec485987599e34da074541 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6723afe385344ed081e46c53b7e05638 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/857bc3e15e424ea2b4caead471fc92bf : SUCCESS in 4m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2529759bfa5d40e68c10c19ddbe0a089 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/08fab69247454d6db2bbeb5dd036322f : SUCCESS in 10m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/be1ff84d74e94ce8bfa3f57ab6e12b6d : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/42b3c5f11d854f69843087fd1621e75e : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84d018745106450eba0e70ad69540668 : SUCCESS in 1h 39m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e37c990f587d4b9a979bbf8711dc4493 : SUCCESS in 1h 32m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e2a03f5e4e0741aeb3d19ff504862614 : SUCCESS in 33m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4b9351b6039d447abf73363ca945d946 : SUCCESS in 33m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e41e48670c4457fa06ce136453e2097 : SUCCESS in 31m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f7ea8dadc3ff47cc9894b786cb9202a5 : SUCCESS in 1h 39m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62b5106f51ff463da4015982f31a9957 : FAILURE in 29m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d588761fb3d4eb5aa2dbaf98e15b4d0 : SUCCESS in 1h 42m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a2f864db98cf4eb3bb87cdc8b1922bdd : FAILURE in 1h 38m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/004ee2e37d2444639563aa6bccb0c431 : FAILURE in 1h 31m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4c3a3db355584396a5e461a49c4c0b45 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5d7b5122076a9775ab224454f41350d30bbed1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:00:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c171315fd78cb789f515524a4c88e8a8287c803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:47:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2da8b0717140318b710c5eb555f1c6 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6618d6b5ef164a58b5363a9b64ece671 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f0b1ded31b345af9087172fb5f3a627 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44be726f5fd54824947e6720d40815ae : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/291cbba97472400c8f1ca67600ae3b23 : SUCCESS in 4m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f1a15a04f204922ab36f7e7a5dac47a : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d11cbdbaa7db476998f3cdf07715dad3 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07ba02c2d8f3461ab8370b9a1d8da196 : SUCCESS in 1h 38m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4df297df84054f6e986db09e6aab83c3 : FAILURE in 1h 45m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/38818dc3f5514b2b92bc74e631ded8a8 : SUCCESS in 36m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c5d2a8c4455047488f9216ff015bb2dd : SUCCESS in 32m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/829f4c71fd4b48a1b678a4bbfda44cda : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"65f8561f9a836cb7069e7bb3d8bfaea1abd40501","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-22 12:58:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90b27e567a64334bd4cee2ca4694999021d04b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 15:09:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf8fedfca602489dafb31819bae9a177 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7b067fbcb8244a88356c77748f9f98f : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6027e28644b6433ca69856b22cca4ab5 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c809b40ba6742b7aba370f91c4b6ccb : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b869902f194401b08acc7a55058c90 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8285ca1a1a674b99ae54fe2801e080fd : SUCCESS in 6m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/49a7870a4a324aa2bca76d35f7bbbf04 : SUCCESS in 6m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ae3e94c6f71f4378b694f15b5ca42c39 : SUCCESS in 8m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/94fb471f1ad34ede865ad76b64a2cbbc : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d68afa9c7c743cda2b6abe673bb912c : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77d0b7b8859c4af2808312b1d939ce05 : SUCCESS in 1h 29m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12182c31a1fb41aeb7152250fb48b7a1 : SUCCESS in 1h 43m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a65a42e8f9184514a69e1f6b56a3c4a8 : SUCCESS in 38m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a8729192684a445e8d387d4af23cac8f : SUCCESS in 33m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0a41c7f9c4448d8a268c22920d0a8b6 : SUCCESS in 34m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bc78b27a952041d9b87e04f29a8c4a00 : SUCCESS in 2h 07m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba851cd75328466e8afa2f2a3ace8da6 : FAILURE in 27m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d6abf0a1d114d3db670afb717f58f1f : SUCCESS in 1h 47m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5898a87ce9334d56a0608417128c212f : FAILURE in 2h 00m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4575fe9a2e7e43eda34267701c0a60a6 : FAILURE in 1h 25m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1f8a4c1400cf48ffb9f8aa6013331bf2 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a3b36eecd8d0d47ff0b9c25528039650361f25b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 15:10:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d01c2e1d6c794342bd8d85f7ab39e5e4797f65e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:42:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726b9ecea62040899427da2bf84fa57e : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae0deae4f7114c94aaaad1a6ab07cb42 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3556214b4b142d4a3ba728f04f8674f : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a1e0562c8d45f6951a681e6f7907d2 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33eb6a73eef04d279bc7fcd5df2591a1 : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/56b44ae9a72543b6b8308ec751b35263 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff4922201441442c9ae810c73fceca3c : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d7928d6b8d445dfb26b7a1743964036 : SUCCESS in 1h 30m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/655b0c1e71d440e0a5cda24c8808db73 : SUCCESS in 1h 27m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39bc367a1e9f45699a27eb87d8b67696 : SUCCESS in 34m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cf70de3b32a74b09afca1e17c7fbe522 : SUCCESS in 32m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3bf73a8f4c9a490d8a21f14e7f474800 : SUCCESS in 34m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0f71c9292f4f0213c3374200f8b84497a27316ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:42:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8beecb799e7085a91aef0747a55effa6ee3a404c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:44:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2758f2a95304e959116a5f56cc42263 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca33ac5619ae4652bfbd022ebdfaaec7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7592904b7b5c9b87a46665ac5bb8df2dcc294211","revisions":{"7592904b7b5c9b87a46665ac5bb8df2dcc294211":{"kind":"REWORK","_number":1,"created":"2022-02-21 13:13:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/64/830164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/830164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/830164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/830164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/830164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/830164/1"}}},"commit":{"parents":[{"commit":"64c792c225a510f8c27a5aa27ea5dae261e0b76c","subject":"Merge \"Fix nft command line with negative priority values\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64c792c225a510f8c27a5aa27ea5dae261e0b76c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-01-28 14:58:01.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-21 13:08:50.000000000","tz":60},"subject":"Fix detection of member operating status DRAIN","message":"Fix detection of member operating status DRAIN\n\nWhen querying the status from HAProxy all values get read as\nstrings. Because the check wrongly assumed integer type values\nit never detected the DRAIN status.\n\nStory: 2009817\nTask: 44389\n\nChange-Id: If7d68c95173b8da66050659d1fb86225880dafdd\n(cherry picked from commit d350f0908f5111cf2791feed9ea83219c8b984fb)\n(cherry picked from commit 9718e036926b2423980656bd3a6c2d3dfaa3f294)\n(cherry picked from commit 0b0131583bcf0863827da0cd81029eb7f285aa8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7592904b7b5c9b87a46665ac5bb8df2dcc294211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7592904b7b5c9b87a46665ac5bb8df2dcc294211"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
