)]}'
{"id":"openstack%2Foctavia~830266","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I293720d61d4109f364201893b7eaa18912aaca88","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-21 16:46:23.000000000","reason":"Gregory Thiemonge replied on the change"}},"hashtags":[],"change_id":"I293720d61d4109f364201893b7eaa18912aaca88","subject":"Optimize DB object to provider dict conversions","status":"MERGED","created":"2022-02-21 17:28:05.000000000","updated":"2022-04-21 21:37:56.000000000","submitted":"2022-04-21 21:36:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830266","meta_rev_id":"71ad9ef0a9557ee3330d25c75ed5194a802fdcd6","_number":830266,"virtual_id_number":830266,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 21:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-15 11:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-04-14 22:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-02 11:35:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-15 11:47:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 19:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 11:35:12.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-14 22:41:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-04-15 11:47:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"a4f06f85ac52ec89d549ddb09048254fc8c4eca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-21 17:28:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c73f1c511ad636ef2c8e2b8d88abee047a3bd2dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 19:24:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8822c8c33cb4d33981f13cacc2baf6f : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/957182cab9f945a487d97aff2c94e77d : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4010015614824056b8ab7d511cda9bbf : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3852891fa9b64dce8c59ea263755bc6b : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f30beac96684b3ab2166b953d73a1ad : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad322473b61845c582fe231d13537fca : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb6935b34ec4fa9846eed176e9ef541 : SUCCESS in 12m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f48924772d5e4247966e1c18c2b9742f : SUCCESS in 6m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ce4e44a54c334ea8bc508087435b2bd9 : SUCCESS in 10m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc494ea17af5450aa4b3d9956b2bf8aa : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26c8f7f8d5d7400d8c28154d41488845 : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/343cdf08fe824c889f40f0f4de763f1c : SUCCESS in 1h 28m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e646aa2eb0c4ce9b51b597e006075b7 : SUCCESS in 1h 31m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3efa4d18d4414ca8b02ffa7a7d92d066 : SUCCESS in 41m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c3213f0b3f664c4e820ef6544eef25f3 : SUCCESS in 32m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/957d84015c9540f19a0f405627c6b057 : SUCCESS in 34m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a51ad6706ec54541ba6b6afb77d60ad1 : SUCCESS in 47m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43a8a87e09044fb1a30bba1b59a03540 : FAILURE in 1h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65b4f6e1402a41f690cca35f5b1c1205 : FAILURE in 29m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/faaf5190cc0d46659cbc1d94a58c3459 : SUCCESS in 1h 36m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/afd3e06f1fda4c79934d4c76438ced0c : SUCCESS in 1h 41m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5cbd8c066fef448598b075d11712fdbd : FAILURE in 1h 30m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a850e9c942544f05a49d62474d3a52c0 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16ed5e312372ee3354df07a3235e3d5b3dd9a493","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-02 11:35:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"210b3c8a02dd0c88a1e59a5c723557740d8d086d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-04-14 22:41:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5db605e8f4848cb896e260d09881e284c270f9da","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-15 11:47:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c61e87f90381dc0c91218aeb77b986c4ec00aba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:48:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02ef376fcea28fcca98484c6ed9a375dca5c60e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 14:06:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ce16d79184843bb9034283d193ad077 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f700b87624045f3a4848ba3fb678b1e : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9d2aec20cb1454c80c5b508c96d5311 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad45d774f780491cb216ef1e76b2a461 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c0ee5a55334b859b844d88d53b445d : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/48cff9ca31734c2d8359c662d94f440b : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddd589a41e6e48d6a39bb5e545b18599 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/feaab88c20f34e95b1753355bafb6420 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09b8780588104cb9a7cf3846a7edb9e0 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1358433c485d42b4890f1edadc4f7dd1 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/773d289ee26d4788b7dc27817a7d6aca : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86535f9d3ea548d3ae1b39d9d86427be : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e96909f66ad4ca184fc3661a018f9fb : FAILURE in 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3e785a0914af0eac77d6ba55d301789357f96007","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-21 16:46:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4295f14a594f788637d0072c1a9825fb131a61d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 18:43:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc27e400b00b4647b948cb19f178685b : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b7c2677c4394370b2bcca69fe15d291 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2db1ffc122e4d4a9adee8c0cde7d373 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/939599e7be34420890098104b2eba7a6 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c0b557f288546598ec70786e391b711 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b396559e5e2047dc87c9d68af34b6b17 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94de8edba5904749a26fbed8d4e508e1 : SUCCESS in 12m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/64b508688c2b473095d4547374a26751 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4b914d18940f431a856ebaec89966dfb : SUCCESS in 8m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c437ccd141146baa8f6eb1d0f3606fc : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3681f384481494da5b5dbd7280aac43 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef63afcc9b0c4c5497bcacf9134d12cd : SUCCESS in 1h 34m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97c770381b23422b9eb295d68d208197 : SUCCESS in 1h 23m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e67d5ae25fa94922b926f1ee624c189b : SUCCESS in 34m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f809978d87f04d938ad1cf28b617409b : SUCCESS in 31m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bba904fd054c48bea762e0ea191972ea : SUCCESS in 32m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/07f67c8cbe2f4a419002b98ee7a7ebed : FAILURE in 42m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ac1a9c0f18614dbaa01aa3124e7cd3c6 : SUCCESS in 1h 37m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1a2d084c92d46efacdcbb713065ea40 : FAILURE in 29m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/27bbee0dbddc47178d7b310346312c5e : SUCCESS in 1h 27m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ecfac68b0f084576b1569d0018c47852 : SUCCESS in 1h 52m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/247bfc98ba0a496c9606ac0df28584de : FAILURE in 1h 31m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/90c9bdca668b46efb0748b6af7dcb806 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdadaa9a2dde044022dab6686ad17e59f1ace6fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 18:44:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32c40ab596104fb819d94c08008b5ffa074e9702","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:36:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d12582cb066e4b03bfcbd7cee6755291 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4264fb19c94f60a091e2a305fc7921 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e2673c9e964d0bb169db6d32e4a9a5 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb917bfaf90742e4bb85e38b65dbd26d : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d93898b2a1460ebddd23d7109bb799 : SUCCESS in 10m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b010377ea35e4486a85fe6e95f63f7af : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d290000e0a9447abb5caa14b1d21c5f : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c16b8b8f91f3479a91ffc21933ab6fab : SUCCESS in 1h 32m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc82b33d81d846038ebae5969edb7661 : SUCCESS in 1h 21m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1eb25464b0942f2a9d370c8d524bbbe : SUCCESS in 38m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2dcb2264ee3f4d96a00761b48554404c : SUCCESS in 38m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1126d4be5b94708b7e22c83a97c54e9 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"44326ea42a7493fe0948401c0ecd4cc1a98c6301","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:36:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"71ad9ef0a9557ee3330d25c75ed5194a802fdcd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:37:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c4d330900541c9b3073e84219eaf3f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82665fed71edd50cb5c05f518e301460e973025b","revisions":{"82665fed71edd50cb5c05f518e301460e973025b":{"kind":"REWORK","_number":1,"created":"2022-02-21 17:28:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/66/830266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/830266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/830266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/830266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/830266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/830266/1"}}},"commit":{"parents":[{"commit":"d14cc33b7bc98f1d62f40cdf9233122e4a3dfbdc","subject":"Fix amphora build in CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d14cc33b7bc98f1d62f40cdf9233122e4a3dfbdc"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-29 11:38:35.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-21 17:26:45.000000000","tz":60},"subject":"Optimize DB object to provider dict conversions","message":"Optimize DB object to provider dict conversions\n\nSome functions in the amphorav2 driver convert DB objects\nto provider dicts many times in a single call: listeners are converted\nwhen converting a load balancer and are also explicitly converted.\nNow listener dicts are fetched from a recursive load balancer dict, it\nuses a single conversion per call.\n\nChange-Id: I293720d61d4109f364201893b7eaa18912aaca88\n(cherry picked from commit 4388ae7fde59622f461b601f1ac3367b06f5c91e)\n(cherry picked from commit 2f9fe00eaf0d1e2e1c71ef8e8052767e79a402bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82665fed71edd50cb5c05f518e301460e973025b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82665fed71edd50cb5c05f518e301460e973025b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
