)]}'
{"id":"openstack%2Foctavia~830417","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Iaa8ea5255b225ef87461c1f85b80f36210b1bde4","project":"openstack/octavia","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-03-28 11:12:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa8ea5255b225ef87461c1f85b80f36210b1bde4","subject":"Fix unplugging member ports","status":"MERGED","created":"2022-02-22 13:28:37.000000000","updated":"2022-03-28 11:15:10.000000000","submitted":"2022-03-28 11:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830417","meta_rev_id":"0d7f95b708ca39d401eb0a7612dfd1429df09515","_number":830417,"virtual_id_number":830417,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-28 11:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-03-28 08:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-03-09 16:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-22 13:46:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-03-28 08:32:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 13:46:02.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-02-22 15:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 16:53:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-03-28 08:32:03.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"58fba9cebc4c58674297e4b45f2b4b7e68a3e548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-22 13:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00fa6751d9d72206488d0abf4f1c8336659c255a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-02-22 13:46:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"69e30c3a7dc52bcdfa3049f051724f505412b746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:42:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75777c8ecb654785b0e44126c968c2aa : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba398f46e18420b9d7a18ff041187c5 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6baf813f81e34111b8e054e45b44e1f1 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18e4ad02a9004846b820bcbd8b6d7c2c : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb57c86fd191454db7c021998c5b66cf : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1371efc6744be09d309099c19ffd1f : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c6b3f34d66d47708ba3be822a630992 : SUCCESS in 5m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/18d0bcdf253d492d9fdef17ba0b8abba : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/844a25e99b7841ed9ae50c8aa3973a2e : SUCCESS in 8m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/460f3512a5fb4397abff006789e73a8a : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91124d43545e456e8b33b17fdaeb1680 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/83d787ae57204898865b8199c7b566b3 : SUCCESS in 1h 16m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e15fa138ec7a409995db7dd6ace044c0 : SUCCESS in 1h 24m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e481a8fe5974aacbab76e497ceb9804 : SUCCESS in 1h 59m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c981f0c300804372a1916b9b8aca4509 : SUCCESS in 1h 51m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/94aaaf25b8aa4956994c7f7a0c1d6ea5 : FAILURE in 15m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c147ca07aee34ebcb10e82b62158a5af : RETRY_LIMIT in 9m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6947c961047046cb843c23fd8e000081 : RETRY_LIMIT in 9m 00s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e2563e71b02e4547853bfdfe52820190 : SUCCESS in 50m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94dc2f2aa35043fc97d1fbd82dea43ac : SUCCESS in 2h 10m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4b63b58a8ab4f5e9457a4f96c7f37aa : FAILURE in 30m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6e6360c6d3dc4a9e9c8f37b0ec4b2056 : FAILURE in 31m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a25dbed2f1a5431786a03772caf7d053 : FAILURE in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30770db87b90efbb50cbe245d1638bf6ec426910","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-09 16:53:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a78886fa1c9b6e2edf2a4d4ed0ab389f78a9e34","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-03-28 08:32:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92f52ac566bb016d42f500164e42d1bbe42dfad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 08:34:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07902a19fb0f4f5dda07d27b9a9f1b37b908cb44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:12:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7871b854a3e43b18d2fe86631d900c8 : SUCCESS in 8m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fd498fd4db1419f9cca62f68a2f5f21 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e71a6c621bb54f688483720fc843bd8e : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e80c16cc068c404da9fa0a2be49c2aa6 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810c28e0346f43deae2aba2f8ba70b57 : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2e3652787654628a36dfc1e9d53bf95 : SUCCESS in 8m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/496d687941d647538f9b590017db8cdb : SUCCESS in 10m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e0ce27a413334feb95d788d9a34908a0 : SUCCESS in 10m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2ece26233244be5ac1bb0f1ce695802 : SUCCESS in 1h 20m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/df31df96d4c44679b13fd339c30615ae : SUCCESS in 1h 12m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7de03b1675b04b15a138515183c82b33 : SUCCESS in 1h 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"eacd5b21c5cf42106cd2ef51402f2f1810fffb19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:12:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0d7f95b708ca39d401eb0a7612dfd1429df09515","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:15:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178b813da0bf4f399538616f48bdbf72 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0de97576b7f24acdb94abf24babf9cbf : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35ca791355c1fa38c988918c910ff685a87e3652","revisions":{"35ca791355c1fa38c988918c910ff685a87e3652":{"kind":"REWORK","_number":1,"created":"2022-02-22 13:28:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/830417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/830417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/830417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/830417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/830417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/830417/1"}}},"commit":{"parents":[{"commit":"1bc14778098a632c4f72d44231239720c0ab28fe","subject":"Preserve haproxy server states during reloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bc14778098a632c4f72d44231239720c0ab28fe"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-17 16:29:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-02-22 13:08:12.000000000","tz":60},"subject":"Fix unplugging member ports","message":"Fix unplugging member ports\n\nUnplugging a port from a VM can take some time before the update is\ncommitted to the DB, 2 (quick) successive member deletions can trigger a\nrace condition when computing the delta of the ports that need to be\nremoved/added: Octavia may get a list of ports connected to an amphora\nthat is not correct.\n\nThis commit ensures that a port is also removed from the DB when\nunplugging a member port.\n\nStory 2009864\nTask 44543\n\nChange-Id: Iaa8ea5255b225ef87461c1f85b80f36210b1bde4\n(cherry picked from commit 17a79b7ebf1b6e2faf8266b66a46ea83411359f3)\n(cherry picked from commit 887ffeaefdc5ee3b3d164cd80533bfa190a3704b)\n(cherry picked from commit a0196a642519357b43b55aad8ffa63320cc5a09a)\n(cherry picked from commit 71065ca9b608f8b32d68f94e8b2fcde669f5df77)\n(cherry picked from commit 2d6ea06f1e060c6bced0873d007408b0f359e10e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35ca791355c1fa38c988918c910ff685a87e3652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/35ca791355c1fa38c988918c910ff685a87e3652"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
