)]}'
{"id":"openstack%2Foctavia~831051","triplet_id":"openstack%2Foctavia~master~I287d89cd83e91473f1375788c969521aa58ca567","project":"openstack/octavia","branch":"master","topic":"review","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-17 06:36:56.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-08-09 10:48:02.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},"34591":{"account":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"last_update":"2022-08-18 11:50:37.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-07-12 07:03:09.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I287d89cd83e91473f1375788c969521aa58ca567","subject":"Add event notifications for load balancers.","status":"MERGED","created":"2022-02-25 21:10:42.000000000","updated":"2022-08-18 11:52:02.000000000","submitted":"2022-08-18 11:50:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831051-review","meta_rev_id":"5819fcf4bb6446fda8e83f9510d4dbf6713c8829","_number":831051,"virtual_id_number":831051,"owner":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 11:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2022-08-04 18:39:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":1,"date":"2022-08-05 12:26:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-18 09:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-08-09 10:48:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2022-08-09 16:41:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-18 09:41:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"date":"2022-08-09 10:48:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":-2,"date":"2022-08-09 10:48:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 23:08:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-08 06:30:11.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2022-03-28 16:38:47.000000000","updated_by":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"reviewer":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"state":"CC"},{"updated":"2022-04-07 14:52:16.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2022-04-07 15:13:12.000000000","updated_by":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"reviewer":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"state":"REMOVED"},{"updated":"2022-04-26 10:52:23.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-27 16:20:17.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-05-11 18:42:39.000000000","updated_by":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"reviewer":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"state":"REVIEWER"},{"updated":"2022-06-22 09:36:18.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-07-12 07:03:09.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-07-28 23:52:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"8342a07dbd69649a81bf5772759ace5c7fb0cf35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-02-25 21:10:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87c25fd46edfea8303bfc9181e1e95dc9b9b45ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-02-25 21:38:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c69f343d7365c509ce40519da484daf987ca103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 23:08:53.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea26facf192b49a19c8b1173aa01cef7 : SUCCESS in 11m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72a8310c70de40289402b6596ba20b22 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5de03dc0ca0468e8f756678a74e71ef : FAILURE in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aa23578e2ba40d38ba8bb569fa476cb : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0f7081695eb4ff38c5fbf778b9f3c7d : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25be20b1b7e943f0b2ecf87e6148194e : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53346e52b97343eab11ce2e6a83fad46 : FAILURE in 5m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3805c76d6b3f4eeda6bd459116fa21db : SUCCESS in 7m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cce5002228449eb86fce804541e0be9 : SUCCESS in 11m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0d32ed6162dc42e7b082567e251a4b9d : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22e287cb503346e4be114e691d41d853 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae9cc1c1c0614cfc902acea10e612199 : FAILURE in 1h 08m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e2f3279ccec409382ea19d2d198106e : FAILURE in 51m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c05c6fed0a854b9eb51063264756cee0 : FAILURE in 49m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ac0b9a81787242ddaffa5f9ad34f3114 : FAILURE in 56m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6dece4ba2a9949caa062bbb1855c9a94 : FAILURE in 1h 05m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fef5d60c88a44d98b9a12a819a57c040 : FAILURE in 17m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c7eeee7c25d844609b6079206bbd2c0e : FAILURE in 1h 05m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eb82bc856f5d4410b9ba16d332f20d10 : FAILURE in 1h 08m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/08cef56f82ff4b10b94de07ea261fb66 : FAILURE in 1h 12m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7fea1a9326d741c08946be32145d63a7 : FAILURE in 1h 25m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/79c403b239984ac8bf9c4638f0659a3d : FAILURE in 1h 07m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b7db5bf0eaf4911b496e6dac72c093d : FAILURE in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7042dcc7ee4ed5e7e06d69a000cf632986a9afdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-02-25 23:28:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"242bc0c92a2f9ca886fa04d75f449004d7862430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-26 05:46:54.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd1e0a4f5dfc4418aede1b20d96e8fa8 : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d88a548b9b7342778d4b19cd9c0fe6b0 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2dc7544ed04561803a8cb6273c71ee : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49a1bae8ce5a414f83876ff457a9f361 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18dd620f86ce450ab81f6f800f17ad3f : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9291d57586f4bf185bb816fea028603 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87949fcd71124cebac97954bf967af6c : FAILURE in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd62a7fbe0e340a68b5e06e296bd3513 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8a6c713f02c4cfabad1932af5f40757 : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e56cbb967d7e4703b1c9c0ece7d32bec : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3cc0db5c846406c8137e7755031d958 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74fc3a87ea98407290091351e84f7627 : FAILURE in 1h 31m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66145020c48a4e04b209382e591a4ad4 : FAILURE in 1h 12m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/844df7f0127d4ad78d8b1bc02b072930 : FAILURE in 43m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f48b53782894301bf70a50701c343cc : FAILURE in 48m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/05ec98368d8d4037acc75e45327edec6 : FAILURE in 45m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/752a3cf2713141a4824019aea5c02900 : FAILURE in 35m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/52528e859b9f469a85f9bf7f8c2459ad : FAILURE in 50m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8a0dfdf0a924133a0b1fa457a4f0e2a : FAILURE in 58m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c5dd66b96e4a4ad1bd57b85090237d2e : TIMED_OUT in 2h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e73e1c99f193451592e20edf3c80a794 : FAILURE in 59m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4a63f374a758478788a07ff9d1e9ff9f : FAILURE in 45m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8d1d2f078ff4902b7fade61a7f8de64 : FAILURE in 59m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fea71bc4f9ed995c533a6f6a2b2eaa062c4df368","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-02-28 22:19:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2db79cef7fe94762758ae9a36aa8f60799c5259c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 23:47:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c91831dc25040fc9148cd1011f05e4d : SUCCESS in 12m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff834ba619654cdab5b55a387ac9171c : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ed09ac83dc4c4492b9d1d19c16b3da : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9319ed0ebdfb45e99e77a48a99fb6c25 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/593149bfc6b74abebb71da55c1b47bd2 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d18fc9ed3d2d4d7b945819fd75c24869 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8fa235a5a54484a41a0c8efc861f8b : FAILURE in 7m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c772bfacad6241fe97a24fe6eb2480e3 : SUCCESS in 7m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bce5a51dc8324aaa9ac51d69b264724f : SUCCESS in 11m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/500ab94ce455445da85a5a0b20becb93 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fba12fef89e647f387cd9d2b861a2e1f : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f25d9fd56bb4b538d51851ca12c7211 : FAILURE in 1h 13m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/115f00ffd0c145f589983c7538ebf696 : FAILURE in 55m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58f62d83c5cf4f01982c87ce7d8dd19c : FAILURE in 45m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/96da47acf6e34808ae57cdebf4be1cbc : FAILURE in 1h 02m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3f7ceada9b984d699e8dd0973774bd88 : FAILURE in 53m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/126f1a4f71a54fb78e5490229bb5c089 : FAILURE in 39m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9cffd0c299a54e30962f64e5422c665e : FAILURE in 56m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40965204f5954c9883f455a523e1e811 : FAILURE in 59m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/72b50d8f4f1b4295bbf970844acd272c : FAILURE in 1h 07m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a741f7225c6b41cbb6513a7d48860fd0 : FAILURE in 1h 25m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0ce581fb84f740a1b8dd4d4f0d29953f : FAILURE in 52m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/38c42eda3926447da6991510adce80fd : FAILURE in 1h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd1ef485812e3973c2952addfc8ee30a7fe91497","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-01 17:19:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c9d1b24d33232e775e55068755a0ff0223c02be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 19:20:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/498a94d74a354dfd82396b1d4ab4b95f : SUCCESS in 11m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0759b1d2ed1b44eab686f716f7b77eb0 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87979a27cf7e4c2db446f0ac68f74039 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb51a3a423dc464bad9ba64e3785a27c : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fea28b94a3f4a35b8d09129d12e34f2 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e240dff68fa94234b96492dc11d72c60 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9d24f6ec9f42ad8a6703c4b13684af : SUCCESS in 13m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2f0d48d6d4f4499ea08b3939550ed829 : SUCCESS in 8m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5b3100fc3a1b416291946d29dab04ada : SUCCESS in 11m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fa0780b13b0547519cab01d04107f452 : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49d449f4710f4964b2f2f0e7f0a74b5b : SUCCESS in 10m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3064f7c99f024d0fb28c224796cf1c4f : FAILURE in 1h 12m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b72d1be78b1435494ef323391227b96 : FAILURE in 1h 58m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f59b859f3de4f0a8b7f86e3cb4c5219 : FAILURE in 38m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7ae3c3b9eab46f0b72486768a20cd5a : FAILURE in 50m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/76860a0bd1884179bdcf6c7b4312d3d7 : FAILURE in 54m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b8d219ce2314e9bb0247819da0abef3 : FAILURE in 36m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a50f0dd055e4e84a1b7250f30dee296 : FAILURE in 54m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d02242bd48f349e5882f58685114e610 : FAILURE in 1h 29m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c73517c336c04dcbab22afd4e3020c8d : FAILURE in 1h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/373ea29d712c4a21bb1626f183ebee1b : FAILURE in 1h 25m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2ba303d7c9a24aca80cc743288146770 : FAILURE in 48m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/192e2c39b47946cfa7ef97dd64dd4b88 : FAILURE in 1h 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f3d935a18d5ace8f2f73cb2aa9d1fd2976d9323d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-01 23:00:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"46504d7bf9cc73f9bd2af52d238a235529ad36ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 01:22:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42a35a899406480d8f229957bffe68b2 : SUCCESS in 11m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ea485bce7dc4cebb3447b36caaf2a82 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01693f090e914321b5a91844e004eb5e : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1d4ded3663c43c29eea4767f654700f : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/362fe5cf222c4c32a45675e8976e0579 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c10f2f14b5584e92b10b452d883b37dc : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42beaa1ec0a487baf7d9afb759f2529 : SUCCESS in 13m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c9f7f0e117a34652813ec0dea8ad7195 : SUCCESS in 7m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/afcb941dfc7544c5b50f5d7a349bab69 : SUCCESS in 11m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c5e34a04d8d2484883511f128e799bfa : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bf368ec645c4694919f7afde2a6e31f : SUCCESS in 9m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e836dc4a4fe42beabd985c3710a4618 : SUCCESS in 2h 10m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca318c50b4a5446e8164455ac205251a : SUCCESS in 2h 14m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f92693a446654258b02c1b40facc0ba4 : SUCCESS in 49m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b16d8ba78b4418a9357e75b815e6118 : SUCCESS in 1h 03m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5d98da8f04274d69ab6f089cd253eb7e : SUCCESS in 1h 14m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1e6dcdd73714f0b8b11c102bfb746a1 : SUCCESS in 46m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/500ac92eb26541f8a0027ba6440b1a40 : SUCCESS in 2h 11m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa1c5e28f190489d93dde11e050821e5 : SUCCESS in 2h 18m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d349921f429d40e3924620e47c44a8ca : SUCCESS in 1h 56m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5b9dcb1e5e3f4da38e73cf4cde55737d : SUCCESS in 2h 12m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7b70ef7b02ab48dda727550815ff30f3 : SUCCESS in 2h 15m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b780a45293142dfa9908e14ade5ef81 : SUCCESS in 2h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cc0a37f7e8d2d2b48467ffd89536213bccedcc04","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-08 06:30:11.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"72a0e37b992409444d4b2b6743eebb8a22398e86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-15 20:21:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5dfcfb46ef3b9569da2db4b4be13c050f012091b","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-15 20:23:15.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a433c33a1e010408f50160cbf90a803558069cd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 22:41:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a77201020ff4c4ba0a1455eb980c38c : FAILURE in 7m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05b53237c54647a9a0bb8aa259abd56c : FAILURE in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b535fbcb584672acdab8be2e1cf83a : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10bf6a858c394c5e874ee75cfd6d9720 : FAILURE in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/780fd81055f64cb199c9eaf4250c39ad : FAILURE in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d0048fefd541229e83e229c18f947b : SUCCESS in 12m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c92129790c8042ccb23e3e7c490c80a8 : FAILURE in 7m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2a89871fd5014973a45656da8e218043 : SUCCESS in 10m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c05c34cd890a4ce694a3232bfd30479b : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f05142b0cace4f779f15ff7245e0534b : SUCCESS in 10m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edf1e02cf4704c8ea1b3e93ac9e677d6 : SUCCESS in 2h 06m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd35dfc038f74d47bcc6099048d0cedd : SUCCESS in 1h 47m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/462fabc3d5aa4857aeaf98ad8c8ed830 : SUCCESS in 38m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e68f196bf9d843b9b2feac45b524f90f : SUCCESS in 1h 05m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/71aa36c952bd4d519100d31324739bfb : SUCCESS in 50m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81def7c8693b455a9d14d07eb190d119 : SUCCESS in 43m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f65306759dd43cda5d04d67c2cb20dd : SUCCESS in 1h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0b675ee4fe2347efb9466d76e101fc0d : SUCCESS in 2h 18m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/42b5049d8fd44c08a60a10c038454f9d : SUCCESS in 1h 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4d878d4dd79e4023bdf1f0450aadb38f : SUCCESS in 1h 55m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d592f9c548a0486187ad539d4c95c6f5 : SUCCESS in 1h 51m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/60e6ac1d5b7d4fdbb8658e3532b031c5 : SUCCESS in 1h 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0373dd0a9af92b60a8ee0b6a04abce40a24ecca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-16 01:00:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f7c5c28deaa08445b47e6377368f8f83636ed8b6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-16 01:56:37.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"251a81b9691071f7f1634571ac037e8b37fbd816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 03:19:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e362c1bca0234d6f89bb3bfe23058074 : SUCCESS in 13m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/979192178047415da9bf4aaca0810541 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29fa2c8608340b5adf0ef8d9ca862fc : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5de243af8ac84e6fac8255a308d122e6 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b341df7b16fe4a7f8192fc7e756ddde3 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cc209119864edaad4f39bd10224ea8 : SUCCESS in 15m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6207f3a36840493ea65258f58ba12029 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f1ff1c9da7044282b6929dc658c52132 : SUCCESS in 9m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2eec50a872554a529a93d65c25de5622 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2be90c5c50774389a0b9989acd158061 : SUCCESS in 9m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/15ed08430a7b492a9b6d255619bebff5 : SUCCESS in 2h 12m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28c3bffa505a429886a1e54c9a477b8c : SUCCESS in 1h 56m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7fa3bd60e4bd4f6f91ad4d5437084189 : SUCCESS in 39m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab764525d2a14b76aa6b880894dff724 : SUCCESS in 55m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bbf39b463c1d468db1e84ef818bf1345 : SUCCESS in 57m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e52107780dde4a90830fa9de55699f5d : SUCCESS in 45m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ee8e8581926b4c9c91d1142466cdef2c : SUCCESS in 1h 49m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2dbad34b08445f58070860b98d3f512 : SUCCESS in 2h 15m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0894113b6b74bb896a3c855c8872c9c : SUCCESS in 2h 05m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/585376e72b494a60802c641d74329ac7 : SUCCESS in 2h 05m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3cd1069f6ea44b09ac568e517b83d3ad : SUCCESS in 1h 46m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffa69470851942da8ac5c3b5f2e387b8 : SUCCESS in 2h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4c8182495fad006a009a9ff6db9998c21a65fb77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-16 16:56:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"00704f2484cd2020df7eaa3aa77ac39f5552d329","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-16 17:01:17.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"42b6336a92bf8a498fe3373c13ee30f8ea564529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 19:34:04.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9499dc28872f47908621c7327c5ba08e : SUCCESS in 13m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60c5b2cc715b43c7a42e2f5ecb56e3f9 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeea3abee3e54932a8e437f6c813b9de : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc04d506f54d4da883926f4904afe284 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a24ff849b604db88cb2a8ae4399a552 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d718407c7174882943b8ac47fc9ecc9 : SUCCESS in 13m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/11b2b13d6eda468991536c8ac34b52bd : SUCCESS in 7m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3c271e9ab07945e7806d83e6038759c0 : SUCCESS in 11m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1da56bd2e35b47c4b666291bdb228bcd : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa876cf61b3c4552b95cda829963d3ef : SUCCESS in 10m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f488a4d239fa43aeb14a08f9ab21a0df : FAILURE in 1h 04m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4acead9af095411bb301d8fadff8675b : FAILURE in 1h 18m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b686b9c802de4d3486ea1b0060779153 : FAILURE in 1h 09m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/acba84f87e934b50959ea4445191a209 : FAILURE in 1h 09m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7ebc083da1c5441eb223668f6eebb5c7 : FAILURE in 1h 02m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e1183cb86504e9c8b02fe816d26851c : FAILURE in 41m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/612bff9593df4de0b3409c9653dff453 : FAILURE in 1h 10m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f0ddd4e8dd34f9cb8cddc10f45804e0 : FAILURE in 59m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5379ebb150034d51b6a407a403352b54 : FAILURE in 1h 40m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eebd1dfdd34b4ae3ae5eb8b6718196dd : FAILURE in 2h 04m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fd7740774b46488094078177018c2a62 : TIMED_OUT in 2h 35m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/224ad3141f7a47c58366637a3ca234a0 : FAILURE in 1h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"62a6d6e2bd34184fd9b9fe7a61c0a57ddd7160d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-16 20:30:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4f85e0089d53a46343afcf8dbb0a147f970fe462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 22:59:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9241cc20deda4e5797846d50d5fe363a : SUCCESS in 12m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8f7a331122446d981f2bdf2f58e8c13 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e67fd1c49174742b236778a2dc08c81 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/971ec081356c4c2a9eb64ebb9c9a36d0 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eedb2010bbab4dd4bf93b4081e63a361 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50faec11b7b741e78d64640044252742 : SUCCESS in 14m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4e8972820d924995b423ddbb0410212b : SUCCESS in 7m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/198818351392430cb02140d74c556791 : SUCCESS in 10m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d39cc56a3afc47339f556519f2fd80e4 : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/83a19a9d25f04f4fa473add3244f2723 : SUCCESS in 9m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f48225f75c645869f016207d5158d5a : SUCCESS in 2h 02m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85bd46c29e3d41a9a3264092faa0a2eb : SUCCESS in 2h 02m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e38a28f3d2845cc96b2ce1d91db8d20 : SUCCESS in 48m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0a5bce39f86e4909a6b2fe1181413660 : SUCCESS in 1h 05m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ab2a6809127a4424b8468c6fb578bc3d : SUCCESS in 1h 10m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3950ced0838b421a9593e0b56167ad46 : SUCCESS in 46m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7fe4248e42c74298b66cbda323a5a5fe : SUCCESS in 1h 47m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe9cd0fac5bb4f0bace902723f0863be : SUCCESS in 2h 24m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2c69b295ac4b9ead80f4626abde0dc : SUCCESS in 2h 03m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9a11f4785b0a4975b99a1c38af2dfd56 : SUCCESS in 1h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6fd530217e6040bdb58a0c42265a3e4b : SUCCESS in 1h 31m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f86e7246dacd4bd1a6918bbe2564a629 : SUCCESS in 1h 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c5eddadf6e4f9fc9e864eb7f6ee8a26c95807d11","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-22 19:43:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4afe49853adbb95028cb525d57a57eb64731d995","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-03-26 05:58:52.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0695fce8f65b7278b930252964fb6e5231ffb47f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-26 21:55:16.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"911e750de6300839e3f1e624552aae2b77450d9f","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-03-26 21:56:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"349bf02ffe2d469db041cf475f8f65a753c8d734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-27 00:16:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1dc2a0bfb7e429e86029d2e67b83601 : SUCCESS in 12m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bb68abb9f344fa69fc2af003df64308 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c86bc13667544db88da2acf978af5d2 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65ee2993a74e487192aa498d87a51630 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c08336d86d6a4c6c9ec4a2548417e649 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e8519da585b4d30ac7a74ef4cede8df : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d990296acd643279b065768d5c050e4 : SUCCESS in 14m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e10dbb18b9124be19188f6c3d20415b0 : SUCCESS in 7m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c16eab7d50ec4dc98affa8dc0d1edf30 : SUCCESS in 9m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4d099850ba494e38b1b4321455fcdbc1 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bf2de20f597420b9bab5bd17bbe36ea : SUCCESS in 11m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d570b9b311b2443698dac1ebc7c2a1db : SUCCESS in 1h 32m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb43620019bf4d31a5ffe6ad1437eaa5 : SUCCESS in 2h 01m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/95da286227a5400ba25a01f564a9fcf8 : SUCCESS in 37m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/554276096304438fb665e60c80504cba : SUCCESS in 47m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c2d31b8553b4229bb5649c9a2166cf5 : SUCCESS in 58m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d50bd242376428095da02357159b912 : SUCCESS in 37m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/21a4c4e02c98463ab1ad34daf141b496 : SUCCESS in 2h 16m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c015bf64e3d426cbfa2a02ed64067ae : SUCCESS in 1h 38m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f20baeac593d449f9b56259236814c98 : SUCCESS in 2h 08m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d3eb9c6343da4b598b3999f411fd8225 : SUCCESS in 1h 51m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ed6ab8d49b714bae9fd0decd9124b9e1 : SUCCESS in 1h 51m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc647ea9994a45509d202a37d5ac3744 : SUCCESS in 1h 55m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f2c025ab5c5e85ec47e7746e5b83c95e35271a28","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-07 14:52:16.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d6edb56585a4b3ff85d35e0d2c6a7cd9cf4a7acd","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-04-07 15:13:12.000000000","message":"Removed cc Susanta gautam.","accounts_in_message":[],"_revision_number":11},{"id":"97fabf39797299169966da4739224a60d8af1c04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-04-22 15:38:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6003854548e2e487ffcdc18243fb5602b3d6a957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 18:12:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35106a19c0f840978d6427686744ba08 : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eff938e8092446d68a5d2c26a1824186 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91faec5ca0374068bd563d3f968eca84 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/759be813168d4bdbb8410e87429cff00 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c13bc2874af94961a1b754fb1f8c42ed : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c721434cd9b349bdaa52ee776bcf31c6 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b57d571fdd44fab374c872980ad392 : SUCCESS in 12m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2707a447fbec4bc5afa8848be74818be : SUCCESS in 7m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/798439f379ed4c11a25356254f7bba1a : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3455569afa3440328eb5a400e711f475 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f64067a333fc45d2a796b67cd62952ac : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba925aa1ac424cd1b90171890f35dacc : SUCCESS in 2h 27m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e393b432744b415694317b281155d142 : SUCCESS in 1h 52m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51f7119740154c60875c6af9276103c4 : SUCCESS in 37m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5817057bf57249b6b179c6c458b7522b : SUCCESS in 51m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/13b6260907c64168a5f7dce175c3971b : POST_FAILURE in 16m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/449faa20528941a581abbc71ec5ada5b : SUCCESS in 40m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4d2fdf58e0140dc8b763dd59f70d8b5 : FAILURE in 1h 43m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29b3e954fa9d45b28ad2425b271f0405 : SUCCESS in 1h 44m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/37a3d831e81f4057bf028d699ce82d34 : SUCCESS in 1h 40m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/30e0f8052ed84ab4996b29cd6e181eb9 : SUCCESS in 2h 03m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/889148fca9f14ab283f71a89bf8bc7a9 : SUCCESS in 1h 36m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/cea4919e1aad4b9ea4e5919dc012faa2 : POST_FAILURE in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3533be68b60f6af7467ba39642fb9a0cc05d7261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-04-22 19:30:22.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d71a909860b30b55078ae85bbe066d7b545c1204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 21:57:46.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1c90406a37747a6be583c2e6c0f2e21 : SUCCESS in 11m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8cca27774b4488fba6174156fad5c31 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34f1426fcdf044f495e0b6c3a716f45e : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/310bcdc33107486c983b7fe5b0d6fbed : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2def20c1621044cba1d9550c829deaab : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fcae1e3d6e44177958ef5ae0bc4cfa9 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7176f5ed5c8c4eb6b47c603d8de7f00f : SUCCESS in 13m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c7fcdd86e2ec4b23ba8b17fdcf1b7d0e : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/94f2b866575f467b9eaf51d3c0dae25c : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e35bd9aa590845ad92701426cec0166a : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3c40b7b5f0347f09cec27be39019229 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a5346cbe81241beb24c2cd0986214e9 : SUCCESS in 2h 25m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40ed6523d2d64ddeb460c8e400a998bc : FAILURE in 19m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b643f405acce4b7aa9c948c28e05ee5a : SUCCESS in 40m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/03fe21691b844cc9b88b406fe199980d : SUCCESS in 1h 08m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3933636e548f490ea98aa6a187a5098a : POST_FAILURE in 17m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c2d080cc1194216ae11cffd3b17a2b7 : SUCCESS in 34m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/256cbdd6abd943fe9f27d0088d89ac37 : SUCCESS in 1h 50m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46ea1ad4c6734e20b11dfd688622b307 : SUCCESS in 2h 06m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ddeab68d30524c06becdd33dd2febc25 : SUCCESS in 1h 56m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e8042d82fb44be4833bca13e2831da0 : SUCCESS in 1h 41m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3f58be0acdd94b278b1243970f524d83 : SUCCESS in 1h 49m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/30fbe883a7974a25b28295de8663fd74 : POST_FAILURE in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"43cf8ff6de3a512277196e90c7daeadd3bad0b24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-04-23 00:11:13.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"68a0c0b506a256d6c4b327aa35aa95c849bad77d","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-04-23 00:11:51.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"9292ba9854199065f2a091c71b4b564a7ee8a507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-23 02:30:56.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/296691f32a074fedbe3fdd325fb370de : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/046f19c610d04aea81799c3b6fbb82c6 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c046bfb07f4f08a9014e8385c181cb : FAILURE in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cbcaef3fbc04b49a873e9b90a9850f1 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a77769bdc4d4267b69bc506c8bbf35b : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7199b0f270c245ac86905cac5a955121 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68864faa470a45519b6040d3ae86d889 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef3fe1916c2e4a0391aa1579ea49cf69 : SUCCESS in 5m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eef49252fcc948aebffe1cf5d193942e : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/deb36805cbb04694abe2d4eb6497167b : SUCCESS in 10m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98724677dde04307978cdfe2672161a2 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56a74356f633436ba9ece39ded92f432 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/764630235ade4729b70b4b5c97f2519b : SUCCESS in 2h 17m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afdf1b61880d411dba9f3b00a8a8c21c : SUCCESS in 1h 40m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fb97b4d36f0c4df486fa0c5f552b74ea : SUCCESS in 34m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7703d6e4277d4d83a888b25b9f678d0d : SUCCESS in 52m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ff72c39d4e141cfa328cf6a80846e3d : POST_FAILURE in 16m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65aee2a65bc248e88d7c2113fad7c446 : SUCCESS in 37m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d9c10a960a741d3813efaff8c71291e : SUCCESS in 1h 39m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d44cc41503ec4b158a23ae76a5b613c2 : SUCCESS in 1h 59m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9e9061babb6f40e4b22842f88f4b6f87 : SUCCESS in 1h 51m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3564d0ae18e140dabf250476e55bbdb0 : SUCCESS in 1h 42m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e11d64cf05b94adf94ec933335cfb624 : SUCCESS in 1h 27m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b8e6623dfd44c159fc29bfa92355991 : POST_FAILURE in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a7b22853cc9f4c1a5534df1afbeaa905bcafba60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-04-23 04:01:42.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"345fc8be7e6df14db4f240696c27fa780757324d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-23 05:53:29.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dfc0d252912481e9e38d062adf76834 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea906406a9154b92a34e339ac3e0417f : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b736201d0f4979b6964cc73ef914a8 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8933d13ba0bb4f50a7a96acb8cb698d6 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406d41cfd5ae421caaec05c8b94b1c84 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03dd016470e14fe495cad3a04430ab8c : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5fe2994d5264abeb2fb06434c42057a : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c98e8de648454a1492d63b1c27aded0d : SUCCESS in 4m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1fc914b288524b5fbfe12e4e8ffc3048 : SUCCESS in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ee36133b7a1f49e0b369d4394da16fc8 : SUCCESS in 8m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/285f03e7e82646888258178c498017e0 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f2c6f77f99e645259a866b84aa677b65 : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e032e0c83b7440fa0ec9555cf713274 : SUCCESS in 1h 31m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73ad213e679349ea9b3dd1b0be148470 : SUCCESS in 1h 39m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cbda5a0614a846b48e2be64e2240646c : SUCCESS in 34m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c27284073ac4b989b6359cbae9dd4ef : SUCCESS in 53m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6da2c0331ec543d1a7145af436df1f73 : POST_FAILURE in 13m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1e76777fb5446679b3c3b99d63cd2a4 : SUCCESS in 34m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/70f9c99b22d04b888c76b8bacc1128ac : SUCCESS in 1h 29m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f71e065342034c3085b4b384cd75af8b : SUCCESS in 1h 50m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/767c757245f8402ebfb376a29e133832 : SUCCESS in 1h 32m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef6fba77eeac4c38909a47b0f42416aa : SUCCESS in 1h 36m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/785d110902984014b8f754362a4ec675 : SUCCESS in 1h 39m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9edf99aaff8248a99a58daee1e1873f3 : POST_FAILURE in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d9abcd08f81cde6dc9484ba58fd1fb7d612f4772","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-04-26 10:52:23.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"89d592ce5cfedb77c21b19ee85eaa9f5ccc3ce37","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-05-11 18:42:39.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"8b23fa96024382c9d24cc0f87dadfe66c7be7374","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-03 09:13:04.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"e4749f1290b366a31784f8f74e0ba4396d9ced4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-06-07 20:14:56.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"03f16ee71b8d94d2d4296115fc1b9ac5e7db8d4a","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-06-07 20:19:21.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"2041768c9cfb562d507cba086618eda8d3f73ae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 22:54:36.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83cb0fa1cf444237be806d9a7b042650 : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6bc3c0bc66a43e3b52f90dd9ebbb053 : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8efd72a70b8844eab95243de8d4755cc : FAILURE in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/702a3763d89c486f90404cc9d048a4d8 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba8fc4eb43fb46c3aad8ab2e0ae93803 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30698f4a28634408a1689add45ea2d09 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/369986c5c99d4a67a6cdba8e9a38767c : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d0fed01fe0d4864bc34512f3530b183 : SUCCESS in 5m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ccad0a3dbe92418c9a6c98991e2cff12 : SUCCESS in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6d546130b3804bfea37e5d4e489db1eb : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8be8a73899334f6b92f611ffa884b2fa : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/998c0a11e0e144449fa3cccbfdb6f23b : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a52b375b97ce4759ad7e476aedff15fb : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cffb9b95294c450abbe42d6850b002ca : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9fc70a91b66430baafe11bbd1dee008 : SUCCESS in 2h 33m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c4d77cd4edc4f99b12d6968d8474bba : SUCCESS in 1h 46m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9fdb2cdf45f34febb0315c27b87aac52 : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cd5d8dee636c49babbb2250d4847dd3c : SUCCESS in 1h 02m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d756df925fe0449ba9df33ae4c242a95 : FAILURE in 17m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f52a4c1384c42c08da6d1d2cf10fd11 : SUCCESS in 39m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c9972c283ad54c429384dbc3c228781c : SUCCESS in 1h 35m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2eac796bb8c41d2a193addbe90c5787 : SUCCESS in 2h 11m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/527c9db64d4f4fd9bfc72ca09a13af96 : SUCCESS in 2h 02m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/42dd410c5b494681a73849bc0a6c1ca4 : SUCCESS in 2h 04m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5ee751672c9b450fb91c9e3434d04eb4 : SUCCESS in 1h 47m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3c646735db740829947816e9058744b : FAILURE in 13m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":16},{"id":"5cad2b20c0a0ceb974ac7bb716564f80425240c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-06-08 16:19:02.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"8e763deff864ff4f60ee81b18924be116d38addb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 18:46:53.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2743eab3ca4affb252a096425f56a8 : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a0292103c134db8afe43fdc5b0efff2 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ecc5c26cd774248ba4638be66866097 : FAILURE in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a71e38cbd5f3488eb6004d1fddefb0f3 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bb61e155e6842a4a2fe4268861eb8cb : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2eff6f7a392b44fa9937c9fd61e27d75 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4a50ee5294e44a6a5cc329965131ec1 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11011374f528405aa47084512814ea75 : SUCCESS in 4m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d9c91f3d6cbf430db457aa36cdb39e15 : SUCCESS in 6m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8686beeebaa4411f9c27f79fd2c705eb : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ab8fbef7e3a4ae5911ca8bc6f30526b : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8a06694808746b1aed6ac01a1c663b7 : SUCCESS in 8m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/393ce1d814344493961ceea2e06b9004 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3802dff798ea40dda9155e3b475fdeec : SUCCESS in 7m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b308180a32748a9973b9e4f83d80f60 : SUCCESS in 1h 57m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/166eb7bfc3044795bb04a6aab7bb5bef : SUCCESS in 1h 26m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a06f5ce245f455ba8f30964d9083cfb : SUCCESS in 42m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b25e38d92f374a2581eaac81de1e2433 : SUCCESS in 53m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3bb76e57470948f486e669cf25f4d07a : FAILURE in 14m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5925beb77024a03952eef5591e89a8f : SUCCESS in 31m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae4572f83c904bfd98ca054c168e3cc1 : SUCCESS in 1h 34m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/01ec2532ef3f4159a65e2e21f028f854 : SUCCESS in 2h 13m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5fd8e2464db742a899be24ec827e57fa : SUCCESS in 1h 30m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e987448d85e443f87f46fb9b365d00b : SUCCESS in 1h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/91ee0b41a52343108b5e4626bd44b7d0 : SUCCESS in 1h 59m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cb1a6d7b94f4bccaa0a813d41571254 : FAILURE in 8m 59s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":17},{"id":"b8a186896d9bbee64f49ff3a837718a5c43051ee","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-06-16 21:20:52.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"10119e382e4ab891b3641c86db4ce56c1b466be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 23:32:13.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ff76a5d16a04934984d99f4b5d44379 : FAILURE in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f68206bf46341789a0e3153992d7b60 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b7b6936d9549bcb82d221098e73a57 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bbd94bd812b4451ba22660f4d02c974 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/010bfefe27764f78b9de72eac5e62171 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc0e40f0e75e4fbea217a9fc05f501bd : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d30df73e1d3473ea7058565ecd315d9 : FAILURE in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd7f6864a7d64a2ba3bd3a170d413d70 : SUCCESS in 5m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/20234e80b86847eb9099be5cfcafdbe8 : SUCCESS in 6m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/17913605774d4c69b1f8c9568bf109ba : FAILURE in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d9be1c3a885848cebe10bc9ff4569704 : SUCCESS in 5m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/36bbebd1f16049dd98ddb3b3e18af43f : FAILURE in 8m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/672e03b300a743c0b1fdd93c95ab80cd : FAILURE in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4264a8c8920941c9a0a0a46fabbf8b55 : FAILURE in 8m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/62e22be04a2e45f89165ef17f280b7a3 : SUCCESS in 2h 06m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d41fa4e7df44ca49d0e5f179bf05f7d : SUCCESS in 1h 28m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f42cb5ff46694b3c9819d61eebf7643b : SUCCESS in 34m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef818052150d405e88726bdfb3ed6432 : SUCCESS in 52m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/03f912633e8049dba7601f84ae452888 : FAILURE in 11m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94979526162d43f0b587328e6b8a9ebc : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e7ea8c31dd841a88110e4c8085e81bb : SUCCESS in 1h 40m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ebe1dfb28c2340b2b8eb171007636a9b : SUCCESS in 2h 05m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f8cfc8c24caa495eb632ab18df3b545b : SUCCESS in 1h 45m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/76f21fbead7949039d0117b2293162f1 : SUCCESS in 1h 35m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5793bbbdf3724b128db1e6ae73062f19 : SUCCESS in 1h 24m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/716248e27a97496a978691b9fbf34c04 : FAILURE in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a5587cf4ac2d9ebc33112aaac55c1ed7462d45f3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-17 06:14:00.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a40fc187d0904a566274ea1d2dd76886ba2daf48","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-06-17 17:51:26.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"979f0b82a47df3ee0cb9014580ec2ad95aab9df7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-21 15:07:13.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7f962634fb8e6183003decf97be5a6f277bcc93b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 17:31:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b85f9d8c4db64537af20ac5b670904e3 : SUCCESS in 12m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58b769e1e67541f78ccb6b51991ebe70 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56666b0e1bfa420b9daefcadebda2a84 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a558cc46610b4115a4ab2151b1059e36 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57e410061de747dd857be92c80037719 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/274bc4015d3c4c4d90975c26537b0998 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa676ee0087417e9448ff9c15f839af : SUCCESS in 16m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a391190f86143ac9fd8bf05a1e02819 : SUCCESS in 4m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3f39b18e41b844c5b4ae2d0a87e6ab6c : SUCCESS in 6m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b8831540b79844568d8286e7d6fbe534 : SUCCESS in 9m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/37362964a8ae4d2db11950866dcd3a21 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/99853b894a10425e921af77f5c5d4b6d : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3610f3b0a71e4293ba2899d6b510c913 : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8678619f4a74b33b3e20254dc9cd2d1 : SUCCESS in 9m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06e3b30efb1d4be4a72461dc8d8c49c7 : SUCCESS in 2h 17m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fb777fd645949f48dbc60054ebc981a : SUCCESS in 1h 26m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/877ca95b36284f67ae0e2a157bcecbe7 : SUCCESS in 38m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6cc5e7a7ecb4f7db8cfdacd822c9542 : SUCCESS in 1h 12m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6ff415e5b713472bbb4266e3e4e338de : RETRY_LIMIT in 9m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d7e1c66f0f14bf2938b2e66b64a3984 : SUCCESS in 40m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/69fc80edebdb40dcb77528ac4715896d : SUCCESS in 1h 59m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/854c5697df8e4792aab39f9250a6dd78 : SUCCESS in 2h 07m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2124f70979164b6281b836932bc7aa00 : SUCCESS in 2h 09m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f775524a3fdc4577880e66c3d9989b91 : SUCCESS in 1h 51m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/18de8944eebe4967917d2ef99bbf7b3c : SUCCESS in 1h 53m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/92bae9c767064c2abc107ffe00e07986 : FAILURE in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0d76f28f36f6ea02b05aa73f11334e3c87e92302","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-06-22 06:50:25.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"fd69fcb1ad0bcd46ba57f06c4a3b00537bf4fbc0","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-06-22 09:36:18.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"612c1c60a6d5f060526e9e71e402c6ac150d39e9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-22 16:02:45.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"5e7b23ca791fc7d28a0d9e10ef35efbcd20e6386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-07-06 16:22:43.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"222afd47ffad34ba07173a0a163e365016488157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 18:36:56.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36939dacd6384399a88c3507fbaef07b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28aa24ca9da74e4b9af83c686c423acc : FAILURE in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c426acb021f44c5ab77d75bc65f74d47 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d77c3a9d382408285bdc9d1b7acfa7a : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0cfda4bca374c6db5eef6e346316cca : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23e9c521530749c186f614d12182cf65 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/480e58f14d204857b7c15f8376379216 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b64c3a3e2a4fa886a10468c0f6309a : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0270ab3ca3a24b21b202d512d334ba0e : SUCCESS in 4m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e23d66c56a334a0fa487b0482c91e785 : SUCCESS in 5m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f94ff8b855494decaa60050b1dcfd83d : FAILURE in 6m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/feb6927c95294c71b9ac9cd59ab21b7e : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/db93a84320cb431abf29653f165807a9 : FAILURE in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1baadcda57f8403b85878ed4dca9e56a : FAILURE in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93393a29237d46a192667b5401a8aa50 : FAILURE in 6m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e384649f6c74b35b8651cc499e2c96b : SUCCESS in 2h 07m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5858143da184a45bf386e028fccd509 : SUCCESS in 1h 51m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b44cd59f99ee45e68948b000d60cd44e : SUCCESS in 35m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6050eebada6f4c8d92b5e85f0f6d298b : SUCCESS in 1h 05m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8017150e7ae14b8bacb186bf7fba2e88 : RETRY_LIMIT in 8m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c8b8ad890744522a0f77afa98f52e7e : SUCCESS in 43m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80d1b828021c476983ac48c259ea4e4d : SUCCESS in 2h 11m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b7379503b44419780ae03252daba5a0 : SUCCESS in 2h 04m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2529912d88154dc2889d11e6bcc98b53 : SUCCESS in 1h 29m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5cc724c1137d4a1d9c3752186922c388 : SUCCESS in 1h 25m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9d1dfc4b5ea84217bc48972613dbc53d : SUCCESS in 1h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd264f018f64496f861a0346444df73c : FAILURE in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0e878cb6db8902c79ea412afb50ed7208b7e09fc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-07 13:47:02.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"ed31c17dd5d16fce49b6b768df72325e5920b100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:52:25.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10800180926a4a3c8b0d3715ef5fd75c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bde7da92ae3e4ccb8a2a66c1634eb2eb : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aceee2b15c04de79fa714cf47a0331b : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c58efdcc17645059ae81947a0e8341d : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a31242609bac4390ab4bfc3bc55fad2d : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82c4f8e6c2c94ae1b48f70b291111fa6 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4fb09d237894314abed0e448f369b60 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d72e4de988b54b55b167f0b4487d514e : SUCCESS in 4m 05s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ce818e9eb54a4b32bd8e369996b2884f : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c1c4de2efc7841f19892c30259a28923 : SUCCESS in 7m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2074ab6fe7f94ce3b2e84e7f9acfb577 : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/951fddfa04d344a3b92f9bb70a38ba76 : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a90189d7f0b4a568c24c73592832f91 : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13f6768dfd5246caa679bb21ad3eb4b1 : SUCCESS in 8m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/752e308c72c34bae887f8edad9fa809a : SUCCESS in 2h 00m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ebae5eda78d464ea2d58b9656574752 : SUCCESS in 1h 19m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a34c62c6006d43a7b297e6125d314525 : SUCCESS in 35m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bbaa2d0b62964edb83019f9d937d0b54 : SUCCESS in 48m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c8ff34bac72447678e83ee429dab6275 : RETRY_LIMIT in 9m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6f7939fc0ef4687ac1fc196a4f70a10 : SUCCESS in 29m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b64706646b72412e9825cd72146e301b : SUCCESS in 1h 23m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ddd40835e5ec4396ad02abfbd8bce7c9 : SUCCESS in 1h 53m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/930078c60cc44636b92ab297e1ef42c3 : SUCCESS in 1h 46m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd41d254c1bc48acaa3a0fc635be4811 : SUCCESS in 1h 27m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/91e7dde7ee5d4a35ab7c4a8698a49f35 : SUCCESS in 1h 23m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/b932d17ea25f4d9d8b20e59d2c97fdc4 : FAILURE in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"d4e652348bd3e79238da888be1268aa248c5de9f","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-07-07 17:35:28.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"4d6f4d6f0c70b4baa6856d9b042966a004b57bec","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-07-12 06:01:26.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"f2550bf532035aaeb0b40f1a8507d282a96d180c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-07-12 07:03:09.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"682c8cf3992ee402134f41629ae086eac3d49a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-07-28 18:18:46.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"87ca6d81399e151e9e73f4f3ba43f180619184f3","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-07-28 19:30:53.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"7e72e3e346824d3e0c8a19328717b69dced31e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 21:00:51.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f250dbcf0cf4e69a5bd1bdcf2cce6a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb0e98b51d38455f9099c8d688918b24 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a4d029137546fcbb7b61dabdb47a3f : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fea419181f74406ba682eada675d18fd : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3b8afe797a749b1b6691773e28f7f12 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1f203b455f94fcdb148d816a3e30a0e : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6524fc5fe260425ebfba1d96aed86d05 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0fd9c91f42f421a8f99fbc9217f9443 : SUCCESS in 6m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5959849ad3f84959a42a4e7d990c626e : FAILURE in 6m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/00ad6a4f6167451b886dc2c6829cccf6 : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/882ca0694f39433aa86b140773e8fb53 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/141b1e89240843da80353af5c2b8931d : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/41c65e6b2ee74ac7889a884aab686764 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/77a0ca3561434dd39113d511aeb5a4bc : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2493a0198a804472aedd77716f188902 : SUCCESS in 2h 22m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5630e9936c604eb9b4f9f025cbc20273 : SUCCESS in 2h 13m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1aec8f6c5fc84b4eb7341e183841ba52 : SUCCESS in 58m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dad9ba5e2e2b46dc85d64cf6b0da07f9 : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dcbce9442574429fb53acbb55c8798f6 : SUCCESS in 1h 33m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/048d146aea8e4de3b7a8e6236cc5e8e3 : SUCCESS in 54m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b33dca7a97d5491c9de9e7eecb601daa : SUCCESS in 2h 05m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b080a65e2da46089ad48c86845420cc : SUCCESS in 2h 38m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/77d6c322955940d1ac7073a11c7c24e4 : SUCCESS in 2h 17m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b481fbc49a01488baee54d16ad343c21 : SUCCESS in 2h 10m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c31a6c3a64a646469be3cb4d9ee744d2 : SUCCESS in 2h 15m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e89be8de3c842f2a8765f39b7515019 : SUCCESS in 2h 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"3051e3c89fbfe4904d82e4f06d58d992b577ea23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-28 23:52:00.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"ddf915cb260ac0deb5cd6217261142b340c33b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 02:53:38.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25e4806180254a54a246ecc029780221\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf5fbd2e31ec41f6ab2aebd35b9535cf : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b80eeafce36443bb871420ceb63932e : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2411be912534a9f857602137323519e : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a6db28e1d26421f94a879166eab880b : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4024cde37fb487093c4589034fe3ce7 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ec0f1f974f49be9c6a6a46134a80e2 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd0494ac041d4a1f95e6c4508a023439 : SUCCESS in 5m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6e561285e3264bbcb99271334456127f : SUCCESS in 6m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d5683a49584f4ebbb1073ca4de67c350 : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ad8760c577d344b599e199eae1fc3857 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37deaad901c444ff9f243668ba713a0b : SUCCESS in 9m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4e8346da3f94d7ea34db1442dd142e8 : SUCCESS in 9m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0fd325704c1d469eac54b0baef770aac : SUCCESS in 7m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/192dee6c0cb14729a877f85acb122f0f : SUCCESS in 1h 54m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edac2cf2e4a04e1b8cb4f9250806d462 : SUCCESS in 1h 43m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/87704f2dd3b24fd88921dde85d76931c : SUCCESS in 43m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c93e4e78bef6439f84994a838cea54c4 : SUCCESS in 1h 07m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9db7c4962612434b89bef7442fcc48e7 : SUCCESS in 1h 16m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd366c046ea34a1ba791e8abfd42942a : SUCCESS in 40m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6cbb5d6d32224a56915e57d1bf953a76 : SUCCESS in 2h 29m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d7186217d8904a4bb98aebc2d827821c : SUCCESS in 2h 34m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/edd2184ab012459b8105ee03db0560fa : SUCCESS in 2h 14m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/834c11eb38384a7e9f089a08134ce119 : SUCCESS in 1h 50m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/840f98d1ccf94b5dbccffc5e53e31de0 : SUCCESS in 1h 49m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c25b889be0b44c03940e8b85a2ccf67e : SUCCESS in 2h 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"099aee7884b860b64a2bf0f40c28b3c693e0fe81","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-29 11:06:27.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"f649fce98a91689fb138a7f9bd1da587dce214e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-07-29 15:07:38.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"0f03cda45d1c750f46dd629f295c230dfc31c688","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-07-29 15:12:29.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a6b7a6cce8cf55c2564373e719cdd4000a4b4e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 17:45:05.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc866360c1ff494dad866430bdf3d089\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90b6fe08164543f58495478d1c36f5ee : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daaa2a3b44724bbb83894012c694f248 : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e76579620c9a4228a6708975c27a8757 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06ba9682514b441c9ce0b5810d58e84f : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79f2636b07b2453a90c05c6334d8dedd : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/940e68f3223b4040b071c67637d926c2 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6769b8467f344b49a07f313a225fc622 : SUCCESS in 6m 05s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/19fca1773fbd4e00877b0eba6ce15d6e : SUCCESS in 6m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cc0cf0c34f434d18979f8ffa1d2bffab : SUCCESS in 9m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b1d64ef506f4f0cbddd43280ffa70b7 : SUCCESS in 6m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1bc27bfcfd5541a7abec0b57ef4211d8 : SUCCESS in 8m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de16bd0e77c5436d9e45dc265fc5c568 : SUCCESS in 9m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/77d2e1a731d2408092b09263c9d0e726 : SUCCESS in 7m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c6ec217ccb3647838fad5cddd4456fa5 : SUCCESS in 2h 16m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8992882387ae4398bb4b3986f93bd8f3 : SUCCESS in 1h 40m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/739b312d10c0473d89825cc9868bf01d : FAILURE in 24m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5ddd60e5d18b4903be803aeadbdd2194 : SUCCESS in 1h 30m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/152b75499fa940f3b217e538f980921c : SUCCESS in 1h 30m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4d2f548c62e416cb01e408219fcecc4 : SUCCESS in 57m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf2383af0a4f404982a48fffe0bd32e8 : FAILURE in 23m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf5fbcd795f44988a66e16c829c83aab : SUCCESS in 1h 57m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1a01c6db79ab4893beacabb3d0a0abc9 : FAILURE in 24m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/23fcf932a1f14b5da7e1af6ba18ba660 : FAILURE in 24m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/32fba5974db14cdf88ffab99820f8039 : SUCCESS in 2h 08m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/115783d6aacf4b88bd4cee84c8d6fa83 : SUCCESS in 2h 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e2979eeb6ef43fcd7df0f13c1fb26275d82f7fbc","author":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"date":"2022-07-29 21:23:51.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"1108ae6a3e8d4583895d7bda515c61c8ec820480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 23:57:15.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da7e0325610421898acbe7221ee606f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/316c374530db472b8f59809c1b7eca1b : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2275d79dbb1c4b6b84067b6f92f597dd : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5175d3b6328c4f5983caebbf2677204c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbb6d0fae7904989a0e2d33bbd62e514 : SUCCESS in 9m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7a94ab5f96444878a5ad28211f65e39 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2bcca0a48ab4cff815dcf2ca0f22a80 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee3e0792d4a949409d3b23b13f193f83 : SUCCESS in 8m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9dc96f7f08904d75bd871ee4e3a14218 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f32ae400c35d44b9949357148ef43323 : SUCCESS in 9m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5640a2e5bab645f6bc9f6a7d5bd403c1 : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b20e5b5b7c340f19f6eac37d7df46f3 : SUCCESS in 6m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00d53455e8c64a0c820c3d2dc5032568 : SUCCESS in 6m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33c06cceab88436d97d57fbfee679a6e : SUCCESS in 6m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b5fe90928f1b4d9982660f4caa422088 : SUCCESS in 1h 37m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ea7b9611b344437922e0ee7f6531cda : SUCCESS in 1h 26m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc8ab57d74084657bf54ea5a15d0ddf2 : SUCCESS in 41m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd577fb94a9545d097186f742b7be617 : SUCCESS in 1h 01m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6b77290633154ba9a01cf7e65710cc12 : SUCCESS in 1h 04m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82cba6c5116b47789d59dadd4aec423e : SUCCESS in 39m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/82cefe7d750649ba9acbead33a246794 : SUCCESS in 1h 37m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ae9063001af4413b9500190ebd9a9b4 : SUCCESS in 2h 17m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5a259399b0f44488ae7e3780ba013f1f : SUCCESS in 1h 44m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd96d17b4ab94f099ed32b1ac254a8b3 : SUCCESS in 1h 31m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ceb5c38f6278424196bbac11f5dc2912 : SUCCESS in 1h 29m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f038896778af45d5ae891dec960f9bed : SUCCESS in 1h 55m 34s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"46318bbc5a7763ec6a16854193ea1c3c1842108a","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2022-08-04 18:39:28.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"bba1fdbc6c89eb23ec8afac0b3f6aa9cc35184e5","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-08-05 12:26:10.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"1508eba1f3a6683963fcac0975c23f59b4fe8aed","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-09 10:48:02.000000000","message":"Patch Set 20: Code-Review+2 Workflow-1","accounts_in_message":[],"_revision_number":20},{"id":"91eca61c1918c0a2eaa9b977dc19bcfe3c9b38e0","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-09 10:48:14.000000000","message":"Removed Workflow-1 by \u003cGERRIT_ACCOUNT_34429\u003e\n","accounts_in_message":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"_revision_number":20},{"id":"c7e5e65b1d13018e9a3b9b37131e9f303864718a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-08-09 10:48:26.000000000","message":"Patch Set 20: Backport-Candidate-2","accounts_in_message":[],"_revision_number":20},{"id":"ab0799a489dd7ff5d948a3b51e0ae468e2bd0d9f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-09 16:41:58.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"5092a497a807afd8be664cc895a7296b955919bc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-17 06:36:56.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"83ae4b7de8ca99a9cc446a763df3676cbccdd05c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-08-18 09:41:50.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"83459ec4a53c099c3ca868f7756c9b1baf7de55b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 09:42:32.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"e7d60d99262677997baec8a0efb1f885c6678d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:50:37.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ed77e7280ed48998f197e2f16d653ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b64f8a7346f74254b5405542d4b51a03 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93e303a9e3b146108c85573d11692209 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9cb126bc72249eb8cf7cd8c42afa2cf : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/108bfe6308df4f72963ef57bd4ac3097 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5810d3ed32432497f03fe54d5d49c6 : SUCCESS in 5m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/39f88123c9e24b9baccc3eedb39ad36e : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98b0c5634eb740e1bd7f83c0917e20f3 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77bbfb80d77440c18b0c890d1277554d : SUCCESS in 2h 07m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/778832c409744a03a444764aa5318063 : SUCCESS in 1h 42m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b78e947752e4fe2bdcd785cc72cacc7 : SUCCESS in 42m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c14ff45b877400b90e6763cd3d2c949 : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":20},{"id":"a24331e85c8ce7c19e6ff7b5c34fd4b8639af06e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:50:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"5819fcf4bb6446fda8e83f9510d4dbf6713c8829","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:52:02.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77cccfea90284dba8c2942c2503898b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee1da60c1ef4052bbd4453156ac1840 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a626fd240fdb46bc9e8f071bdc9582c6 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"70257eb6b36eebcbd38650e694d7f951a8931cb9","revisions":{"b182e01cbd20256fc534b23edf49d59cd6aad1f2":{"kind":"REWORK","_number":1,"created":"2022-02-25 21:10:42.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/1"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-25 21:05:33.000000000","tz":-360},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b182e01cbd20256fc534b23edf49d59cd6aad1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b182e01cbd20256fc534b23edf49d59cd6aad1f2"}]},"branch":"refs/heads/master"},"7be83e6bc041408dad5f4f1afb43e194b546e68b":{"kind":"REWORK","_number":2,"created":"2022-02-25 21:38:21.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/2"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-25 21:36:18.000000000","tz":-360},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7be83e6bc041408dad5f4f1afb43e194b546e68b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7be83e6bc041408dad5f4f1afb43e194b546e68b"}]},"branch":"refs/heads/master"},"adeaf4b8b14e0bc7a199ddf96bf79bf1911c340e":{"kind":"REWORK","_number":3,"created":"2022-02-25 23:28:22.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/3"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-25 23:27:23.000000000","tz":-360},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adeaf4b8b14e0bc7a199ddf96bf79bf1911c340e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adeaf4b8b14e0bc7a199ddf96bf79bf1911c340e"}]},"branch":"refs/heads/master"},"094aef564bf21d149929f65e3a7a0d593315d3e4":{"kind":"REWORK","_number":4,"created":"2022-02-28 22:19:52.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/4"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-28 22:17:28.000000000","tz":-360},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/094aef564bf21d149929f65e3a7a0d593315d3e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/094aef564bf21d149929f65e3a7a0d593315d3e4"}]},"branch":"refs/heads/master"},"63ff5e164b7a1fd6d9ff1bd2d408d77aff0db911":{"kind":"REWORK","_number":5,"created":"2022-03-01 17:19:22.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/5"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-03-01 17:18:00.000000000","tz":-360},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63ff5e164b7a1fd6d9ff1bd2d408d77aff0db911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63ff5e164b7a1fd6d9ff1bd2d408d77aff0db911"}]},"branch":"refs/heads/master"},"152337fc8d0ef817e63223a341f49341fa4ce732":{"kind":"REWORK","_number":6,"created":"2022-03-01 23:00:23.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/6"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-03-01 22:59:19.000000000","tz":-360},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/152337fc8d0ef817e63223a341f49341fa4ce732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/152337fc8d0ef817e63223a341f49341fa4ce732"}]},"branch":"refs/heads/master"},"1cc7b62cd31dbc275703b27b61278a955f64e89e":{"kind":"REWORK","_number":7,"created":"2022-03-15 20:21:56.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/7"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-03-15 20:21:02.000000000","tz":-300},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cc7b62cd31dbc275703b27b61278a955f64e89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cc7b62cd31dbc275703b27b61278a955f64e89e"}]},"branch":"refs/heads/master"},"63799d316761d7576c47a84a440416298c7cb564":{"kind":"REWORK","_number":8,"created":"2022-03-16 01:00:05.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/8"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-03-16 00:58:59.000000000","tz":-300},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63799d316761d7576c47a84a440416298c7cb564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63799d316761d7576c47a84a440416298c7cb564"}]},"branch":"refs/heads/master"},"f0682d620080e1a4852ce438ba29371bcc06c089":{"kind":"REWORK","_number":9,"created":"2022-03-16 16:56:51.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/9"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-03-16 16:54:59.000000000","tz":-300},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0682d620080e1a4852ce438ba29371bcc06c089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0682d620080e1a4852ce438ba29371bcc06c089"}]},"branch":"refs/heads/master"},"48975e19720368d3916558c213642d24597edcbd":{"kind":"REWORK","_number":10,"created":"2022-03-16 20:30:45.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/10"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-03-16 20:30:01.000000000","tz":-300},"subject":"Add rpc notification for load balancers.","message":"Add rpc notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48975e19720368d3916558c213642d24597edcbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48975e19720368d3916558c213642d24597edcbd"}]},"branch":"refs/heads/master"},"108cbbade7737f1597a4d398ee4e885bc97ce33c":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2022-03-26 21:55:16.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/11"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-03-26 21:54:11.000000000","tz":-300},"subject":"Add notification for load balancers.","message":"Add notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/108cbbade7737f1597a4d398ee4e885bc97ce33c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/108cbbade7737f1597a4d398ee4e885bc97ce33c"}]},"branch":"refs/heads/master"},"47203dd466d4ad8e953a9159092d5936a03f202d":{"kind":"REWORK","_number":12,"created":"2022-04-22 15:38:23.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/12"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-04-22 15:36:36.000000000","tz":-300},"subject":"Add notification for load balancers.","message":"Add notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47203dd466d4ad8e953a9159092d5936a03f202d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47203dd466d4ad8e953a9159092d5936a03f202d"}]},"branch":"refs/heads/master"},"cf75ccfb6f343bdf3b0294920913cb4170a85e33":{"kind":"REWORK","_number":13,"created":"2022-04-22 19:30:22.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/13"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-04-22 19:29:28.000000000","tz":-300},"subject":"Add notification for load balancers.","message":"Add notification for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows. These task modules can then be used in provider drivers to support notifications in those as well.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf75ccfb6f343bdf3b0294920913cb4170a85e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf75ccfb6f343bdf3b0294920913cb4170a85e33"}]},"branch":"refs/heads/master"},"bcb512fe57280b98f9a4422b0209699a88de23ab":{"kind":"REWORK","_number":14,"created":"2022-04-23 00:11:13.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/14"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-04-23 00:10:20.000000000","tz":-300},"subject":"Add event notifications for load balancers.","message":"Add event notifications for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcb512fe57280b98f9a4422b0209699a88de23ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcb512fe57280b98f9a4422b0209699a88de23ab"}]},"branch":"refs/heads/master"},"2195ca4a0759cbf58f1429c94243591cb85b17cc":{"kind":"REWORK","_number":15,"created":"2022-04-23 04:01:42.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/15"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-04-23 04:01:03.000000000","tz":-300},"subject":"Add event notifications for load balancers.","message":"Add event notifications for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2195ca4a0759cbf58f1429c94243591cb85b17cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2195ca4a0759cbf58f1429c94243591cb85b17cc"}]},"branch":"refs/heads/master"},"4540652e5bbb8ae1ac14cbc7b06b4e6d4c24c2eb":{"kind":"REWORK","_number":16,"created":"2022-06-07 20:14:56.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/16"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-06-07 20:13:19.000000000","tz":-300},"subject":"Add event notifications for load balancers.","message":"Add event notifications for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4540652e5bbb8ae1ac14cbc7b06b4e6d4c24c2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4540652e5bbb8ae1ac14cbc7b06b4e6d4c24c2eb"}]},"branch":"refs/heads/master"},"373e97750c853d8ba02994ac0fc5bb32ec6f6a83":{"kind":"REWORK","_number":17,"created":"2022-06-08 16:19:02.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/17"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-06-08 16:18:32.000000000","tz":-300},"subject":"Add event notifications for load balancers.","message":"Add event notifications for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/373e97750c853d8ba02994ac0fc5bb32ec6f6a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/373e97750c853d8ba02994ac0fc5bb32ec6f6a83"}]},"branch":"refs/heads/master"},"a0c1390ad805886663334044c1781450fbac0483":{"kind":"REWORK","_number":18,"created":"2022-07-06 16:22:43.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/18"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-07-06 16:17:31.000000000","tz":-300},"subject":"Add event notifications for load balancers.","message":"Add event notifications for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0c1390ad805886663334044c1781450fbac0483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0c1390ad805886663334044c1781450fbac0483"}]},"branch":"refs/heads/master"},"58f0f258c9e41a56298519fcf8a1dcf02df69800":{"kind":"REWORK","_number":19,"created":"2022-07-28 18:18:46.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/19"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-07-28 18:17:16.000000000","tz":-300},"subject":"Add event notifications for load balancers.","message":"Add event notifications for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58f0f258c9e41a56298519fcf8a1dcf02df69800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58f0f258c9e41a56298519fcf8a1dcf02df69800"}]},"branch":"refs/heads/master"},"70257eb6b36eebcbd38650e694d7f951a8931cb9":{"kind":"REWORK","_number":20,"created":"2022-07-29 15:07:38.000000000","uploader":{"_account_id":34591,"name":"Spencer Harmon","email":"spharmon@blizzard.com","username":"spencerharmon"},"ref":"refs/changes/51/831051/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/831051/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/831051/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/831051/20"}}},"commit":{"parents":[{"commit":"92038b24b6480fe11ea80fc96c2d5f01c964763a","subject":"Merge \"Optimize DB object to provider dict conversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92038b24b6480fe11ea80fc96c2d5f01c964763a"}]}],"author":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-02-10 20:47:51.000000000","tz":-360},"committer":{"name":"Spencer Harmon","email":"spharmon@blizzard.com","date":"2022-07-29 15:07:02.000000000","tz":-300},"subject":"Add event notifications for load balancers.","message":"Add event notifications for load balancers.\n\nThis patch creates tasks for load balancer notifications and adds them to the amphora loadbalancer create/delete/update flows.\n\nChange-Id: I287d89cd83e91473f1375788c969521aa58ca567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70257eb6b36eebcbd38650e694d7f951a8931cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70257eb6b36eebcbd38650e694d7f951a8931cb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
