)]}'
{"id":"openstack%2Foctavia~831087","triplet_id":"openstack%2Foctavia~master~I403398aa00b0c39804f1a1699b11f7da7fc33cd9","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-08 13:22:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I403398aa00b0c39804f1a1699b11f7da7fc33cd9","subject":"Fix a typo in a prometheus help string","status":"MERGED","created":"2022-02-27 01:14:08.000000000","updated":"2022-07-08 13:25:00.000000000","submitted":"2022-07-08 13:22:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831087","meta_rev_id":"fed3d90f080d9fa859c0c6b7b73943abbb5b05a2","_number":831087,"virtual_id_number":831087,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-08 13:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-03-02 11:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-08 10:52:42.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-08 10:52:42.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-27 03:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 09:44:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-03-02 11:30:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"a100c9f3de08865a6f9e40cd636ed80f0e890840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-27 01:14:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b08a83d53521b53c1d6fbdd5db62c92efaf505fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-27 03:55:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4c5af26ae2e4241b69664abf1cfdf2e : FAILURE in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cff1a4d6037141edb8faed8bed84c7a4 : FAILURE in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a54195d2fe6422e89fe8b32792c382c : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/544df7c74ab1450bbeb6c08a1642e4df : FAILURE in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e405c9eb7d40435fa84e0c32e968f907 : FAILURE in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c084ddac5fc6406aa89a5b9fbb2e208b : FAILURE in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440ca81d0b814d50b482d221bbc3d931 : SUCCESS in 10m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/97ba5f214cb546f7bee9b4ff81a3982f : FAILURE in 6m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f60c52ce2b947dcbc4a95f2dba832c0 : SUCCESS in 11m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d81927b76ccb4435b5101db009c3d8a1 : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18f109e2e2e64537ad53d8c0d7971ecf : SUCCESS in 13m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aa7a049e3a804785b9e61e6b1d1d91af : SUCCESS in 2h 38m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5fb0ca3d2ad448285ceab6e931857ed : SUCCESS in 2h 13m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8cd511d1988d41e8bca3a100c9ba1f59 : SUCCESS in 46m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d03f3216bd2947b7bd3c23d184a77a68 : SUCCESS in 1h 09m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/70b101dda0c645009a47a345a6b41547 : SUCCESS in 1h 13m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bacd0463fa684b90888b154f046ad75d : SUCCESS in 34m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/04baf6d186b14f90a7d13a4bcf21390e : SUCCESS in 2h 20m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e0d6bad823a4135a818ac5aa3d3eabf : FAILURE in 2h 04m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/88d08a00f1034136a29dc3a6c70b74c8 : FAILURE in 23m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/07768f27426146439d17a78d4da249de : SUCCESS in 2h 17m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6437c5fdadc2411e88c7ddd248f359bf : SUCCESS in 1h 49m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b57ba8d03f648f6a27a03a9b863c421 : SUCCESS in 2h 06m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3049e26f3481ac5340f1d03c6bfc500af00c5eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-01 22:03:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"faa9b23d01b600d4e049ef55c82135fcb92088f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 00:28:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c45ed45a251548a4ad19504b33eb88e0 : SUCCESS in 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ede24dd51a234aeca0259ef75ed93455 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3a7961469646f1b687791ce8ef21c3 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d17b501a1da7483baf654f069f2527e1 : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af4ed5ec40a749bdb208e128e9e4d915 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f75eb1ca43314ae28546df4291062204 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53e88ac0fa34528b00fd162ed61ead6 : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2e5e6b359ceb4f11b3bbade45da172f4 : SUCCESS in 8m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/addd19bb96b544dcb1abf6326fe508a9 : SUCCESS in 10m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23f21f9aa79c4056a5def7d9f6795d32 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b89e8ff9ee14012bdee4fe3ed2db9b7 : SUCCESS in 11m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6f950f41a734450bd82743a7a6dcb74 : SUCCESS in 2h 04m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6612a23267d487a97a6a28155e21d13 : SUCCESS in 1h 40m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b50e62355f904fd3a3fa5e2231e65c0d : SUCCESS in 37m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4f7437d025964bcfa62eec27735ddabb : SUCCESS in 46m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/edb8624cfe3e4fac807b7c06a6e46838 : SUCCESS in 1h 01m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c1335058b7245dcb9609e6d61f48acb : SUCCESS in 32m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2cf3e04e78464176881202a5cb19042f : SUCCESS in 1h 36m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac21672f9dfd4a4da019bf26acc182d7 : SUCCESS in 2h 16m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3762318d508f4086b899e0887a6df6e3 : SUCCESS in 2h 05m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df283b9187b24d07aa018e33cb25cb90 : SUCCESS in 1h 55m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b5485822b2984c4aaa95d3e6e93c5095 : SUCCESS in 2h 23m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/257f29bf0eee4960bce6b4bfc714a197 : SUCCESS in 2h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e138721b58fbab028462a522057fa384add8530","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-02 09:44:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"605cb57919ece4fc3abaec21f6d3657d88642240","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-02 11:30:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df70dbc500ad03da421a402edc1162e97c168601","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-08 10:52:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11639deadc16f6e9336ff168988fdb1a5e7bf41a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 10:53:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a98b659873f38baa36521d46774ff07e6de00ab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:22:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b5cc0bdfe0403491ab5cd34e2169b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95916cd633d24c8792d1d849f5b9aca9 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e1f46f88b94ce2a3df3ce607148e3a : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2917be5242964acd8de4d40b4c8ed229 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2332926dc9c54982bb108c79eb59dde5 : SUCCESS in 11m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/510067bb8ed8453fa5187e00895f11d3 : SUCCESS in 5m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1465b220e58f4305b4fb85ed655cb4ec : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a7612bf265b4f2281b4f6ebcdd17d92 : SUCCESS in 2h 18m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4eb59e15464649ba93ceacfad588a22e : SUCCESS in 1h 20m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b58d41674a6f40a780aaa415f48ce4dc : SUCCESS in 36m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f39ee96b9bad48ca81ed7bd364ec3243 : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":2},{"id":"651fc93493788f877cfe5c643af86f08dea3b561","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:22:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fed3d90f080d9fa859c0c6b7b73943abbb5b05a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:25:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbea750dbd6f4569929bd7ff921a26ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b990d83ced6a4436911ffac420267fee : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"41e3225c8867c412f72d82c940ea80d2c12e5148","revisions":{"02054a394b5f573cfade42d55830079be10548e6":{"kind":"REWORK","_number":1,"created":"2022-02-27 01:14:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/831087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/831087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/831087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/831087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/831087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/831087/1"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-27 01:12:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-27 01:14:06.000000000","tz":0},"subject":"Fix a typo in a prometheus help string","message":"Fix a typo in a prometheus help string\n\nThis patch fixes a typo in the prometheus help string for pool member abort\nmetrics.\nPreviously the help string had \"octavia_pool_server_aborts_total\" where\nit should be \"octavia_pool_member_aborts_total\".\n\nChange-Id: I403398aa00b0c39804f1a1699b11f7da7fc33cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02054a394b5f573cfade42d55830079be10548e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02054a394b5f573cfade42d55830079be10548e6"}]},"branch":"refs/heads/master"},"41e3225c8867c412f72d82c940ea80d2c12e5148":{"kind":"REWORK","_number":2,"created":"2022-03-01 22:03:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/831087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/831087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/831087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/831087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/831087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/831087/2"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-27 01:12:48.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-03-01 22:03:04.000000000","tz":0},"subject":"Fix a typo in a prometheus help string","message":"Fix a typo in a prometheus help string\n\nThis patch fixes a typo in the prometheus help string for pool member abort\nmetrics.\nPreviously the help string had \"octavia_pool_server_aborts_total\" where\nit should be \"octavia_pool_member_aborts_total\".\n\nChange-Id: I403398aa00b0c39804f1a1699b11f7da7fc33cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41e3225c8867c412f72d82c940ea80d2c12e5148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41e3225c8867c412f72d82c940ea80d2c12e5148"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
