)]}'
{"id":"openstack%2Foctavia~831233","triplet_id":"openstack%2Foctavia~master~I6d4de38dd44adecc677c637a25233494065ba3fb","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-04-08 13:37:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d4de38dd44adecc677c637a25233494065ba3fb","subject":"Add a Grafana dashboard for Octavia load balancers","status":"MERGED","created":"2022-02-28 17:20:40.000000000","updated":"2022-04-08 13:39:10.000000000","submitted":"2022-04-08 13:37:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831233","meta_rev_id":"c24022337ebf7a0ec7f71828a2519a5683c708ee","_number":831233,"virtual_id_number":831233,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 13:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-04-08 12:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-03-31 15:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-04-08 12:22:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-28 19:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 15:13:40.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-04-08 12:22:41.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"e637d715d1aa3b18693abb720ab199e8dde61b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-28 17:20:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d92b8ea9b92ae5ce3570ce4cb1c9a7be42024a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-28 17:27:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41a266d024b203b0b762cdb141f7459adbfb7aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 19:40:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0991a575ea1443385add8273a80f6a0 : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e544ad3cfe9b4e4a842ee469f265bbab : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c52e03a2fdd4c90aeadab7924b4d907 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f16ad6d8b054e62a835029305b3a6ff : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f2b4f51dac8478685a6a089935950bb : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b02784c89d4c358b90b473fd269670 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08bcdf2747845a798d5d660f3fc061d : SUCCESS in 13m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dbae946c4bed4ebd9785cb7bbbef9c7a : SUCCESS in 7m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/18621ff6c1d34def93ec090511a9de69 : SUCCESS in 12m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a513d179bff64667a0094055152c3687 : SUCCESS in 2h 07m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bcbb88614644fc3be5b7b855bd2a8de : SUCCESS in 1h 26m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/506d09a1aeec48029130934f13a0a43e : SUCCESS in 37m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/62854b2c9e7440009358e9a27d0ce9bd : SUCCESS in 45m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5ad84051234c4d96a7c5121b55278842 : SUCCESS in 58m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b8b974db38244cdb9e2256e7f7e0b8d : SUCCESS in 32m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b1bbc63bb92e4db1b8b6e37e58e3faab : SUCCESS in 1h 30m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f73d2eaf45604eca90dd3e370607ba26 : SUCCESS in 1h 55m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82cd97b2ceca473aaadfe7cb5c8efca3 : SUCCESS in 2h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c93b519eb24d4c2b8e0af535c9bf9309 : SUCCESS in 1h 35m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5a9cab9d9c704f70aa4ba622f49c2ac2 : SUCCESS in 1h 25m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee2ed5e54d8d49eb856ea238a96c471d : SUCCESS in 2h 05m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6bc210120fdd62af1b2201e0d0f48d402140a13","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-31 15:13:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd163a2567ee5db5093fb52d32eae48f728a25d3","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-04-08 12:22:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64493fbd8fff88ea9ad4aece238782a66c121980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:23:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eebd1de07e7d5d8c0c23ec921ac496df1aefc8da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:37:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6410a713b9941cc94be90bca71a0e2b : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373d6b7d1a7042abbcbd115f2d08deae : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb02d833167540b0a04f54e3c0d66fc4 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed208b294b2548d7b3fdd30d49f02b49 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0499a0636a4846a52c5c1a18893e15 : SUCCESS in 13m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b434974b0a4c49dbaa35aee3d577bd0e : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fbdd6ded1912beda99c1d3bf0a57a960db8cd71f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:37:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c24022337ebf7a0ec7f71828a2519a5683c708ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:39:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96aac45345fa4f1db21eebe8f98edcca : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f97c826fed7d6d09492ed40564c0c9b23fc1c8fb","revisions":{"2e2f43b014751f256e5e50d951a39abc1822a855":{"kind":"REWORK","_number":1,"created":"2022-02-28 17:20:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/831233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/831233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/831233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/831233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/831233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/831233/1"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-28 01:46:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-28 17:20:38.000000000","tz":0},"subject":"Add a Grafana dashboard for Octavia load balancers","message":"Add a Grafana dashboard for Octavia load balancers\n\nThis patch adds a Grafana dashboard for Octavia load balancers that can\nbe imported into Grafana.\n\nChange-Id: I6d4de38dd44adecc677c637a25233494065ba3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e2f43b014751f256e5e50d951a39abc1822a855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e2f43b014751f256e5e50d951a39abc1822a855"}]},"branch":"refs/heads/master"},"f97c826fed7d6d09492ed40564c0c9b23fc1c8fb":{"kind":"REWORK","_number":2,"created":"2022-02-28 17:27:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/33/831233/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/831233/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/831233/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/831233/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/831233/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/831233/2"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-28 01:46:01.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-02-28 17:27:20.000000000","tz":0},"subject":"Add a Grafana dashboard for Octavia load balancers","message":"Add a Grafana dashboard for Octavia load balancers\n\nThis patch adds a Grafana dashboard for Octavia load balancers that can\nbe imported into Grafana.\n\nChange-Id: I6d4de38dd44adecc677c637a25233494065ba3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f97c826fed7d6d09492ed40564c0c9b23fc1c8fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f97c826fed7d6d09492ed40564c0c9b23fc1c8fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
