)]}'
{"id":"openstack%2Foctavia~831840","triplet_id":"openstack%2Foctavia~master~Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745","project":"openstack/octavia","branch":"master","topic":"fips-compatibility-cs9","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-07-22 18:20:19.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-07-22 18:20:19.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-22 18:20:19.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-07-22 16:06:16.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745","subject":"Change FIPS jobs to centos-9-stream","status":"MERGED","created":"2022-03-03 18:49:31.000000000","updated":"2022-07-22 18:23:02.000000000","submitted":"2022-07-22 18:20:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831840-fips-compatibility-cs9","meta_rev_id":"891d0bdb0f838e0de7386e480c73556785801c82","_number":831840,"virtual_id_number":831840,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-22 18:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2022-07-22 14:30:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-07-22 16:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-07-22 06:27:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2022-07-22 10:39:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-07-22 16:06:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2022-07-22 14:30:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-03 22:01:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 15:01:07.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-03-28 08:53:00.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-04-20 12:22:12.000000000","updated_by":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"reviewer":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"state":"CC"},{"updated":"2022-07-06 10:25:46.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-07-22 10:39:30.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-07-22 14:30:00.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"1f8600dd5f4a357ee499eb77d138146090bea8b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-03-03 18:49:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaab0b47330ab1fc770d0f105a8284469e82df4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-03-03 18:50:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5a2b08ace58f0a549ead263cf9a5e17a5a32ddbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 22:01:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7ea95382bb84f93a3c4aea4092aa155 : SUCCESS in 11m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e26d5d2d5684c21970eb22d399885b2 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc7c838e91741f9beeba1df3e868d4d : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99fcc9ee156d4b8f84e871526b0d6684 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74469811136540ce826aeaf1f16aaa45 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fe384131c4349e1825ca20e800fa61e : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c714507fbc54361aa0906b216697565 : SUCCESS in 12m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b82cf96c9e46460bb5c0b92868a76683 : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7459d83ad2ee461da41249e51649e3b6 : SUCCESS in 10m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/642f85817fb34247a6410fdc7b68f175 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/127c4de5d4e746c4b8013af95c2707f6 : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0dbc1883c334d6e9ea81a927a2ee354 : SUCCESS in 1h 56m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11bd9fd555874d2b9206de51f5639888 : SUCCESS in 1h 38m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45854526120d49578ea9f3ee8bb54ec6 : SUCCESS in 36m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/415afaa2ea6a4cb4a3242def6e9f513c : SUCCESS in 47m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2462f651ff3d4aa7a04640a7a877d720 : FAILURE in 55m 48s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/90ff9caf8f464c8188e3a20b14c69acb : TIMED_OUT in 3h 06m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79b9b05ce5d94d35a5eda1cea8b09764 : SUCCESS in 35m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aa8c644675554f1a8035e91570e4a834 : SUCCESS in 1h 40m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d12077777004185990f4bb1253db442 : SUCCESS in 1h 49m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b5091b0d2fa43ac9a1325d3bb80b7c1 : SUCCESS in 1h 31m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/29104aad591e4ec199457b7f6f429312 : SUCCESS in 1h 32m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cec3abb3cd894d2ba8528642446732f1 : SUCCESS in 1h 32m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec79833759b444568e51ee0c46da3e3c : SUCCESS in 2h 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99befc4360ab78da85d798a3f1d12a504a92f154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-18 11:03:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"130ddec9fa2906ce7734099dcfc7478ee75741e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 14:15:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c6ec3b68ee74e9cade6dd044778352d : SUCCESS in 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b96bffb74ac942c594a7f203355c6556 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb171f5241cb42df92a53900de589c49 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c25909c4039428aa13d9f6682e85172 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9833df7a4c6e48bcaefad24b8135e80b : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dd1a254863142bd9fd390101eab4018 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e420e0a97b7f4bd1a4b7bd6d572fca1b : SUCCESS in 13m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c186444a87e54719ad1842c6dc696138 : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2600d20d1c954878a228ee7dd946784a : SUCCESS in 12m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ed1c9e951124b5da32142e341204844 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac5bd64fe8c64cdeb7678aaa54019857 : SUCCESS in 10m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fa12cdb5466a4b448ab3d99b9a63a4d6 : SUCCESS in 2h 22m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0606f475d0e473b8271fb4d4782cfca : SUCCESS in 2h 00m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6fe5e0779c274a69b480dc613888227e : SUCCESS in 43m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9b2e36462784954a7fa7939e6616eab : SUCCESS in 1h 06m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b5841f18ba554422a3bdecd8d6b17404 : FAILURE in 1h 09m 29s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/5827a74c17464d31957d96ba2185e1ba : TIMED_OUT in 3h 08m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/878ce6b127504763b0ef9c932edc852e : SUCCESS in 1h 03m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6a624fcd39d5426e93c2a211f7b8cd4b : SUCCESS in 1h 39m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/832998a67b1642039ad027a892917d0d : SUCCESS in 1h 54m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f53d29b3c74040e59cfbee8dfb59c0cb : SUCCESS in 1h 28m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/da28cf84f42e4ad1ba1edc038ea2da59 : SUCCESS in 1h 27m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a2a733b41e82417382db5978cabaa0ad : SUCCESS in 1h 30m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/278b0a5f86b14d34b2bb4c489b667fc4 : SUCCESS in 1h 57m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"82153cb4b7ba14f9d8193a3ab5480f31e8e77c9a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-28 08:53:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1323b0175b87cf1650c199e0c549cf20cc76be38","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-04-20 12:22:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6329a2c2eb19901202a529af82092e8e41f3f62e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:37:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44fa228a552a485c90821642ac466303 : SUCCESS in 13m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7cc7a6bc6cc426da04f90c1a85473ee : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43015be377a1445eab1717536de653de : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fedddd8ae754e01ac852c50628b7314 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df86849d7d81472cb7912fa207d61661 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a2efaec809844cb8fda205a550840fc : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae6deb8bfc141b1bf8f65541b28b866 : SUCCESS in 10m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2cdc423269b5402f9f0eb57f57a742db : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c2fd50c2b08240499ee4dee4253bece2 : SUCCESS in 10m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/12152e8ccafd4961ab5171297f34b087 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2ee577f2b5d4ee4b6f3f67e19039450 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/820abd76a0fe41ce975ff2dcaa32cbb0 : SUCCESS in 1h 40m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fe633b8245c4846a13eb043d9f1eea8 : FAILURE in 1h 59m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ae41e8b6af74b4bbc626d7c3e1e5299 : SUCCESS in 39m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3949a624141d475f8fb47be9dc5f8060 : SUCCESS in 1h 05m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0a3206381d4a411ea5284d5fac030688 : FAILURE in 59m 14s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/9affe95859cf468da93afeae9e575079 : TIMED_OUT in 3h 05m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15dd793666b9466493c27165c8fd7b1f : SUCCESS in 31m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a656ccae4cfe45dcbf4dbd4879e34884 : SUCCESS in 1h 39m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0760eaccad734849af111c86fd6ebeb8 : SUCCESS in 1h 44m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3753e82b50a4f058d812bc56d2a5739 : SUCCESS in 2h 00m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a9066634ed224e6580ca6226ca418fb7 : SUCCESS in 1h 50m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/189cbc4570484a18a2fed35360e3ee1a : SUCCESS in 1h 42m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cdfeeaeb84f4cb0a7248bb088515326 : FAILURE in 10m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3fa6a8e7445ca5038f34f69be43af2d8dc9efbe4","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-05-10 12:26:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b9d96465f315ed75fdb766ea970bba60f3ced3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 15:40:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77a54d71ad224550b6d64c645ca78fc9 : SUCCESS in 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8cf0cec5cd45a2a97af55025196a19 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06adc59024504a94aac9b9ac5705f344 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/516f0b851616470f94ba55cd2b03591b : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85a1a221868c45b795f7b894a902b62c : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cfc441046e646289f4da2bd516c09da : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/223837ffe0f04736af5f4367c33cea0a : SUCCESS in 11m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/526f025b028a45b9a657ebbb88459386 : SUCCESS in 6m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b013483bbbe742cba1fec922b30de0b6 : SUCCESS in 11m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/adc981fa9902491181f4bf6e0008ad64 : TIMED_OUT in 31m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ca3c89e0cf049e6b111978fadcf339e : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d83e6798c3ac418d8f76480092b7897f : SUCCESS in 10m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8dff81aec8db4766a8ac184a9852860c : SUCCESS in 9m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8542860c1a9e4111a7f3b0833d94ba8a : SUCCESS in 2h 09m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2de9e092a19541979d68bc307803784f : SUCCESS in 1h 45m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d697a28a580466e915c3de536d3ca10 : SUCCESS in 46m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2573a540ee1f4d3da901a06f8f3154a3 : SUCCESS in 52m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e00aacfc60304f369c51a7c1b0199840 : FAILURE in 13m 13s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/e82d6e1a90e946679f201421711d0392 : TIMED_OUT in 3h 09m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/121778c22fc9402ebc3f8d4201c397fa : SUCCESS in 43m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d22488636e754cedb5c7f7ab680f470e : SUCCESS in 1h 55m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b715ebf95c14419a6da3006ddfcbe4a : SUCCESS in 2h 26m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3cc311ef1c7c4a60910886c9dcf912fc : SUCCESS in 2h 02m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9aa609e6499c4411852fc1ef2e219ba8 : SUCCESS in 1h 28m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bcc7d80d2561486ea46586cb1acb7c82 : SUCCESS in 1h 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/669330b98fd24b4082df2d92a43baea6 : FAILURE in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e702cb6cbed3be150b5683d9d571b4f3f6f5172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-05-23 20:44:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3981accbbd6b1edd55e21d6c8799db10dc0fe0a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 23:56:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0be57af7858f45718d8fdbc2857002ff : SUCCESS in 11m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26db37621f31457c8189aae531262cae : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6939ab3916374778ad1142e765c254ee : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a4e2f35d14c4a20955efb5809cd508c : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f93303677b6c46738b0ad0a07aa2ec24 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3dae9ab6bf804cee9254a24ff7ef5730 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eacf7349d8974a12b08d44571f6deee8 : SUCCESS in 11m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cb4dafeffd2746f7a4ed64a63a7bec59 : SUCCESS in 6m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8373504676d14c4f83691478e97c1c36 : SUCCESS in 8m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/03a7127caaaf4e3f9124dafe9fcac574 : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ef7b073cacf4404bf0d9d9540a8ab6f : SUCCESS in 8m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b58223f2ffe4a998dc56184ebcfc713 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/32af2ec1294b43bf99e74bde89ab6b64 : SUCCESS in 11m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5706898fc39e45f6bc05b2e60b09ebff : SUCCESS in 2h 09m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f220a1e1a85b4d4b93ec7025b1874b20 : SUCCESS in 2h 14m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ff08209278042eaab42a8bdd3f5d9e9 : SUCCESS in 45m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3d767e3036dc48d381e5bdc819b8fc43 : SUCCESS in 1h 18m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c641d5b210c6428c8c27a8500dcd58fc : FAILURE in 1h 06m 19s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/4ba445955d014d74be09e09534756bd3 : TIMED_OUT in 3h 08m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35efc2620b1a4727878c879c706dfc6f : SUCCESS in 44m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fd2608863a6e4f56bbd9a7bd819b6bcc : SUCCESS in 1h 44m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ae74cd067877418d80d23ae41bc2d2f3 : SUCCESS in 2h 41m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9000251499d247caa788b87042363fcf : SUCCESS in 2h 16m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5898c8703716476a9d3d46a7f374aa32 : SUCCESS in 1h 46m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/34fb69f0c7674e0aa6eb5701a2d3e859 : SUCCESS in 1h 22m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb52f7f3329f4728a6212570c2732be9 : FAILURE in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51111b07bc506193f046f861692183535cf29905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-04 17:53:33.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9966ad54933a7d529458183afae427b135f312d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 20:22:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73be962c51934dc89e6d791f086a9860 : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ef532ff0bf24187a9ad5960cd7f73d2 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ecc8a693e7a4bbb96bbc2bbb86e8d4e : FAILURE in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8edaa737f07c42dc86df44f7c4e8e9e8 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1fde7df49ee4df882a2d3b766236d0d : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02a2fef7fe804857a66da6c04f36f571 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4976c7586be0425389d37ef706584954 : SUCCESS in 13m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/31b707e961ea4b6f9fa5a7300a3ef7f0 : SUCCESS in 7m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/220384c3606a4b9f9b62a1d27792b2dc : SUCCESS in 8m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3518d298d95c47e1ba97faf9ef55f802 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/78d193700cbf4c929473546f89111d0c : SUCCESS in 8m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee7cb6040808435d8a7633ae8dd16bfd : SUCCESS in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b79711900fa43d4809a916bb75ce879 : SUCCESS in 7m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4261c7e75c54658a6d95da4f57e87ec : SUCCESS in 2h 02m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45000ea7a6564e79a70c4e20afc3d2c8 : SUCCESS in 1h 32m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44c1dee2cbb54c56855ba3b345735600 : SUCCESS in 42m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3ebdf8d3a6a4f3a86398da00a35038e : SUCCESS in 49m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f778955846954055826c669fc8e0b408 : FAILURE in 1h 13m 00s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/e610b6fcf2284aafab793bc58c74f2df : RETRY_LIMIT in 11m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d99942bd2044f888f049a3de558dbbf : SUCCESS in 34m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae357d09b4b64fbfbb82161911f57de2 : SUCCESS in 1h 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a8bdb69e4a1148fe8f79e84161428c47 : SUCCESS in 1h 57m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7f7b626caa7d429f8740c9f33e0acb8f : SUCCESS in 1h 56m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cf773b15195d4b999f5d94f28b8132fb : SUCCESS in 1h 41m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/84a03cabfd4f4742b40e5d66a7884fb1 : SUCCESS in 1h 14m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8bf325d93de4a3a8bf74a336a244f93 : FAILURE in 12m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":5},{"id":"8fa16e2071e63f91aee78f074447cd87c0881023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-04 22:00:25.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"acfad1118c05c7670754527c7a620e47d3f0ddf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-05 01:05:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecdf0014d41f47dfbe1f319130b03ab8 : SUCCESS in 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a34fc0c22574691acd35adde60dc57a : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a5624c3be1b412391b9d804f476fca5 : FAILURE in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a577156dea34de8bfa6b097be498cff : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ecd8c4503f243cc9d0b3f63cd479d71 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/849ec1c598ac4dd8b8ee8f9faf2bfdbe : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c43558e9244452b45a45cb6873e3cd : SUCCESS in 9m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a8dcaf3d999240fcaf5437fc57dd27a1 : SUCCESS in 6m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6f9edad2777c42e09babd763bbc52402 : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b4c8103be9514708ad43982bb19639a7 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aea08ae1f8194991ab6d36b23cc1b2aa : SUCCESS in 9m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/91543e7dd2b147c792d7ade7daf7cec6 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ac2ac52ee9244629d379d9401987a93 : SUCCESS in 6m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54203e58f21a4ee583ee7888a55b1118 : SUCCESS in 1h 57m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ede564e1ce7456480401c498c08b410 : SUCCESS in 1h 39m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ef973ed29304c3187084271a9f1c54e : SUCCESS in 33m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0587545a6d924197892d6b293518af25 : SUCCESS in 50m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/72fb765323d243cdaf8700810246d337 : FAILURE in 1h 05m 58s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/ccbd5d0a5bcb488d8530338db4520442 : SUCCESS in 2h 53m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e44694abf1b8450a921e89214d18e318 : SUCCESS in 31m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a32f5c878a0f4738817daa8f8dcbd39c : SUCCESS in 1h 23m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b40661792e864f4db6c95ca01404fca3 : SUCCESS in 1h 58m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9f42991846174f3c9e386b046f7d50fe : SUCCESS in 2h 10m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/551a8ccc6e22403095171be32f6d68f8 : SUCCESS in 1h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ae48aed4000546748275cfad8aa30d89 : SUCCESS in 1h 38m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7517aa5801224fe79bc0afd2f095190d : FAILURE in 9m 41s (non-voting)\n\nWarning:\n  Comments left for invalid file .pylintrc\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py","accounts_in_message":[],"_revision_number":6},{"id":"c8ddedd818f9a69111d03cad7b4ca7a9abaf0cb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-21 18:16:02.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"5d00d6d3aaec09c0bc0c92845c61a37d2585a0a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-21 18:16:09.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"f28f25b321f6f037dee5a340ce80f8cdfcd01e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 20:33:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b618489145654901a3387550326772ae : SUCCESS in 12m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c57a7a5a7e02412e96fe06aa1b6a93ab : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd905701d044caa96cc1dfe03f8e839 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e429b0216b1749b3a48bd26ea9dc7f8a : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a05ce5e40a845799d3ad6b115fee3ff : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/300a99dc48fb40aebcae74b4dd987eb3 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a87c9ea9284c839393735d7b41a786 : SUCCESS in 13m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/541ec0c985ff491eb5b888b8e72b756a : SUCCESS in 6m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7a4524ca3e0946ad84c179f27f52cb5b : SUCCESS in 8m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c92407159efe44e487b4ed47ed6b9085 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee72b0c87d0643c2a6244b9b436bb232 : SUCCESS in 8m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1aef905b8a2a4272b08bd3f497a17666 : SUCCESS in 10m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5250bea2322f4e3cab11570771e293af : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9dee63866924e00bceb2c57efced6c5 : SUCCESS in 2h 10m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62369a121f4b410b993a4336ff593db9 : SUCCESS in 1h 31m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3ebe30e7f1b49e584a860e681acb6ad : SUCCESS in 37m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce7cd49223a34a73af22c7361dd47db8 : SUCCESS in 52m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eef5039caa964910a938d6f60cf295f4 : FAILURE in 53m 58s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/c643662bef5344cc85962eecd5518639 : FAILURE in 33m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf443a29a1e145db916e54eb2d6a4712 : SUCCESS in 31m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d0546dcbb0a542d69fcd7d20997c76f3 : SUCCESS in 1h 39m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9eb0874842d8401abd3ba8ca75ee9c40 : SUCCESS in 2h 03m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f03a4d224a4d4e8a8e081f7fedfa8697 : SUCCESS in 1h 31m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c66394c50e1445985c83baf0b49cb80 : SUCCESS in 1h 34m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/680d7d4c17844f9f8b8b90559291d84b : SUCCESS in 1h 49m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8d56c6dbab94d3ebf1e3d159148a40c : FAILURE in 9m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a33f2d206bf4676e73f91d2272489acd683b92b6","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-06-22 21:09:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ba7103a0efb98b90425fbc7518f5526843f45d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:13:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa38e873a0484f0bbd04a46c64985f6c : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8595f0e010374c3eaba01699b50780f5 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f3b91064e2438cbfa2372aede73934 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/361675f5703849718966be2622826888 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f587acfe43e94551bc86b867cd684e40 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0ba1f8023d84a42aa0abe799c11ff6f : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e63f21c88d4621b0f9711fa72a4b6a : SUCCESS in 11m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1d2618f39d714f0c9b7169f662b73c25 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/26d5f30bac0042f7bbdcf3517396861a : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/89c2c9379cf145f88c72bb1aaf19f9c5 : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c6795ab43b8b4acf9f44adc2923f4e99 : SUCCESS in 9m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35c40165d1cf4712a2a5f37c3a9227b4 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/50401402d7bc4980b705fdcb9e96f701 : SUCCESS in 6m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c528426df8764be091d0a49264629393 : SUCCESS in 1h 41m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59904696cc0d4188ad498ad2327273cd : SUCCESS in 1h 35m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59c51705a0564f13abeeb421fcda0289 : SUCCESS in 35m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cd1b65794046454a847a049c042771c8 : SUCCESS in 1h 04m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d935a191c83a4d3eb26b89947beb94b6 : FAILURE in 54m 35s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/304e2e6869dd4791a39dcd54b5cda53a : FAILURE in 1h 07m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79d4d0e804d64bf4826ae8f1f1133bdb : SUCCESS in 33m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e89f82e7c6a34c4b938271e05c52265e : SUCCESS in 1h 53m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59adbe58a72943b19121114844968642 : SUCCESS in 1h 47m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8ca5b16254764088b99b5112104e2943 : SUCCESS in 1h 43m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c54b0d29c5be4f899e264ac2cb2758b3 : SUCCESS in 1h 22m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c0cb15aa1752490d9c1d074e6490117a : SUCCESS in 1h 40m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/d80df8c60d8f4b28afe560515841fa47 : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"529834be2a96a9ff12c7b9260cdd31c2ca336f5e","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-07-01 18:44:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5b277c5bc3e4e5f74c8ded8cda931ae8aaf54e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 22:03:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c67d9bdcaf447588f489f1e2fe00f17 : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d6100281d62486ab622e0961bea0227 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b725a9b39cd4793be0840b441ebdd64 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92d3b79123c847cca3da611e196c556a : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb14a1259baa46c4bdb6835dd7dea9e8 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3aab6e9ebbf4264b8c20f47aea018df : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42636b29c1f43e9aff5c9ed92caaca2 : SUCCESS in 11m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8af74af159a040f29dded2b99b681e8e : SUCCESS in 5m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3710b09ee05f403ca6354004c073f05f : SUCCESS in 8m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/24c5d0ad72ab4c8195dbf80f619e44f8 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/989042e9197d4c7fb99b42737e53c834 : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c2b1c6a2df3435895afd66cc9a6fee7 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87f66d0a06724df783ccc6c10f3f6bb5 : SUCCESS in 7m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6974e85efd3c40aaa4b62cba5caee9a6 : SUCCESS in 1h 53m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38b5288f88aa40088f7423b775ce3dba : SUCCESS in 1h 55m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/40d1b90f63f349908cd6fb89865f93b3 : SUCCESS in 38m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e66c4ec84a74139a0f60fed7deebd9e : SUCCESS in 1h 17m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f506e5ad6d5b4ad1afd9c8bfe0fe4b01 : FAILURE in 1h 04m 23s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/a3f49a67543d4bd4ab8eebd13f70fc13 : TIMED_OUT in 3h 08m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3686357f5a7b4a44b57db06f74a86a4e : SUCCESS in 38m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a4d3884c6424388bd5e7100cdbcbdb9 : SUCCESS in 2h 03m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30162a6a66144dc6a4cc6304fa6a4e65 : SUCCESS in 2h 13m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3731c3e2d00745108f014daf06534291 : FAILURE in 25m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64302f4dbb684b39a7c5de6e8f29377d : SUCCESS in 2h 25m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4c90b84414684d7eb6b93153b9d0619f : SUCCESS in 1h 40m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/99452822db51402697a0827ad932d4f2 : FAILURE in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6859b7fd0a9af300a0a324e97116ddcb938ac935","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-06 10:25:46.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3602b791a86abd60979ff21dd044adf7c35e1ac6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-06 10:26:04.000000000","message":"Patch Set 8: -Code-Review","accounts_in_message":[],"_revision_number":8},{"id":"731184cf1466b5891e73ee1cc943095e3acaf79b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-08 14:13:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b525669ad8d9f8d63383b28c3cde2cef5edde74b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 17:22:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72e63d8d41d42f6a2cd2d124ff96fa6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe9b66a199834462819dad15e5fe199b : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69592c350244a64a712cba5cfc8b83b : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/856775e3b9bb4d83badb2f1d4620c5d1 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e6564a693054735a818398792150195 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9f6a04e4ab54d93a62fed3e57ae7514 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661b3734849b45e7918eee08a77b49a4 : SUCCESS in 11m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d2b9b32310774ae48153c30e58240b3f : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b8af64a3f58b464981cda2790e770821 : SUCCESS in 6m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c253363f821487fbeeb0ab4c43afea6 : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/831e9d3ca23042f3a301ce58b8cd021e : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3669d197ec1d4ac4b80dcb8bd7c4565c : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b4e1795cb56c49e1a63f5ba27a4c0f56 : SUCCESS in 6m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9154531bba5845658239605570cbd351 : SUCCESS in 1h 50m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45c9d938fe6643b2a990330f12c0b382 : SUCCESS in 1h 21m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0628eda53628409a9a9b00e71442b515 : SUCCESS in 33m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/428c56043cb24547a491e5e6ce753d1c : SUCCESS in 52m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f022f7508ace408dba530c89b448bfa4 : FAILURE in 56m 04s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/ace8852fde3b46ac965a28a211fefc32 : TIMED_OUT in 3h 05m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92f8c20c54544e2d86244cfd5ded7158 : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ca396f92ce8a44e99b93c3c90f0e6374 : SUCCESS in 1h 23m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/425a8cd3c81b4a75a6e28c059561a692 : SUCCESS in 2h 09m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/601cd3da48f7403fa6e8ec41243f1413 : SUCCESS in 1h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1c2143a4bf874933a9fa390e7966e7a3 : SUCCESS in 1h 45m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/94ced12950c04ebbbc124c18dbb55aca : SUCCESS in 1h 20m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/23b66663c03e4a389d200760f36b8b94 : SUCCESS in 1h 48m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ae58197d51957ee111532488907354713e65a18d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 09:43:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7220677e710e5d9d8e2d72e593c215e608ee339d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:08:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86420dd57c614dceaadebe14b8a9b53f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9007f7daca5345e6bfeaff6af8651708 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ded59e0923d4865ab9945cdeae3301b : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d2e118413d447f8a407f14e73b7761d : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce8a64e3dfce4b2298a784aa0ed58304 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/affc1f7efdfd4eee83959f6c7d5dfb2a : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d540ca4167e4957a9d287ee85001a3b : SUCCESS in 12m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/edbd95578c6c4bda85151f4aa3773fd3 : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c323699f69974b17b2f07ebf4ab0aee2 : SUCCESS in 9m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/73dadfb8695f4cfdbf67e0255aca4270 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bff400ecbf8c410cb2d407ea52c036a4 : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2694046661df4029a53ec826076e3d87 : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/539a66bdaebd46d09ab97a0a708d2368 : SUCCESS in 8m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e42f6b4e7ce477e97ae7f4dd4aeac4d : SUCCESS in 2h 15m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d236cc309e484a8d9b29c0fa7d7bdea6 : SUCCESS in 2h 16m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b366c0020264c4e88e6399cf3295dc8 : SUCCESS in 57m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f672f671eba48579a4e87aa1d0ec6d3 : SUCCESS in 1h 30m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/aa4ebbbd7f624a10abd8e6af853e5408 : SUCCESS in 1h 27m 47s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/914121e81aba4f9e87699485fbe9967f : TIMED_OUT in 3h 23m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/846db99c0f9c4566bc818a2926983daf : SUCCESS in 39m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ffd6b4bab2a4b0793f9030b46f3bcbd : SUCCESS in 2h 20m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f1db6e56da14467bba4e5c197a3137d : SUCCESS in 2h 31m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2c5cd4fd5dff4e75b9e311397581b44f : SUCCESS in 2h 36m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4b814d76c6434391a2226883908cbac8 : SUCCESS in 2h 02m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fced688a69fb4551b58bccde005c75d4 : SUCCESS in 2h 12m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/46aba9ef59764da0bd1efd09050715ad : SUCCESS in 2h 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ee1ead4a06c8fe3f5af1d458c0af14124b9c95c6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-11 14:29:38.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"707038872360663997fd3afdfdd29b015e754c08","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 14:33:50.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f0e444e2320752180a342364c785c652318ebc6d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-11 14:36:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f27129e7ee959c6b430cae38a2a0e80a2c7c5d22","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-12 08:51:45.000000000","message":"Patch Set 8: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b7ec3095e6c622c1e2896fab81de0285ecb5a7c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-21 01:00:17.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"c86f0019e314b7c0df46a5aad173795691be6c53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 04:24:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05fbc45743964bbda842cdbe9be7d196\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23d3d0ff916041818ba42a07fc26ff6f : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5c1e429abe4c0faf7e1be91522dd80 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48be59904db048a79d0d0ac0c8043db8 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16f0224c8ab9493792c291e632418d0f : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d68d3330c484d858b6734217630a866 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24536f2e25ef490aa323ec99178810cb : SUCCESS in 12m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e17b3110fdc94080854df674fda32d29 : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bd243451ac3c48bf81eb1746d1ad64f0 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0d56eac7faa3440faa9f5aa173a72829 : SUCCESS in 5m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7d2490be3ec465fb2818333ef818187 : SUCCESS in 8m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/755d28464a25428bb5fdbb9cf2444c7f : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/411f122138fe4a978036a3b728999850 : SUCCESS in 10m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92dbe300de50468c864e0b55adb468d8 : SUCCESS in 2h 16m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c35dbb5fb5a54f1388b0ffee1509a3cb : SUCCESS in 1h 52m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bef46b0ae40f463bb0498b9bd8df9b27 : SUCCESS in 42m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c622f885bfe44c7bd96cc8db7809f5a : SUCCESS in 58m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67ae40be07d1479d90f4445591bcd48d : SUCCESS in 1h 14m 14s (non-voting)\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/0033db840d9a43d386e03e84e49524b4 : TIMED_OUT in 3h 14m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bb02090714f49e9b629a863dcb5b7b0 : SUCCESS in 33m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4f14c3a58d824d298f140d80484e32ab : SUCCESS in 2h 06m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d61d6d914edf4ee2a266d40c1f41499b : SUCCESS in 2h 00m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4bed9e3186744915b22601c4e4a7e872 : SUCCESS in 1h 56m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/92e3d44cff1d4c00978b0dff7ded21cf : SUCCESS in 1h 58m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2473d2c11cf74283bb9261996bdddacd : SUCCESS in 1h 39m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/12ba90dcfd544d3780eb6a9b50a5be5f : SUCCESS in 2h 08m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f0365a32de8c0e8f4a1308db7d00b33a61a91b03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-21 13:37:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9a2a149d3980c42d6d542b6b7a4bf99faf23b0e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:57:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89844344ab624ab39409c83ecea3869c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54a7ef83b4e14397855c1516eae106df : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9483b64fdb0e45a3810d3f9244fd33de : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee0eaf83d73d4fa0a19156123d487c87 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b78f65e305834cb48f58932c17f57fb8 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29868c7652ff40dbab691e8782de9f6a : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e57d6489cc44861832f4e23cf2fec9a : SUCCESS in 11m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/50b2a7c27e3a4cacbd16d557dcebe212 : SUCCESS in 6m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/420460a0d419470e9a88953fc4e45031 : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/36a3039904734b9694e46b55d2b84144 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ac0ee608c444f65892d29fd598f625e : SUCCESS in 8m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc02b7ccbf4b48aaa5780adac7c46ab1 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ed611efcc75446fb7c07bd87430ccf5 : SUCCESS in 8m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f333759de0bc413eae23ee80ee9b8129 : SUCCESS in 2h 05m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8345f292ab0b4c75a4f039b3ea410b37 : SUCCESS in 1h 45m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/085674b19640479a884cacb7a850df99 : SUCCESS in 37m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/23568453a41c43329ebba89ee449b77d : SUCCESS in 55m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f723d6e7655f4d16b68ee843f94cc3ab : SUCCESS in 1h 35m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c376ca0d1c0f4044b054a703585f2f96 : SUCCESS in 38m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96c52e10dc994fb3aebbc1a04ccf3eef : SUCCESS in 1h 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fd3bda9193454f4e8ff8caa1cec80029 : SUCCESS in 2h 06m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c783d367021a48c2aa4f3294bcba2fca : SUCCESS in 1h 49m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ca8f1c4515da48e4930568732364baf5 : SUCCESS in 1h 45m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ce6ed90664ba4fcd8108d65aecf40073 : SUCCESS in 1h 30m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/abf60dcef41642768b3ba9256b3e847a : SUCCESS in 2h 17m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"03e6fb41a2e699225004a7e8a1047af45855cde7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-22 06:27:44.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"e938e46f1ca7e301e8e2ead8f0d78ea6f4bdc620","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-07-22 10:39:30.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"9fce1d6eb9839105640a04f855450d73c7bc6528","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-07-22 14:30:00.000000000","message":"Patch Set 10: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"8055e927b1ffcea0821f455b6749c25aaa6cd816","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-07-22 16:06:16.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"22698df7f3ca132dcdfeb6f3a1a5fbc7ac9d7408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 16:07:01.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"89b538d9fb77da5c954996e8185d9183ab117ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 18:20:19.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a0586512ce45888380d53fb1819dd4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3a0f42f1cf49a983bd3513889d6762 : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a83e58e8976e4902b172c3a19f491dd2 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13e56b21783b443881139c51f70bf38b : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3d313b7bf9483695f4bb25f41b94f0 : SUCCESS in 15m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f78aa699327043c79d7a29ff352d100a : SUCCESS in 5m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c1661b3b741c44999d81fe6fd0b2dee3 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c185a2686a1845a787eac3b04a22c50b : SUCCESS in 2h 06m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5849553f74d40f591faa53dfcaa6c1e : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b85a244d43f42b7bb16bb0c34125674 : SUCCESS in 41m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cd750c52c2fd430c8403eeb17e6f23cb : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":10},{"id":"6ac2184f5d40939cdc874340410503015e47bc8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 18:20:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"891d0bdb0f838e0de7386e480c73556785801c82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 18:23:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba30cbea4a794541a2fdb6170e601fd6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a3b374368a454bb5ff3e7c2ad04002 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"6d7464303ee554de2a598502486574e2a9a4e3f7","revisions":{"1a105c4c542adf3dfb9e423370635f50cf07ab3c":{"kind":"REWORK","_number":1,"created":"2022-03-03 18:49:31.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/40/831840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/1"}}},"commit":{"parents":[{"commit":"328ffbab5a6737878b02edb058d2c91dcba76313","subject":"Merge \"Remove ureadahead package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/328ffbab5a6737878b02edb058d2c91dcba76313"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:49:24.000000000","tz":-300},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a105c4c542adf3dfb9e423370635f50cf07ab3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a105c4c542adf3dfb9e423370635f50cf07ab3c"}]},"branch":"refs/heads/master"},"81f570cc320100ab1a61002ae18c42dbc7c4b799":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-03 18:50:51.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/40/831840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/2"}}},"commit":{"parents":[{"commit":"328ffbab5a6737878b02edb058d2c91dcba76313","subject":"Merge \"Remove ureadahead package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/328ffbab5a6737878b02edb058d2c91dcba76313"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:50:31.000000000","tz":-300},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/831245\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81f570cc320100ab1a61002ae18c42dbc7c4b799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81f570cc320100ab1a61002ae18c42dbc7c4b799"}]},"branch":"refs/heads/master"},"142a66733a554210b07390a9b2fa57598488bd30":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-18 11:03:30.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/40/831840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/3"}}},"commit":{"parents":[{"commit":"328ffbab5a6737878b02edb058d2c91dcba76313","subject":"Merge \"Remove ureadahead package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/328ffbab5a6737878b02edb058d2c91dcba76313"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-03-18 11:03:30.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/142a66733a554210b07390a9b2fa57598488bd30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/142a66733a554210b07390a9b2fa57598488bd30"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4a1e358bc6cecf68a473b0b34db7f54c86cfbf3c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-23 20:44:58.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/40/831840/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/4"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-05-23 20:44:58.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a1e358bc6cecf68a473b0b34db7f54c86cfbf3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a1e358bc6cecf68a473b0b34db7f54c86cfbf3c"}]},"branch":"refs/heads/master","description":"Rebase"},"20df23e3b2258441e8401140625d408a16b2bcdb":{"kind":"REWORK","_number":5,"created":"2022-06-04 17:53:33.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/40/831840/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/5"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-04 17:53:17.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/844704\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20df23e3b2258441e8401140625d408a16b2bcdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20df23e3b2258441e8401140625d408a16b2bcdb"}]},"branch":"refs/heads/master"},"12ea7890349118a5f99a9c6b3699cdab79a1fac1":{"kind":"REWORK","_number":6,"created":"2022-06-04 22:00:25.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/40/831840/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/6"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-04 22:00:25.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/844704\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12ea7890349118a5f99a9c6b3699cdab79a1fac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12ea7890349118a5f99a9c6b3699cdab79a1fac1"}]},"branch":"refs/heads/master"},"648288b04b6d5434269851c6ee259eb7d0f66944":{"kind":"REWORK","_number":7,"created":"2022-06-21 18:16:02.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/40/831840/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/7"}}},"commit":{"parents":[{"commit":"6253560d22f255c499c91612ac4286dd0d8329e1","subject":"Merge \"Make amphora timezone configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6253560d22f255c499c91612ac4286dd0d8329e1"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-21 18:15:51.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/648288b04b6d5434269851c6ee259eb7d0f66944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/648288b04b6d5434269851c6ee259eb7d0f66944"}]},"branch":"refs/heads/master"},"11797389fd28a1375881d780956e1208b1a3729c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-06-21 18:16:09.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/40/831840/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/8"}}},"commit":{"parents":[{"commit":"2877bf77bc277dd500fa1d24345ef7648d924573","subject":"Merge \"Update default image versions for CentOS and RHEL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2877bf77bc277dd500fa1d24345ef7648d924573"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-06-21 18:16:09.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11797389fd28a1375881d780956e1208b1a3729c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11797389fd28a1375881d780956e1208b1a3729c"}]},"branch":"refs/heads/master","description":"Rebase"},"3de9e94952e6eddac0f891b2f9b9d3fc213b43d4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-07-21 01:00:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/40/831840/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/9"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-07-21 01:00:17.000000000","tz":0},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nMoved job back from periodic because centos-9 is supposed to be faster.\nLets see how it goes.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3de9e94952e6eddac0f891b2f9b9d3fc213b43d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3de9e94952e6eddac0f891b2f9b9d3fc213b43d4"}]},"branch":"refs/heads/master","description":"Rebase"},"6d7464303ee554de2a598502486574e2a9a4e3f7":{"kind":"REWORK","_number":10,"created":"2022-07-21 13:37:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/40/831840/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/831840/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/831840/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/831840/10"}}},"commit":{"parents":[{"commit":"3f4a44db94298433e64de4e811c9799de5722f8a","subject":"Merge \"Move system scoped secure-RBAC to separate file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4a44db94298433e64de4e811c9799de5722f8a"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-03-03 18:47:44.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-07-21 13:36:36.000000000","tz":120},"subject":"Change FIPS jobs to centos-9-stream","message":"Change FIPS jobs to centos-9-stream\n\nIt also fixed the missing nslookup_target variable for ansible.\n\nChange-Id: Icbbab19fabc90f4f4ea63a0ffdf75bb00144b745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d7464303ee554de2a598502486574e2a9a4e3f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d7464303ee554de2a598502486574e2a9a4e3f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[]}
