)]}'
{"id":"openstack%2Foctavia~832509","triplet_id":"openstack%2Foctavia~master~I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270","project":"openstack/octavia","branch":"master","topic":"docs","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-03-28 08:52:38.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-03-23 09:06:49.000000000","reason":"Gregory Thiemonge replied on the change"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-03-28 08:52:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270","subject":"Documentation updates","status":"MERGED","created":"2022-03-08 10:33:46.000000000","updated":"2022-03-28 08:54:35.000000000","submitted":"2022-03-28 08:52:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832509-docs","meta_rev_id":"cce5cdbff584c713c036974765120a21dd320fe3","_number":832509,"virtual_id_number":832509,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-28 08:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-03-28 08:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-03-23 09:47:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-03-23 09:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-03-28 08:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 10:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 17:32:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-03-23 07:54:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-03-23 09:47:41.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-03-28 08:31:18.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"43e94c2b97ae2757adc6d4d2ab9075091145c732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 10:33:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a19620a6efacd6f9f5c3d0cc36e84c882327491e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 10:33:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"1cb4b504229c0e33ce36894514b7a60a30afd0e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 10:48:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f6bbf649e94a1494c67097e7105fb7 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133d5b2eab754863b0a971c4d34519ce : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"aad8a47e3c27f5a432d3bdd9b5d2cd3702b99102","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 11:00:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"627b4db8e2b8529ff8bbd89e47294a106619d3dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 11:26:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae46ad721b54028bb2fa565b8c0111d : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e4e2b3defb4136942c630d6f4535ad : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6c6c142a012ee8bfa539f683037ad89a2dcdee1e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-10 17:32:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e962f1cc905e33a2b68e1623b7fa6dc35833abb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-11 09:24:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe4c6d3ddec3d8050a38215329703932fe55871f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-11 16:20:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b470c5d914a36e4aeeebafde8d53b85aeeb5b96","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-11 16:21:17.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"ae316f497888a2e9a384859abe118ed8991d11b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 16:37:21.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf0b54407de40289690232abde29a10 : FAILURE in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7c2be88a8149ea8f6a7d96351037b4 : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":3},{"id":"91a29e3c35f1e45e09ed24b2d0248e4f51f9c044","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-14 08:35:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f9a42054971eaf76f4576ea87b69a93e950ee48d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-14 08:38:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12eecc5170d4554897ae7fdab10b19e5ddfbd4c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 08:52:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f217216491004d55a48e2776f48b223e : FAILURE in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0565cfbd3eb43a9b4916eb5fff32df1 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":5},{"id":"99402ec563b5bae301eca44977c1223b5719e64d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-14 15:27:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d4bac314f044e741ea4b1d2704c54888c83f872c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 15:44:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4930037d6c9d472a87b39c778d8f38ff : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/564ed581b461437bac44970b2833007a : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":6},{"id":"c75c0a976af512f7227f3d53f511f134793208b9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-23 07:54:14.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"90a9a636b7b4ef042c8ee7173eb846cdb344c5c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-23 09:00:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"713e26e5c78c6ac1990ba459d886067d414dc206","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-23 09:01:51.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b369425dd2868c2bddd26b095552f7590f7a968d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-23 09:06:49.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4db2b90337de2176e99ae6bd785ae81acee8c85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 09:21:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a785b80e0f4b8188a22fe77c406a42 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d8f880ee6bb48eb8e5ff7f83cca9590 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":7},{"id":"6d84173e794ddc6aaf4926e2175042668450fc61","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-03-23 09:47:41.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"07a656ffad55c1c1fb851023ece48ca6fffc4334","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-03-28 08:31:18.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6bf6736f2a220145abc5306f342ecc5b69fec97b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 08:34:16.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5a8f1ae4b8708e9f4da7f6445656053f09d05e93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 08:52:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0139f26c36f242518f1adf62cd6b65f4 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ca995bae5141dc999b7b30546263f1 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":7},{"id":"9d8f70322df34bf825271031c98fae891202d444","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 08:52:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"cce5cdbff584c713c036974765120a21dd320fe3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 08:54:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd83eb77cf55462e9a484eea2e4503aa : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d556c622b15a48e1ba76c529952da511237b5d00","revisions":{"24d0142ea42188e29c5bf019ecb07f29cac2efb9":{"kind":"REWORK","_number":1,"created":"2022-03-08 10:33:46.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/832509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/832509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/832509/1"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"subject":"Documenation updates","message":"Documenation updates\n\nMinor corrections and update of the remote debugging guide.\n\nChange-Id: I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24d0142ea42188e29c5bf019ecb07f29cac2efb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24d0142ea42188e29c5bf019ecb07f29cac2efb9"}]},"branch":"refs/heads/master"},"a83d0efd79b87cb101d72047f3c7b8c684992187":{"kind":"REWORK","_number":2,"created":"2022-03-08 11:00:16.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/832509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/832509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/832509/2"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 11:00:00.000000000","tz":60},"subject":"Documentation updates","message":"Documentation updates\n\nMinor corrections and update of the remote debugging guide.\n\nChange-Id: I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a83d0efd79b87cb101d72047f3c7b8c684992187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a83d0efd79b87cb101d72047f3c7b8c684992187"}]},"branch":"refs/heads/master"},"f0cacd16a4f967dc0cc85e91bdae47be8dee0078":{"kind":"REWORK","_number":3,"created":"2022-03-11 16:20:37.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/832509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/832509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/832509/3"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-11 16:20:23.000000000","tz":60},"subject":"Documentation updates","message":"Documentation updates\n\nMinor corrections and update of the remote debugging guide.\n\nChange-Id: I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0cacd16a4f967dc0cc85e91bdae47be8dee0078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0cacd16a4f967dc0cc85e91bdae47be8dee0078"}]},"branch":"refs/heads/master"},"244d79677d9fe6c8dd1a8f62d53ecb02a43087b1":{"kind":"REWORK","_number":4,"created":"2022-03-14 08:35:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/832509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/832509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/832509/4"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-14 08:35:22.000000000","tz":60},"subject":"Documentation updates","message":"Documentation updates\n\nMinor corrections and update of the remote debugging guide.\n\nChange-Id: I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/244d79677d9fe6c8dd1a8f62d53ecb02a43087b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/244d79677d9fe6c8dd1a8f62d53ecb02a43087b1"}]},"branch":"refs/heads/master"},"70dbbf46db24daa7c7ca18fc4a367bc5a554e2e5":{"kind":"REWORK","_number":5,"created":"2022-03-14 08:38:12.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/832509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/832509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/832509/5"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-14 08:37:58.000000000","tz":60},"subject":"Documentation updates","message":"Documentation updates\n\nMinor corrections and update of the remote debugging guide.\n\nChange-Id: I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70dbbf46db24daa7c7ca18fc4a367bc5a554e2e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70dbbf46db24daa7c7ca18fc4a367bc5a554e2e5"}]},"branch":"refs/heads/master"},"c48e1d82df9eae9b6eeae7b0149b910d89a4a47d":{"kind":"REWORK","_number":6,"created":"2022-03-14 15:27:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/832509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/832509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/832509/6"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-14 15:26:58.000000000","tz":60},"subject":"Documentation updates","message":"Documentation updates\n\nMinor corrections and update of the remote debugging guide.\n\nChange-Id: I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c48e1d82df9eae9b6eeae7b0149b910d89a4a47d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c48e1d82df9eae9b6eeae7b0149b910d89a4a47d"}]},"branch":"refs/heads/master"},"d556c622b15a48e1ba76c529952da511237b5d00":{"kind":"REWORK","_number":7,"created":"2022-03-23 09:00:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/09/832509/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/832509/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/832509/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/832509/7"}}},"commit":{"parents":[{"commit":"c5561a3ecb6df6b1d37c751f36b19a32eb1bd133","subject":"Merge \"Use centos amphora image in the FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5561a3ecb6df6b1d37c751f36b19a32eb1bd133"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 10:32:45.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-23 09:00:47.000000000","tz":60},"subject":"Documentation updates","message":"Documentation updates\n\nMinor corrections and update of the remote debugging guide.\n\nChange-Id: I0d08aaf4f02f5575f1a68b65ebc16aafad8a0270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d556c622b15a48e1ba76c529952da511237b5d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d556c622b15a48e1ba76c529952da511237b5d00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
