)]}'
{"id":"openstack%2Foctavia~832537","triplet_id":"openstack%2Foctavia~stable%2Fxena~I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-03-15 17:21:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","subject":"Fix Amphora RSyslog configuration for TCP failover","status":"MERGED","created":"2022-03-08 11:51:25.000000000","updated":"2022-03-15 17:23:35.000000000","submitted":"2022-03-15 17:21:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832537","meta_rev_id":"d4f909f75766c38bba722d1bab792aaeea351a95","_number":832537,"virtual_id_number":832537,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 17:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-03-09 09:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-03-15 15:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-03-15 15:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 14:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 09:35:08.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-03-15 15:06:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d5c77c37f3d861ae4f3abf1a646ed7c3eaea819e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 11:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f85c0fc8fdd1e1556098c1ef9fbacf23e3b26ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f627df6aeccd4431b2fc7c0ad4c435f2 : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac614352b8e84803b8705d652ef68b74 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f70c73ef3b4e17b04e3c688b8432ac : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2153eb76f0441308f7429343ede9275 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc18d162fa244c17b03d930dfbc8e016 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de38c59f11c64b459272b2fdcca26e52 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca00afe0f1b1493f8648d4864634f050 : SUCCESS in 14m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ef51e4c15c44f398d4c8d764340734a : SUCCESS in 6m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c92215dcb732434ebc10d3aab1e2d3f8 : SUCCESS in 9m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b97398c7b0fa4835b06dd7af68e0e12d : SUCCESS in 10m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/54cbd4ae1fa8433981f24f82ed80b164 : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0d4403777714c2ea6680c22009873a3 : SUCCESS in 10m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dd9bc979df00427d90dd1aadf6837df3 : SUCCESS in 2h 27m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c07169a87f124b7a8a987cc64bc9681b : SUCCESS in 1h 28m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca334cdec2d540d39113fc8e1b54f373 : SUCCESS in 39m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/be6a8556ef644d6696078b9c2b4310ec : SUCCESS in 53m 10s\n- octavia-v2-dsvm-scenario-fips https://zuul.opendev.org/t/openstack/build/9a631db196b34a8b9bc13ffe6442385f : SUCCESS in 1h 51m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bcab27ec8a1a424082912adc926320d3 : SUCCESS in 44m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c12c78e3aef34994aef9cc95e6453bf8 : SUCCESS in 44m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d0d41135d8d4a33924107b19efd933c : FAILURE in 1h 32m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e907405828b548b3b236315dd1376cf7 : SUCCESS in 2h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/79b714515044426fb1f744cecc6f6b8c : SUCCESS in 1h 28m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0cf90502203748a0a72eb940235af7e4 : SUCCESS in 1h 31m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/82beba0673f843f896d66ddb0631eb55 : SUCCESS in 1h 22m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d87d002301b9406092fb2bdb778edfe8 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46e41abe9058be0983c967a3e121ce00cd116da3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-09 09:35:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f09ef32e803bb401c9a666240d030bfa09d0684b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-15 15:06:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ff9d3bd92841227b2268047b73c0eefe58f3af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:08:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"040f5d336b269acd60018d95bb898c3dd4670217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6501c598ed7442c0ba1cb62eeb557c95 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f07021cf71439d9d9b9eb9698a8a9d : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18598e6c65c04bf7bd1d0c770e021685 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f5775c4a5474b85a0cad63b3848eb8e : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df03a3a92b35498e8514461343b5e7a2 : SUCCESS in 14m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfd86fdb1981464aa91c8177c3892e1c : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/96aa81d950b241f3b65f4087dc2fe394 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37272eb46a154b62b04e254bafa8d1e3 : SUCCESS in 11m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6f930e7756b44ddb6ac49655284d264 : SUCCESS in 2h 07m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a09aca19808488bb7b65328a5f114f5 : SUCCESS in 1h 45m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4bf3f0e59b64f119423d75f0aa0dff7 : SUCCESS in 34m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/322ec89b718942d6b41f288e708d06fa : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9ebc2096db382d931d4ae7a5f0304832c19d3730","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4f909f75766c38bba722d1bab792aaeea351a95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:23:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b784ebfec44000ba05e3b8bd8df2ec : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bf7971725eb4f8ab0444b46a11ba3a5 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ade96be53f2fd75a49964d9c7ea98f684ec970e","revisions":{"5ade96be53f2fd75a49964d9c7ea98f684ec970e":{"kind":"REWORK","_number":1,"created":"2022-03-08 11:51:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/37/832537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/832537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/832537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/832537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/832537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/832537/1"}}},"commit":{"parents":[{"commit":"1986c89e3f93264fdc0d55921d7ff5c21f3fd6b1","subject":"Merge \"Add fips jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1986c89e3f93264fdc0d55921d7ff5c21f3fd6b1"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-28 12:13:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 11:50:27.000000000","tz":60},"subject":"Fix Amphora RSyslog configuration for TCP failover","message":"Fix Amphora RSyslog configuration for TCP failover\n\nRSyslog failover does not work if queues are defined on the actions. A\nworkaround that allows to use queues with multiple RSyslog servers over\nTCP is to combine all actions in a single ruleset.\n\nStory: 2009876\nTask: 44599\nChange-Id: I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f\n(cherry picked from commit 145f4f804347358689165d1418756fe2a91c07da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ade96be53f2fd75a49964d9c7ea98f684ec970e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ade96be53f2fd75a49964d9c7ea98f684ec970e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
