)]}'
{"id":"openstack%2Foctavia~832589","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-03-15 17:21:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f","subject":"Fix Amphora RSyslog configuration for TCP failover","status":"MERGED","created":"2022-03-08 12:10:44.000000000","updated":"2022-03-15 17:23:41.000000000","submitted":"2022-03-15 17:21:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832589","meta_rev_id":"c0a9ff43c8c79cdd81daeddb03027d9d3147d3a0","_number":832589,"virtual_id_number":832589,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 17:21:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2022-03-09 09:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-03-15 15:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-03-15 15:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 14:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 09:36:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-03-15 15:08:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ee56b7a746f7ef94e71480d38d7e3b2b200a2c44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-03-08 12:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c528fb3964865230b4e39ca90ec650ea7f9acf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:08:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33ad679972824e07aa88a0ae8bd5412a : SUCCESS in 11m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46dcf66116da44d4a91f6c9f6f293edc : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1526ac16b64d72bd60270659791a9d : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9422711033c4cac83dc29fdb1b178da : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70735109cb6e4733b57093edb89e76e7 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02b2abeaa0464b788b45207ab4ac1b95 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b40739d4578448f96cd12bb75a0dddb : SUCCESS in 14m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37c30357e0fb4e3b86c925c93fe7c1e3 : SUCCESS in 7m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d0e920b676dc45ef83a0f6b30e7ba0ee : SUCCESS in 7m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e76e290185ec401c9e9962abcfa399d3 : SUCCESS in 9m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3bff6a2ca1a540ea856c112d41affd16 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f5ca2ff12824816948607e316625f5c : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e73666632dc14454a56fa7ff6558c714 : SUCCESS in 1h 30m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c8908b221244c2db5666b62754c25ab : SUCCESS in 1h 30m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33eb6ab8e789479097c08dfac3964d92 : SUCCESS in 36m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e11f0e5f3f3a4f10b2e1b16d6b60bae8 : SUCCESS in 33m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db2e5204b4054d1981041b0bffaf7558 : SUCCESS in 37m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ca6e25e6959411487962379d65056c8 : SUCCESS in 51m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/838af9a893cc437db929c3b990ef8683 : SUCCESS in 1h 30m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/50035bcaa713414a83e6ce24a916fd06 : FAILURE in 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d3a8781212d4e98a7e6636566c8fe7a : SUCCESS in 1h 27m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/af1ed3d6a0e944869b733829e9a14b8f : SUCCESS in 1h 35m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d1a1573ddef144948b435db49bcab852 : SUCCESS in 1h 37m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/90783ce049144d069d139841f3c0f673 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c69855cb108f8c47ff9d3791bac5903e3886cbf9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-09 09:36:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a05df1b5b49fba9e8ff71ff09d1efab7eaefb4d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-03-15 15:08:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1297bfc583f3d55b14d60de908abfe3c30eb7598","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:13:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b5146f940a490d4cba9c27a5275ba967c619c34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/691c096770ce4b33be98e01090273da9 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c9413938e24a12a2a93f5f25fb17fd : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e41699557f8849cc973fdec31279a9a5 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9b52faf3f554dc1979fdd05587c101d : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e67e934c8742a082a483f8144f642e : SUCCESS in 13m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8c2d24bc58f4bdd8b590feac546f2eb : SUCCESS in 6m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29f2aac207744c378f61a5c7d66068b6 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2463ff43f49949fea9378ec331d8ca1d : SUCCESS in 9m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fd80ec0821c47a88e9fc90bef61fd15 : SUCCESS in 1h 20m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ff201efebed43fb85b8427a37cb5a79 : SUCCESS in 1h 59m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6485dec60d04c8895d8c440f7458d89 : SUCCESS in 41m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/12a8a730b9b54f9b902587aeea99471a : SUCCESS in 47m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68e159a4700b4c2cbd8abafdfdb0ef5c : SUCCESS in 40m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fff6b0b618bc4fa4a598db904d325b50 : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5fddccf323954d330eb9a3f31c99c92183c32b8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:21:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0a9ff43c8c79cdd81daeddb03027d9d3147d3a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:23:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e32029d75c4e79b6eeba832b5d96e7 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe005af3be1e4e7c98a7d2f9aec08c6d : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f134a292f907c2eb8e05aa85b80e7a28c8b44ec","revisions":{"3f134a292f907c2eb8e05aa85b80e7a28c8b44ec":{"kind":"REWORK","_number":1,"created":"2022-03-08 12:10:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/89/832589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/832589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/832589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/832589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/832589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/832589/1"}}},"commit":{"parents":[{"commit":"ce9a43566956d3918998a33aa9ad5f99e713a179","subject":"Fix amphora-agent elements for RHEL9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce9a43566956d3918998a33aa9ad5f99e713a179"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-02-28 12:13:08.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-03-08 12:10:20.000000000","tz":60},"subject":"Fix Amphora RSyslog configuration for TCP failover","message":"Fix Amphora RSyslog configuration for TCP failover\n\nRSyslog failover does not work if queues are defined on the actions. A\nworkaround that allows to use queues with multiple RSyslog servers over\nTCP is to combine all actions in a single ruleset.\n\nConflicts:\n\toctavia/common/jinja/logging/templates/10-rsyslog.conf.template\n\nStory: 2009876\nTask: 44599\nChange-Id: I4a9d14f0f918bda64ec489777ef7c10ceeb9b96f\n(cherry picked from commit 145f4f804347358689165d1418756fe2a91c07da)\n(cherry picked from commit 5ade96be53f2fd75a49964d9c7ea98f684ec970e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f134a292f907c2eb8e05aa85b80e7a28c8b44ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f134a292f907c2eb8e05aa85b80e7a28c8b44ec"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
